)]}'
{"id":"openstack%2Fkolla~810578","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I4325fe031647a04766516afa528639d6ff0b5ae6","project":"openstack/kolla","branch":"stable/wallaby","topic":"bug-1942846-stable/wallaby","hashtags":[],"change_id":"I4325fe031647a04766516afa528639d6ff0b5ae6","subject":"toolbox: Move custom Ansible config to global location","status":"MERGED","created":"2021-09-23 10:15:26.000000000","updated":"2021-09-27 11:41:51.000000000","submitted":"2021-09-27 11:40:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810578-bug-1942846-stable/wallaby","cherry_pick_of_change":807681,"cherry_pick_of_patch_set":8,"meta_rev_id":"6477bb91861365d0d82a855b3f5221d85204ae42","_number":810578,"virtual_id_number":810578,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-27 11:40:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-09-27 10:08:22.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-09-27 09:38:50.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-27 10:24:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-09-27 10:08:22.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-23 11:44:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-23 12:00:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-27 09:38:50.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-09-27 10:08:22.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-09-27 10:24:48.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"0e39ac65065b93f0053487e824a4f09fb839ba84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-23 10:15:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f3751d4f8e5d1cc7d1cce1ea8622ab0450715e27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 11:44:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/83a18a74495641f78a6895f498306497 : SUCCESS in 1h 20m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/53800269ad0749ba9a2e3327aa9c1f7b : SUCCESS in 36m 36s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5ad258d60eaa4faa889b15ceec2a7360 : SUCCESS in 1h 22m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/dfb6e9e65ee947c0bf6f2e5a7d6b437d : SUCCESS in 40m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1dbe3d61b6208ba8ce2e9d6991d7ac514b89276","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 12:00:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4a29c44dce8a461ca5eb274e6cd7b7c1 : SUCCESS in 1h 27m 04s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/dcfcf90316454b119bfa597329fe7667 : SUCCESS in 1h 41m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/dd8b805c12534591a412a7fac2273d96 : SUCCESS in 1h 13m 03s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/bb0cecc1aa7d4723a1ddd6a68c769e07 : SUCCESS in 1h 13m 14s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/cb36ed32d629466eac1da385ca5d1adc : SUCCESS in 1h 18m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c24a6aadce7442b38ca37ac0b551ca27 : SUCCESS in 37m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a70437a4d844418da40b969bfa945dca : SUCCESS in 17m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ecec8eee87e84f23b712ba72cdcc7b6d : SUCCESS in 1h 06m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8efef370ab745889717e616ac911eca : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a973ed1e472a48cf9a3ef62b3e079256 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/058e3f770f234984a67208df1f5a1619 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97da53906a5c43e38120cb3b15c88925 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e79bc5c372ca45cd9e9430ee880eb07b : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1bc09d7544146dfa5bc6ffac07d3936 : SUCCESS in 4m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/960ec140d30b493f839a436d42511a2a : SUCCESS in 20m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/42f2c03b61b3409691b4caa0bcbb62f6 : SUCCESS in 38m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7d787563ce764f269376ed7463bf0ce1 : SUCCESS in 56m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/94e69475b5ab41d8b71cca59e6d923a9 : SUCCESS in 55m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e465ddce1dc7447687adc23c006cd0fb : SUCCESS in 1h 10m 19s","accounts_in_message":[],"_revision_number":1},{"id":"00ac614e98c855bd7071a90ba223f164d34b5ca8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-27 09:38:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bd546a8875d766c42bcbd133e5873af178ba6ca1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-09-27 10:08:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3c3362c1a92743d988731a95d207f94e46aff05d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 10:08:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70ef051ad7a0349f4dc6bad662ea5711e528ec96","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-09-27 10:24:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6b19d1b217f739dd618c34786c05df4389e589f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 11:40:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6c8e72a85b2a46788f39321b58c2d0f1 : SUCCESS in 50m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/bcbca3cb2d844e8c8485689fa9c4d873 : SUCCESS in 1h 12m 36s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/dda42db8175b4bb8a9b66efd8f5e8de2 : SUCCESS in 1h 15m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e523fad471c4d80b98d107ecc6b2e3c : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae59f4ea164748e29b701ca3afab8ebc : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f8d3748eaaa4039919c20c895e8323b : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56260541d8354efa9d962815faf83eec : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dec3e2bbfbd44e98c772b67b38068e6 : SUCCESS in 4m 41s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7dca94eeba424d6abf50e8ccc2f6627c : SUCCESS in 36m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/45a662536691405f8e8a26dd093f9abd : SUCCESS in 55m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8b7de031a392484ea205eb966e57330e : SUCCESS in 56m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d2954e6363d49017f2472d8671270438c0c7a05a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 11:40:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6477bb91861365d0d82a855b3f5221d85204ae42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 11:41:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8a59214b739488892241ab4e7794af2 : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6797cfb52ed64f9486750e5b3cbc7504 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22a003276cafa4e9c82b4e7d2de043f0b450459f","revisions":{"22a003276cafa4e9c82b4e7d2de043f0b450459f":{"kind":"REWORK","_number":1,"created":"2021-09-23 10:15:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/78/810578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/810578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/810578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/810578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/810578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/810578/1"}}},"commit":{"parents":[{"commit":"c978baf14349a4185dbb01bd47aecf7adb14a42d","subject":"Fix Elasticsearch Curator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c978baf14349a4185dbb01bd47aecf7adb14a42d"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 07:47:41.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-23 10:15:26.000000000","tz":0},"subject":"toolbox: Move custom Ansible config to global location","message":"toolbox: Move custom Ansible config to global location\n\nIn order to achieve logging to /var/log/kolla/ansible.log from different\nusers than default ansible - the config needs to be moved to global\nlocation.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/807679\nCloses-Bug: #1942846\nChange-Id: I4325fe031647a04766516afa528639d6ff0b5ae6\n(cherry picked from commit 88f1e32aa997b2b8c3627cd4b2128402262fdef4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22a003276cafa4e9c82b4e7d2de043f0b450459f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22a003276cafa4e9c82b4e7d2de043f0b450459f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
