)]}'
{"id":"openstack%2Fkolla~810744","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I9704e5371a410cb1f69ec0367174e2b368961dc1","project":"openstack/kolla","branch":"stable/wallaby","topic":"debian-mariadb-upstream-stable/wallaby","hashtags":[],"change_id":"I9704e5371a410cb1f69ec0367174e2b368961dc1","subject":"[debian] Use upstream MariaDB","status":"MERGED","created":"2021-09-24 06:38:07.000000000","updated":"2021-09-27 11:56:49.000000000","submitted":"2021-09-27 11:55:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810744-debian-mariadb-upstream-stable/wallaby","cherry_pick_of_change":810161,"cherry_pick_of_patch_set":2,"meta_rev_id":"5cca32e28bdb8570e3d28deaf94797a1c9d40cd2","_number":810744,"virtual_id_number":810744,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-27 11:55:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-09-27 10:09:52.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-09-24 15:31:16.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-27 10:24:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-09-27 10:09:52.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-24 08:37:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-24 15:31:16.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2021-09-27 10:09:52.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-09-27 10:24:31.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"b08350e7d444d374c802b560791560fb084d64fc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-24 06:38:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* kolla/template/repos.yaml","accounts_in_message":[],"_revision_number":1},{"id":"50d2f742f6614859eba2fc3da16a3096765a97db","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-24 06:48:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8aaff8fac92eac1ad7f908ff1cb8bcd870de6a82","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-24 06:49:00.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"7e58ddc14fa8ff69c4540bfeb283d4aca40c462e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-24 07:19:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2ee16508a1100ef91bf7a35d33ff9dae318817f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 08:37:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ae52774940f5415ab9b11e17103ab889 : SUCCESS in 27m 15s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d698310524af47b28d9d5299401542ff : SUCCESS in 35m 56s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5f3be727c9404b438a8b3e1102e90297 : SUCCESS in 1h 14m 51s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6aaadd6400514ff389e4d4ec2cad30b7 : SUCCESS in 1h 16m 36s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/eb7caffe247042b18cf7f0a5da5034a9 : SUCCESS in 1h 11m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/261ad6c94c794d54a4ccb4d7068cc577 : SUCCESS in 47m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/93faa20e430040febc47b423f2d116ef : SUCCESS in 20m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f5058c8435cc4051acc7d22e26184e8c : SUCCESS in 1h 07m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc8918dda2d241cbaaaccf6dd57d9f17 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d691e4fe5af247dbb0e7c0e36d741a42 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56ce5c75a7cf45daa345d59c7c0cdfc6 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa1a1686d1ff42888f0df80c6817ad12 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68291fef46cd4c9c9fa9472ec148c9e3 : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/623345d8a5a04d76ad08b6cb3e250042 : SUCCESS in 4m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e15fb0296e904255a2fd792e1916b31b : SUCCESS in 21m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e918604e4159480e990fcdcbe051b11b : SUCCESS in 27m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2ed6076ccb2a4bd8bbb0cbf44f04c197 : SUCCESS in 54m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/681cafe1c26f4fec8bd5e1d49a82e01a : SUCCESS in 52m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1a0b94e7617c49e9a107fa14c7fd3036 : SUCCESS in 1h 06m 08s","accounts_in_message":[],"_revision_number":3},{"id":"8352b715cdce10cb76b2453a5c2a77edc0288e61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 08:57:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/042a26798bdc46408fb7cfcbe43c0022 : SUCCESS in 1h 03m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ad22f11492bb4ce5a172e9829c98daea : SUCCESS in 1h 20m 19s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8d16e91fb9e4444bb8c8cc560cb01341 : SUCCESS in 1h 34m 15s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/eb320a54ac3e4c89b4c3b351a6c1fb7c : SUCCESS in 1h 32m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0fcf567aeb487bdfb7befc6fad7e2e6864e76589","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2021-09-24 15:31:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e8167e1d4b469967f39829bb0b30d98750af3103","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-09-27 10:09:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9f94290564f0396e0cac4440588e837a4d7710e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 10:10:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3c31001cacecdc05fb8973bfdd764fbaae467e28","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-09-27 10:24:31.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"920d8345c5b5f091e4b049c6dc154575d1116e50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 11:55:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2ef45d1b66c7498297d4eb356116e090 : SUCCESS in 31m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/bf4330201bd0476189e59adc11648527 : SUCCESS in 1h 08m 18s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d2296819adaa4fc48bcb0d168a3a869e : SUCCESS in 1h 12m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f866e28b3d42168127bd3eedb9ae4e : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fe1e725c0b74eacbca05f25e28e311f : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f7e2779c5c14985953fc2d59a8f7d3e : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c68cf9517a44b1cb13e0345fdadf146 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9e53a7f3bda4684a2cf19908b316b05 : SUCCESS in 4m 54s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6e20a2f633b3468db1002da1419091a7 : SUCCESS in 57m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b68e7476b6904e9da436631ec5a658e7 : SUCCESS in 53m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5891206d80954af98175128e1b22afd7 : SUCCESS in 1h 00m 27s","accounts_in_message":[],"_revision_number":3},{"id":"ed163cddf2aa7b683b471e1bda2b02fe27e35735","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 11:55:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"5cca32e28bdb8570e3d28deaf94797a1c9d40cd2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 11:56:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09bc0dae3be54c96927d35c9757efd48 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5843f4d7a9d94bc399735f5e404fca40 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fe7b663235df3b924438cfe10c97d5f1f21e7226","revisions":{"391ae496625d1dee56603abae04d693f2bebb794":{"kind":"REWORK","_number":1,"created":"2021-09-24 06:38:07.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/44/810744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/810744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/810744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/810744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/810744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/810744/1"}}},"commit":{"parents":[{"commit":"c978baf14349a4185dbb01bd47aecf7adb14a42d","subject":"Fix Elasticsearch Curator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c978baf14349a4185dbb01bd47aecf7adb14a42d"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-21 07:40:58.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-24 06:38:07.000000000","tz":0},"subject":"[debian] Use upstream MariaDB","message":"[debian] Use upstream MariaDB\n\nThis preserves the Debian-provided minor version - 10.5 in case\nof Bullseye in Wallaby and Xena - and will be backported as\na fix to Victoria and Ussuri (where it will follow Buster and its\n10.3).\n\nRelated-Bug: #1944410\nChange-Id: I9704e5371a410cb1f69ec0367174e2b368961dc1\n(cherry picked from commit b5422a6bc27797144da1ea973cec24a5cf46d2eb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/391ae496625d1dee56603abae04d693f2bebb794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/391ae496625d1dee56603abae04d693f2bebb794"}]},"branch":"refs/heads/stable/wallaby"},"d24a4d1915ea487b66688f3339adeb553bcd5359":{"kind":"REWORK","_number":2,"created":"2021-09-24 06:48:50.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/44/810744/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/810744/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/810744/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/810744/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/810744/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/810744/2"}}},"commit":{"parents":[{"commit":"c978baf14349a4185dbb01bd47aecf7adb14a42d","subject":"Fix Elasticsearch Curator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c978baf14349a4185dbb01bd47aecf7adb14a42d"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-21 07:40:58.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-24 06:48:45.000000000","tz":0},"subject":"[debian] Use upstream MariaDB","message":"[debian] Use upstream MariaDB\n\nThis preserves the Debian-provided minor version - 10.5 in case\nof Bullseye in Wallaby and Xena - and will be backported as\na fix to Victoria and Ussuri (where it will follow Buster and its\n10.3).\n\nRelated-Bug: #1944410\nChange-Id: I9704e5371a410cb1f69ec0367174e2b368961dc1\n(cherry picked from commit b5422a6bc27797144da1ea973cec24a5cf46d2eb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d24a4d1915ea487b66688f3339adeb553bcd5359"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d24a4d1915ea487b66688f3339adeb553bcd5359"}]},"branch":"refs/heads/stable/wallaby"},"fe7b663235df3b924438cfe10c97d5f1f21e7226":{"kind":"REWORK","_number":3,"created":"2021-09-24 07:19:31.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/44/810744/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/810744/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/810744/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/810744/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/810744/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/810744/3"}}},"commit":{"parents":[{"commit":"c978baf14349a4185dbb01bd47aecf7adb14a42d","subject":"Fix Elasticsearch Curator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c978baf14349a4185dbb01bd47aecf7adb14a42d"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-21 07:40:58.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-24 07:19:22.000000000","tz":0},"subject":"[debian] Use upstream MariaDB","message":"[debian] Use upstream MariaDB\n\nThis preserves the Debian-provided minor version - 10.5 in case\nof Bullseye in Wallaby and Xena - and will be backported as\na fix to Victoria and Ussuri (where it will follow Buster and its\n10.3).\n\nRelated-Bug: #1944410\nChange-Id: I9704e5371a410cb1f69ec0367174e2b368961dc1\n(cherry picked from commit b5422a6bc27797144da1ea973cec24a5cf46d2eb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe7b663235df3b924438cfe10c97d5f1f21e7226"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe7b663235df3b924438cfe10c97d5f1f21e7226"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
