)]}'
{"id":"openstack%2Fkolla~810747","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I9704e5371a410cb1f69ec0367174e2b368961dc1","project":"openstack/kolla","branch":"stable/victoria","topic":"debian-mariadb-upstream-stable/wallaby-stable/victoria","hashtags":[],"change_id":"I9704e5371a410cb1f69ec0367174e2b368961dc1","subject":"[debian] Use upstream MariaDB","status":"MERGED","created":"2021-09-24 07:45:18.000000000","updated":"2021-09-27 11:56:47.000000000","submitted":"2021-09-27 11:55:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810747-debian-mariadb-upstream-stable/wallaby-stable/victoria","cherry_pick_of_change":810744,"cherry_pick_of_patch_set":3,"meta_rev_id":"4b32f5f7d79bd8acdadca4f9aa262c99d7bae8e1","_number":810747,"virtual_id_number":810747,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-27 11:55:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-09-27 10:09:39.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-09-24 15:31:28.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-27 10:24:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-09-27 10:09:39.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-24 08:08:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-24 08:23:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-24 15:31:28.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2021-09-27 10:09:39.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-09-27 10:24:42.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"1beab15b428517b53e0a423e4e305d1d0b2fbc5b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-24 07:45:18.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.\n\nThe following files contain Git conflicts:\n* kolla/template/repos.yaml","accounts_in_message":[],"_revision_number":1},{"id":"243f2dc2798ae758fef9ea1b8575d76f38096773","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 08:08:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5fc3a90780ed47b2bd0cad17d7d384fe : FAILURE in 7m 20s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/bc426724b7554c7c87869d364b41186c : FAILURE in 7m 14s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3630f4302ab045b1b6d8306e1cf49378 : FAILURE in 5m 57s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9ac0757057c04e94993d24bd53c33c5f : FAILURE in 11m 24s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/32055e5a9c354239b4ce021678262b1a : FAILURE in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53c8b0c2153cafc6481e1ffa3a0ab93481070495","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 08:23:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/46e61035d08744b9b81121bdf0ae5236 : FAILURE in 7m 35s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/da7c8a6cbf0c437cb38b922fe6647faf : FAILURE in 5m 54s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e9c78d8a1cee47338f8ca72141354d46 : FAILURE in 6m 06s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9be84b45e0d142d0bc9d2c32a5563903 : FAILURE in 7m 07s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0e17b51dfd414491bb931a51667e8b61 : FAILURE in 11m 43s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/5727082e511f4b93ac3f82e89741207d : FAILURE in 11m 02s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e3fb77c464134bd3bb35fac1fe864937 : FAILURE in 10m 43s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7593b7b140424b558de35953a889ef40 : FAILURE in 10m 32s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b245c9223f0d4ced8f3ad861ea5dc33a : FAILURE in 10m 56s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/901463214cfa4128aad6592c1384c0c4 : FAILURE in 14m 04s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f41dea0aeddb41298a67ccb837fd53a0 : FAILURE in 4m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/116d8447384d4045b6c9f1d94569a30c : FAILURE in 4m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2c20678b8eaa4843a8d81720a27c297f : FAILURE in 9m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4060e49cf9664bda8b91058a0a7fecb0 : FAILURE in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0af709423f2a429390019fb6e92c1ad7 : FAILURE in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec96fb83c2544400876cc606a3ccdc90 : FAILURE in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9eedcc422944284aa9c7461dc6c0019 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7840b22a32345fabb0782b74ec8afd9 : SUCCESS in 4m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/323b23d1faa84a7db19c7b82a977bf98 : FAILURE in 4m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/64b2f28c08cc47f0aebd1a6727a53ffd : FAILURE in 4m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9a047e0ff6cd44bf9c6d7cccc67ec5fd : FAILURE in 8m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/31fad1bdd5404436a7514449258c6a2c : FAILURE in 12m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/754b9b8260d64e4f94c80ea15b672863 : FAILURE in 10m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d74e4fd7228d6e30a51f4222fc422d847a647f22","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-24 09:40:51.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b8259e2559d775aba763031e0e34472cc8e243fb","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-24 09:44:45.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"d48a3ea662c27d3a212d46472c4a802d55c33285","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-24 10:03:29.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"eb8c085a963a75d7e39033f1bd6e0d3720e2fb4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 11:25:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/164ba462b52e4f3c991f8f6e77736d5c : SUCCESS in 23m 32s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2db6fd931072458e8c3839170d281735 : SUCCESS in 27m 28s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/bec28101565e46d08b7610e4b3f8178e : SUCCESS in 27m 52s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a09410acc7934a9f90b04f9531737f21 : SUCCESS in 27m 24s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b39450fee479404bbf329e730b9f4986 : SUCCESS in 1h 17m 26s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1e2eca60996e4e2eac610e5a11367c9e : SUCCESS in 59m 36s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9708112b31f14105b52426987685323f : SUCCESS in 1h 04m 43s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9be44ad2c38444dd9c0100df88ec0ffd : SUCCESS in 1h 15m 38s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/3113d31356664bd9800631ed34b65193 : SUCCESS in 1h 03m 21s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/02e48787aeff48db8113caa90485bcf7 : SUCCESS in 1h 05m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/17412b96218f420583296a6fe8d64307 : SUCCESS in 21m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/649a3bc79cbe45cb9dce9d525b9d6c3d : SUCCESS in 19m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fc7aabe74ba64434b1e42b87ac809998 : SUCCESS in 57m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/117397b55c6846228a192f4881ee50f6 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e4c354177434c6b87880c992a2bf5e9 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d340df1908fc435ebe2a0a230d82e752 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d91d77d706045429cc54dd08c68e0b2 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c67fee4b4c44c0aa86cb36aa851463b : SUCCESS in 4m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/403d19c488fc43058a09f91c8dd48278 : SUCCESS in 21m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a94decc8d33b441a88bc7cb3a548c3ea : SUCCESS in 21m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b7fc3f4c14b4423aaf6513cc504464a : SUCCESS in 51m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/881080b628a74d369104a04ea938de19 : SUCCESS in 48m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3a788a21c92a45b881f75401f7740a48 : SUCCESS in 1h 05m 50s","accounts_in_message":[],"_revision_number":3},{"id":"fe03c82d1808767fac9a8f5d7ccb1b8a8da76ec3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 11:56:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/a9dddf82bbb4470abfdeb8415d48753f : SUCCESS in 50m 26s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/504a208b83494ef6ac6213ce5d0d8df6 : SUCCESS in 1h 40m 57s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8e70b71bba894ecabc4c506b13971630 : SUCCESS in 1h 18m 00s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/348754c55e31436eb9ae358564fd77f4 : SUCCESS in 1h 47m 24s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/451712e9359048ac83d31e15b8c87b2f : SUCCESS in 44m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6097802d5a5a66bce95264a9412fa0b4ad152ef8","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2021-09-24 15:31:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2cecdb9a51d95a5390551bcca03531cc5f3d8b6e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-09-27 10:09:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dd6c7322039a9c4de81f3aef32b3c5b17588a26d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 10:10:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fe2cfe8f3ccf10d9905e1db7c34a665a3e13f73f","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-09-27 10:24:42.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2ce0c31b1e6321406a4a8d5e4dc7b7619b8d095f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 11:55:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/055194b68b584a248d8fe1071a68ef8e : SUCCESS in 32m 52s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/02438e1a060f40b0b4540bcfaba0d465 : SUCCESS in 29m 17s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/61f98e343bff40149e095b48331fd852 : SUCCESS in 1h 04m 50s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0be21656c5ae4b008457da86f8f49e71 : SUCCESS in 1h 44m 54s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c69519143f844c4696a9c682bddb8dfa : SUCCESS in 1h 22m 24s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/38bea237c63d4ec3974bfecbbff33c81 : SUCCESS in 1h 05m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23d769ed800f4ab694a44e87d292726c : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be6a6c1a88b94478b970a77103f65c7e : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b83493390805493b9b2b296c3314ac8e : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e7768cb160c4392b7b28739790d6607 : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b800c00cc5b4b6488d69ec291ab28bf : SUCCESS in 4m 29s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/51466bb945404038b1db210e87559728 : SUCCESS in 21m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1478e2f1652d48fab1ee88e9e360a4b9 : SUCCESS in 46m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d633e0b9e3834548b1f6e056f6659a5d : SUCCESS in 1h 07m 08s","accounts_in_message":[],"_revision_number":3},{"id":"974f99d9d1f4a221fe00b1ac66755b91d22832f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 11:55:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4b32f5f7d79bd8acdadca4f9aa262c99d7bae8e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 11:56:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2dbf3e429164bb7bc9d66221ddbb4d8 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c19d6eb20ec74ad8bcd4389aa9895a1f : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8c9bcb0e7dcb652227995db1af3e9f6205943371","revisions":{"21a318d5788e65a1c289fe8517caa2c61dadf721":{"kind":"REWORK","_number":1,"created":"2021-09-24 07:45:18.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/47/810747/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/810747/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/810747/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/810747/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/810747/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/810747/1"}}},"commit":{"parents":[{"commit":"988564428a6b752d653f8be434511f435ee63f68","subject":"Merge \"[doc] Fix Ceph sources\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/988564428a6b752d653f8be434511f435ee63f68"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-21 07:40:58.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-24 07:45:18.000000000","tz":0},"subject":"[debian] Use upstream MariaDB","message":"[debian] Use upstream MariaDB\n\nThis preserves the Debian-provided minor version - 10.5 in case\nof Bullseye in Wallaby and Xena - and will be backported as\na fix to Victoria and Ussuri (where it will follow Buster and its\n10.3).\n\nRelated-Bug: #1944410\nChange-Id: I9704e5371a410cb1f69ec0367174e2b368961dc1\n(cherry picked from commit b5422a6bc27797144da1ea973cec24a5cf46d2eb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21a318d5788e65a1c289fe8517caa2c61dadf721"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21a318d5788e65a1c289fe8517caa2c61dadf721"}]},"branch":"refs/heads/stable/victoria"},"e66dfc29927d43b1aa58bf69d7c7b57f034c8377":{"kind":"REWORK","_number":2,"created":"2021-09-24 09:40:51.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/47/810747/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/810747/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/810747/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/810747/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/810747/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/810747/2"}}},"commit":{"parents":[{"commit":"988564428a6b752d653f8be434511f435ee63f68","subject":"Merge \"[doc] Fix Ceph sources\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/988564428a6b752d653f8be434511f435ee63f68"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-21 07:40:58.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-24 09:40:40.000000000","tz":0},"subject":"[debian] Use upstream MariaDB","message":"[debian] Use upstream MariaDB\n\nThis preserves the Debian-provided minor version - 10.5 in case\nof Bullseye in Wallaby and Xena - and will be backported as\na fix to Victoria and Ussuri (where it will follow Buster and its\n10.3).\n\nRelated-Bug: #1944410\nChange-Id: I9704e5371a410cb1f69ec0367174e2b368961dc1\n(cherry picked from commit b5422a6bc27797144da1ea973cec24a5cf46d2eb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e66dfc29927d43b1aa58bf69d7c7b57f034c8377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e66dfc29927d43b1aa58bf69d7c7b57f034c8377"}]},"branch":"refs/heads/stable/victoria"},"8c9bcb0e7dcb652227995db1af3e9f6205943371":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-09-24 10:03:29.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/47/810747/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/810747/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/810747/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/810747/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/810747/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/810747/3"}}},"commit":{"parents":[{"commit":"988564428a6b752d653f8be434511f435ee63f68","subject":"Merge \"[doc] Fix Ceph sources\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/988564428a6b752d653f8be434511f435ee63f68"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-21 07:40:58.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-24 10:03:26.000000000","tz":0},"subject":"[debian] Use upstream MariaDB","message":"[debian] Use upstream MariaDB\n\nThis preserves the Debian-provided minor version - 10.5 in case\nof Bullseye in Wallaby and Xena - and is backported as\na fix to Victoria and Ussuri (where it follows Buster and its\n10.3).\n\nCloses-Bug: #1944410\nChange-Id: I9704e5371a410cb1f69ec0367174e2b368961dc1\n(cherry picked from commit b5422a6bc27797144da1ea973cec24a5cf46d2eb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c9bcb0e7dcb652227995db1af3e9f6205943371"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c9bcb0e7dcb652227995db1af3e9f6205943371"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
