)]}'
{"id":"openstack%2Fkolla~811015","triplet_id":"openstack%2Fkolla~master~Ibcfbaa1903c7059e437bb2118bc4c9de4a9b854e","project":"openstack/kolla","branch":"master","topic":"mariadb-10.6","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2021-11-22 09:57:07.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2021-11-16 09:31:03.000000000","reason":"Marcin Juszkiewicz replied on the change"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-11-22 09:57:07.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-11-22 09:57:07.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2021-11-04 19:06:06.000000000","reason":"Michal Arbet replied on the change"}},"hashtags":[],"change_id":"Ibcfbaa1903c7059e437bb2118bc4c9de4a9b854e","subject":"Use MariaDB 10.6","status":"MERGED","created":"2021-09-25 19:47:31.000000000","updated":"2021-11-22 09:58:44.000000000","submitted":"2021-11-22 09:57:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"811015-mariadb-10.6","meta_rev_id":"ef50f01f142f114d2f7137ab111baa4f42b986f7","_number":811015,"virtual_id_number":811015,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29880,"name":"Adrian Andreias","email":"adrian@fleio.com","username":"adrian-fleio"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-22 09:57:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":29880,"name":"Adrian Andreias","email":"adrian@fleio.com","username":"adrian-fleio"},"all":[{"value":1,"date":"2021-11-09 23:01:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29880,"name":"Adrian Andreias","email":"adrian@fleio.com","username":"adrian-fleio"},{"value":2,"date":"2021-11-16 09:31:03.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2021-11-04 19:06:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2021-11-22 08:22:12.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":29880,"name":"Adrian Andreias","email":"adrian@fleio.com","username":"adrian-fleio"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":1,"date":"2021-11-22 08:22:12.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29880,"name":"Adrian Andreias","email":"adrian@fleio.com","username":"adrian-fleio"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29880,"name":"Adrian Andreias","email":"adrian@fleio.com","username":"adrian-fleio"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":29880,"name":"Adrian Andreias","email":"adrian@fleio.com","username":"adrian-fleio"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-25 21:04:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-04 13:08:57.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-11-04 13:08:57.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-11-04 13:08:57.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2021-11-04 13:08:57.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-11-09 23:01:44.000000000","updated_by":{"_account_id":29880,"name":"Adrian Andreias","email":"adrian@fleio.com","username":"adrian-fleio"},"reviewer":{"_account_id":29880,"name":"Adrian Andreias","email":"adrian@fleio.com","username":"adrian-fleio"},"state":"REVIEWER"},{"updated":"2021-11-22 08:22:12.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"}],"messages":[{"id":"5c37e9bf5adc915aa0e3bf73b1ea1206892a5fb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-25 19:47:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"821e894d9c4570311e02ffa5528fbac47a1e7462","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-25 21:04:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9b02e793c614424dbe44123f09f0ab20 : FAILURE in 30m 05s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a4b5d9e06fee4346aeb7786b9a6bfb97 : FAILURE in 45m 19s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d7da262f0419453a96845def040ee1a0 : FAILURE in 25m 18s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c9f8eb4fb4924258879a8b8b5fd712bb : FAILURE in 34m 36s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3cb1c58fd2574ffea7fe6b7decda7237 : FAILURE in 27m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/10308daf4cec44c693e9568467689845 : SUCCESS in 23m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e8475b847d00429a943a1ea89c8f3ec1 : SUCCESS in 18m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5069bc9f3b6d4425bafa5069d7e41ef9 : SUCCESS in 1h 07m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87b1f6131b5e4683b1230a5e8651e5f3 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98d24cc0fd084cf2b90a74385c131ccd : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df4e34ff30fb4407ab352a665ebad605 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07a474dcfff14fe9a518b010d0576fea : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/413df9b1cbfe4a7c89f15ff89febcb8b : SUCCESS in 5m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b383af5df70a4cd89ff560e62285ffc8 : SUCCESS in 18m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6660418962b7404d9a1ed4a869341feb : SUCCESS in 27m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/40423bf9c2414881a77dd797b4cdb4a1 : FAILURE in 31m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bbc98e8ff76e4efd8ffa6680e8876098 : FAILURE in 29m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d57504c0699041ad857d9eaecca8e106 : SUCCESS in 1h 02m 57s","accounts_in_message":[],"_revision_number":1},{"id":"05e3d8ded86b533317fa06804c666867fc3efe6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-25 22:21:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/bb0312001d0f4460a81886c2d6a6b298 : FAILURE in 2h 30m 53s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/926c53ab8a9d40209cc3b29e622c8c9c : SUCCESS in 2h 11m 51s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b0b07431779748aaa2a540f7c82b85f3 : SUCCESS in 2h 14m 17s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/38c7dfd70797487caeef3ad1c0e4cde2 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a52f2ee1654c4cea8cce62f147ab2cdd : SUCCESS in 6m 57s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/96dfc71ac8754d138323b6a3d5e33c6e : SUCCESS in 1h 43m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b719e4ed213751e693fe467624053346a30f896","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-26 08:28:37.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e6d91b24245aad18bd314cc035816576ccb1679f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-26 10:04:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/5a4131c389ff41ae84cdcae02ab71258 : SUCCESS in 1h 21m 36s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6308cb6a5c9943b79dbee6d77dc5ce62 : SUCCESS in 1h 34m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8e02594d55f8480a91c3aa4135844dc7 : FAILURE in 30m 36s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a9c5b222282946549eb82fd4e93fdd39 : FAILURE in 1h 13m 16s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/36e07bba8bea404bba92b61a1d6bf39f : FAILURE in 1h 20m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5b59f10068494786905fb368f84b7f26 : SUCCESS in 29m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b41d6ba0abdb43b9904cdef81904d3ae : SUCCESS in 15m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/db1c93e6813c432d8ecb8d71d77e9cfe : SUCCESS in 1h 04m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/754b9deb151949e085f3690c53bf138f : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b20d048ab2c4402faf00b581b4b8f6c5 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e1bf749823541008788539eb8c92a92 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3c1f01e1fa04e5eb15c46144a5f1ea6 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc718e1c866a4e00bb72220780b96db0 : SUCCESS in 6m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/76d4223c56844008b31779a7935fd243 : SUCCESS in 37m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/be9d0ef6dcdf4288acc4df0a44e4e9f0 : SUCCESS in 27m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8ddc1e44ac424f1abe7489854a3e3817 : SUCCESS in 46m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5a50b22fbc7d49c9b4ac3ac26ec23713 : SUCCESS in 1h 01m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f82a294029ff401a8a3658eea19b0110 : SUCCESS in 58m 19s","accounts_in_message":[],"_revision_number":2},{"id":"5d170731dfa6409592b0efe785965aed2f7e78f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-26 10:33:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9eb067ccb1a34adcaf0c748ffa0f3a2e : SUCCESS in 1h 21m 55s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/42af8756b1664d1a81c1e47c05083303 : SUCCESS in 1h 04m 34s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/232d0758572f4ed2aaf73bb319eb5126 : SUCCESS in 1h 25m 02s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd7509025cc64a159fec9226fa01042f : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad80a9d775064c95b54dd9143ff6eb70 : SUCCESS in 7m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/eab7eab57d9a4ec8a5bccb931553005e : SUCCESS in 50m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e4d8052b5bc692fda3c7775ff1fbad1777f66737","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-26 13:43:55.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"50869c7e23626cc8c570f7c5c9390a84dcefd179","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-26 15:15:15.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fc372b0d60da48758b0a5d69ccb9ad17 : SUCCESS in 27m 42s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d777a7e05c4342a18ec7a8e25cbc9dfd : SUCCESS in 39m 15s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/54e24d0eafa54adb80ab079ac7fb42dc : FAILURE in 33m 58s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8fcf1b4f8ba2425bad6a08e0e6267dd9 : FAILURE in 34m 07s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5b38a9ddb8b84ec69bf57fdc652a8d8f : FAILURE in 1h 25m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/83e9c3c9f37f4bebaee6f8f5f37319af : SUCCESS in 32m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/316c468c75d6474cbe60329687a358c4 : SUCCESS in 22m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6664e61cb44642208351e6a17b048da9 : SUCCESS in 58m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fc6dfb7caeb47a98335630e9c8437db : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d13026c3eeab420ab224da13f9af5d33 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ec832989f1c45fca69e6bb44b423177 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/753bb51ca25f41acad67b8a2fdd14656 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca2af9c09dc84dc2a037901626eca611 : SUCCESS in 6m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/91ba12012416424aa84e4b2285a81be9 : SUCCESS in 18m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8bc8c5f5166d4e31927493fb759cbc19 : SUCCESS in 34m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/62e5fef9da3645d89d626a5fb35a92ae : SUCCESS in 51m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f825e2621f30430c83f34a0b4e561850 : SUCCESS in 51m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/661614d616db4215a64ebb78c428c081 : SUCCESS in 1h 23m 13s","accounts_in_message":[],"_revision_number":3},{"id":"b65f1023dfebc51a6412c95a53b89908ab1e8b7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-26 15:55:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/30157bcdf1e14d148c6bc3d457c61427 : SUCCESS in 1h 54m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cf052b3ed24b40c8a32d81bbe9f9efa0 : SUCCESS in 1h 39m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dc7886f6520348a8907db644d975aa19 : SUCCESS in 2h 09m 12s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/17c70c36f38940f9b071a43ebf957a64 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e00efe88719d471282c518b1bfa28fae : SUCCESS in 7m 52s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/fea1d10edc654e97b372f31570532672 : SUCCESS in 33m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9506c993575f0e7bc843bbbd4ae2064d5752f4ce","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-26 16:39:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"42f39d075438fa957e081547d55e9118e2553bfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-26 16:39:21.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"54165abe07548c5fc5a24eb45b172633f3f2fd70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-26 16:39:57.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"0bce9efc37251b7f87aa056fae6ff00b03377db3","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-26 17:37:44.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":5},{"id":"5b729240d8ba93020128b07558a20d2661b5a157","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-26 17:37:50.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"e5cfca8d591cce5d6bb2f7ce7fb2a3d8621ce46f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-26 17:55:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9a75bc35d53d40f3bc7e7a90360fae0b : SUCCESS in 44m 59s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/928824b5940b401ea732433de97c8fea : SUCCESS in 46m 04s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/94607b2291ed485691d46e38d3c80afd : FAILURE in 43m 20s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c8e64a5ce4d44e63a878d3a305a3982a : FAILURE in 51m 47s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9d6e74ff7d51441296321b561a665c82 : FAILURE in 1h 12m 31s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e77899d439594611b53b199da18e8736 : SUCCESS in 35m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c9609fb34d45432abb1439e2c40dabb7 : SUCCESS in 21m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f4d852ce5315453fad135bd35da19398 : SUCCESS in 1h 02m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fad41cfe3b342df9f611e13f387550d : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3adc6cf4ee14d2391a9a235a4926b0a : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c1fc3c96cce41c8951e8f3760bd3229 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98c359a2c24b4bcab76a5f6caf1814d4 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa48ba1368814d11b7bef7a37feeca69 : SUCCESS in 6m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a354e88f083e4e8397e0a262c4c9855b : SUCCESS in 20m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e6d37d92aca74401bcc79e6c26241b51 : SUCCESS in 30m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8362cbaa0c1945d9bb96432573fed5e8 : SUCCESS in 53m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9903a2cd01204a849c77d55834890e53 : SUCCESS in 53m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6e811cd7f50d4c6f9ff19d3b0b32c9b5 : SUCCESS in 1h 02m 57s","accounts_in_message":[],"_revision_number":5},{"id":"cdb878e3aa00551d9934c4edc31ec19aac77c9f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-26 18:42:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/26b719e9ae744b6480d35a8c5993999c : SUCCESS in 1h 53m 31s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2edc52e581d14e12bb873868786e5bbe : SUCCESS in 1h 32m 56s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3999f6e9ee9e4d298502ab9b359904f3 : SUCCESS in 1h 59m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ddff23ca10f45d5b4f9d88a138ce134 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0017c029c6ee40b58aa38333969a3e50 : SUCCESS in 7m 24s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f8aa82aa8bcf45c09c6cd3147535e7ae : SUCCESS in 1h 31m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"518af689e6d6a6a3ac44c4fecbac46350fe0cea2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-28 08:54:29.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"b13ffcb96ea0572b9f0f70169c710638a5a2e791","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 10:15:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/07a3b3d1119c4145b115bf74775ece9e : SUCCESS in 25m 34s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/315661decd184a998135ddc227f3d460 : SUCCESS in 34m 12s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4090ef4c441346f2baad69086363b34e : FAILURE in 43m 32s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/307cda6ac05a4d42b2750cae55c611e7 : FAILURE in 43m 55s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/42bd4a1818274c3b931883159cec3bed : FAILURE in 1h 16m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0a098b0ebf024da7a42e348562097475 : SUCCESS in 33m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5fa7dd36ca8a4f04b26bf921cb54b3fc : SUCCESS in 19m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/52174ea008154b2f946ddbc24ed251c5 : SUCCESS in 49m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cd7b8632bba41f8b91ef9ebb75b3051 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65f02dfe4e3a46e1a9513d3b2f5b0795 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/913d079eca80497c9dfb13c6bcc5ff0b : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36b2e1a6ae8c4ca4a73c557d22f0cf15 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/858a40facd554c8586e8f9e5103a79f7 : SUCCESS in 6m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8b57eb1a4fa744dca599bc57c917dfac : SUCCESS in 19m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d80c2a58bd3746be904aa6628a0bc7e1 : SUCCESS in 36m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/40c285139b3b4d0b9166680df82cf38f : SUCCESS in 53m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/43d4d52b4184427f96f6930ade0f3cc7 : SUCCESS in 53m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/02af77626e174fcfba0b92af087f19d8 : SUCCESS in 1h 06m 12s","accounts_in_message":[],"_revision_number":6},{"id":"2780127a14f5830bb35534ee5a98d437c00d8b88","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-28 11:11:16.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"5b76dfe3f9a8c98ce970ea406487bea8f9f4c702","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 11:18:13.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/eabbb32dd95c4665ba4cef8834802c81 : SUCCESS in 2h 21m 12s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/28eb1b72b52945ec8de1c3358c159386 : SUCCESS in 44m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3014a468ca9f481396fa3f85ce98f362 : SUCCESS in 2h 19m 08s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67acc2fdb8a342a69c51e8d784c40b5f : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f1fc696ed3443f49be85d3406750cf3 : SUCCESS in 7m 50s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/35953a4ad52f421d8eda46732111e3aa : SUCCESS in 1h 48m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"78b5037d6be73e2567a3799f6a6e43ae6742b3f8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-28 11:41:25.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"0d1e66b92900c75caef513fcdcec272073c7339b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 13:07:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8e88f2bc19da4126b749806f824ab3b1 : SUCCESS in 27m 05s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/72ad44ca92d747d8b040600a40cb0827 : SUCCESS in 32m 13s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/01909e3812fa4debbb096a9f8d0df441 : SUCCESS in 1h 11m 29s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b908e1937ceb426eb127c5ac6caa5011 : SUCCESS in 1h 07m 04s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ce22b1f761fe4ff1b2478a9eb5cf2542 : SUCCESS in 1h 22m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0acd49e36c8640aeb9f599c96d5b612d : SUCCESS in 30m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6e82758bcd6f4b6e9c8d2139330d40ec : SUCCESS in 18m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b67ee37b5d534557a5d304ad2f94605c : SUCCESS in 57m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a76993838a3b43768457e6a40afd7e74 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8babebf062c8479dbe6362a7484e772c : SUCCESS in 10m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3c5713b24ae41c0a535f7410e28d35f : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f6b8dbdc7934cd5ad7e79d5f7cf8f5b : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/858e1b94fc0948a0a729a599cef297d7 : SUCCESS in 7m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6f7692281ad74cc4a2deb5f88a3cb2ae : SUCCESS in 21m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a735be6c7920475ca9b7c528ed414959 : SUCCESS in 31m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9a6b579563ca4b5a894d03ebe4d644b4 : SUCCESS in 48m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/afb4c33fde2b4aae9bd447aceadd8398 : SUCCESS in 53m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/23d7cc938b3a47868937407236d984b1 : SUCCESS in 1h 03m 41s","accounts_in_message":[],"_revision_number":7},{"id":"d23fe4d13a826b3c65d2c61040082f7cecba8dd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 13:38:16.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e748e604e0dc49059eda6c68b2557e28 : SUCCESS in 1h 54m 03s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/99c35ba505e3451b91cdd01047d8c587 : SUCCESS in 1h 19m 23s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cd843d14bc6044c0ac97d0d0d4a8a45d : FAILURE in 1h 33m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46b5379c72484011b1430d6ff384d77c : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b1b8ffde3bc47248637d51d56a91220 : SUCCESS in 5m 27s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/16586ad321634478bf841aadc5642647 : SUCCESS in 42m 11s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"778448a2232b014abaec97b9eff683719158f8b0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-28 17:40:39.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"a9e5f56c3b4edc1610e7a8e923c24fcf65c12715","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-28 17:52:23.000000000","message":"Patch Set 8:\n\nextensive testing in https://review.opendev.org/c/openstack/kolla-ansible/+/811466","accounts_in_message":[],"_revision_number":8},{"id":"ae9a51387ef38f52f71c795b981ea7be588dbce1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 19:45:26.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/cb3ebaa0488e44d19d2d7452167e708e : SUCCESS in 24m 06s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/37ed6c4e2d844e4094287f8880937b15 : SUCCESS in 39m 23s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f284b548fb234b0e945ce6c58b4cad0e : SUCCESS in 1h 14m 31s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2ab8c2d5b87f450da48a15243fd0707c : SUCCESS in 1h 06m 43s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d46d1928d01e41bda4d12c1731462b5b : SUCCESS in 1h 58m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9f723b641a0f470b85718883b5f9e232 : SUCCESS in 28m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9b4300b1d5fe42b0bae20956efdb4b6a : SUCCESS in 16m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0944fac12cfa430a9465ee7f75bb58f9 : SUCCESS in 57m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e365263fa4e421e961a671ad8f8246a : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ce31b9530d542b888da25fd4f94e95b : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf85afa8da1a48c488d88983dfd231d5 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b559d35a44446959a462d18f55fcb72 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93eaea458d3c4e32b07127517ee91e78 : SUCCESS in 8m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bf07abc743144b43b8d6852a52a69f90 : SUCCESS in 31m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fa53a3f109d04b949ef512dc42a95472 : SUCCESS in 37m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/92c84d8d17ff4fbb9358a2d82f34fa69 : SUCCESS in 49m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/361382870d974437bd5a43fc9e43131d : SUCCESS in 1h 08m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/890ccc35c781445fb6cd6b69e2e61d8b : SUCCESS in 1h 07m 11s","accounts_in_message":[],"_revision_number":8},{"id":"093cda967820a347535e11d5ddf160cfdfdbd163","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 20:02:28.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/91e6f612c46143dd80f45017e6b35ecf : SUCCESS in 2h 18m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/504541474c264cbaac84eb280875e572 : SUCCESS in 2h 02m 28s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3cd8c5309c6d42a8b44e3b01f3b0df20 : SUCCESS in 2h 14m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/16fc02cc2c444a898d7809c3b8d4bdc9 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b95f31b6d1db46b2a7c7a0fc262336bc : SUCCESS in 6m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a20343c36e754c6ba822592d39a6fc1c : SUCCESS in 1h 30m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bb14e421e2339509f35284f0a3b32f4ba77f5209","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-08 15:13:55.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"1e259647020796946ffed967dcd594f8a9322bc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 16:50:57.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/41d2381198a84c4294c400cef7af21b3 : SUCCESS in 37m 20s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fc03994d45c842218a6f5b6bfca004f6 : SUCCESS in 50m 54s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/deb7b48b2f754a98a75e8689d6bf6b46 : SUCCESS in 1h 18m 19s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/28cccf9868ab4c47b53aced733f9c406 : SUCCESS in 1h 16m 58s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/707f9f8f345a47788f1f7bfaf09c50d9 : FAILURE in 1h 35m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a49017becacd473dadf6ad955f12cea0 : SUCCESS in 21m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3a338a3a0ac74e36ad8235bbeb118e96 : SUCCESS in 17m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/39e105f15e864bc0ad3a945e46174fe5 : SUCCESS in 1h 00m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99731b49242d4fc6abb4c59091efeacf : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f96133a30ab8458f86297d90a8b9b58e : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa0fb695c2c8433798181abe672c1414 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa0caca71f634f539cdc77c30ac64f9e : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2880fc87c07e4816b5a856ce40358536 : SUCCESS in 6m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a4ec1aa2fb964687bf05c1b051799bc9 : FAILURE in 20m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/95f7632bac4f4251a2943d49dc1d3332 : SUCCESS in 36m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6ba098ec27cc4f99b0d9aef39078cf47 : SUCCESS in 50m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/97edcab1e95e4fbb9669ca7b044f2b06 : SUCCESS in 51m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1281b2078fc64b61a45d435f0478bcd7 : FAILURE in 1h 08m 46s","accounts_in_message":[],"_revision_number":9},{"id":"a2f59ac7eeedb4d201b17f7df2b305c1807d2dc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 18:24:40.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4257bfcea662429eb1583552ecf56ca5 : SUCCESS in 2h 48m 57s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b6f71c955fc24d45b108068d566119fb : SUCCESS in 2h 21m 36s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/38515acdd43642b28c04d96c0b8f1e03 : SUCCESS in 2h 55m 02s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f204b18ac9646f497d4cf5ddc42eb88 : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c9f93cb907d74d449b4f223f25d9927c : SUCCESS in 6m 07s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/39cda11b20454862b56ac5669bfc3ec8 : SUCCESS in 2h 22m 42s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"08577943b4162e9ba7cd678e27333dc2eb051191","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-17 07:06:26.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8f55bf0d0a88a4a349dd05e9a7c763effa53ddd2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-17 07:06:41.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"2fbdcefea4998c6676b338bde8e76bd44ffc8c25","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-17 07:07:34.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"c6aca3d442481fc7747ff150147c083fee4c5e86","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-17 07:07:40.000000000","message":"Topic dnm-mariadb-10.6 removed","accounts_in_message":[],"_revision_number":11},{"id":"09321bf86d22bd3aed4607b0fd109651952fb581","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-17 07:07:44.000000000","message":"Topic set to mariadb-10.6","accounts_in_message":[],"_revision_number":11},{"id":"ef1d7bc2cf51f23dcbff362e93b7fc3bf957615f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-17 08:40:15.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/66bb1f0e7b9e4ebcb5e419c3520ba909 : SUCCESS in 24m 58s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9992eb871f2748408621e20e574bb0ca : SUCCESS in 41m 48s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b37ee3d06b2d408f851c682d1615c8dd : SUCCESS in 1h 03m 52s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/15a4d7f7f13e4ca8b33829a2c52f4491 : SUCCESS in 1h 07m 58s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2935475957d54baf8a5c09b8cfb11aae : SUCCESS in 1h 30m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/57418df50f53408bae5766d3982d8044 : SUCCESS in 25m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/605f051e2f9a4d54808b689c3e550973 : FAILURE in 19m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/936b457c81c340caa2252bb03886f366 : SUCCESS in 1h 10m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/770272902f5c41afa2d1e3b32129afe9 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cac6bd807eb94068952fefd14029ee54 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb5f89b7640a4ea99cdd67837e701944 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2de499897e994f97a03e9590dcbb2c23 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55987c7b08ea4c4193e814a9e192f1f8 : SUCCESS in 6m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4b6c527fc83a40828b9796f929ff44d6 : SUCCESS in 27m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ac76fee93820430a95d513e66e8a1b7f : SUCCESS in 25m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8ed0f33cd6c04608bb65a81861b1da97 : SUCCESS in 49m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/987830a8cef3477fb289d2979f1a0cdc : SUCCESS in 54m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/afa9450bb7bc428c81b81a6fd9fde5a7 : SUCCESS in 51m 01s","accounts_in_message":[],"_revision_number":11},{"id":"99dc58b5f49d2a23e559cf92fb8650838c849ee5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-17 09:22:37.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/92ae89abe78643ca940194c895728f81 : SUCCESS in 2h 12m 13s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/42a7689fc3504b1abe6c900d91890efd : SUCCESS in 1h 48m 51s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4291ba28ad104ba68b17ac7efb9e5431 : SUCCESS in 2h 05m 58s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/44d2a45d14944aa883104b7103d6e410 : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d061ef485d11423abcc981cc81655d93 : SUCCESS in 7m 20s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/cd00050faea84fd99c8ff18151afb5e0 : SUCCESS in 1h 59m 42s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ece49bbc4e873dfe191f0722f803473cc497116f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-04 12:53:24.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"59250179df8351fae36d2fb8d62e4588f5f2c443","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-04 13:08:24.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":12},{"id":"d0429282e23daee67469fef8047acb5d082a3714","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-04 13:09:24.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c79b86eb1c5d40355a57000d67be11f17c5a3eed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 14:28:15.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/0f57324bad4e476eb1708dee73b76658 : SUCCESS in 30m 37s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8ec0206653694b56a3bbf53c7dfab941 : SUCCESS in 38m 53s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9ee1580c3f4a46faa766c91a10647561 : SUCCESS in 1h 04m 55s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/5ac4619e253a4bb994717fd1700073fb : SUCCESS in 1h 03m 20s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/bbfefedb501e46d19449b234e6118e2a : SUCCESS in 1h 23m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/217c5394ad8b4d049220436965c1015f : SUCCESS in 32m 33s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0f74a9c546d449d58498503ce07f5c3b : FAILURE in 25m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b4b0b6a568994d50a9fee9f998b6c6eb : SUCCESS in 54m 05s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c7d337c1c62443fba7d80e35ec92c9be : SUCCESS in 1h 22m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac05e921870b4da18e6b3e77bc6b8a6a : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0cd4480e6bc40348e4d7bb08a424cc7 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4c029fc088848c0ba44fa0e65fad199 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc6b505b0f5e456fa86abd6bf3c8a811 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70c78ead1d8a42f5afc32196be94e505 : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47491b6e1a7d4b40926b9ec4aefcb83c : SUCCESS in 5m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/313b02d1538d40479480050d8bc0f038 : SUCCESS in 18m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/88a5f948f02d41f8a97809c4ab235784 : SUCCESS in 42m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7a50ab3bf327475281e25c843119ed9b : SUCCESS in 56m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1bbd4ddcabd94b83aa990647fb7c9bbc : SUCCESS in 53m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4596795ae3e2471cac76c95717f37df0 : SUCCESS in 1h 04m 38s","accounts_in_message":[],"_revision_number":12},{"id":"738c14f5d505716333d1d26b35dd42bb0baf0e80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 14:34:22.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/03066ab561d44c70af282693cdd7f82a : SUCCESS in 1h 27m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9983d3ca1d5049139adc5c624c3f6a60 : SUCCESS in 1h 24m 19s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f02d17600af54a129732569a3336289c : SUCCESS in 1h 35m 08s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fee24f7dd523484ba6400d23919a3df9 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/167e1b6ba27b4c98b8c36a8606b2d257 : SUCCESS in 7m 23s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/80f50df5dfe94710bb675e3494cba384 : SUCCESS in 40m 16s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"9b0192872de5e62795eb8e59fae0fd8c61568891","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-11-04 19:06:06.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"dfb8ed25613fe668066daaa81c23ac512e801eba","author":{"_account_id":29880,"name":"Adrian Andreias","email":"adrian@fleio.com","username":"adrian-fleio"},"date":"2021-11-09 23:01:44.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"5c462a09fb6139124eeda5facf8a2220593ff75d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-16 09:31:03.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"d2ff72175221d60ab95713c26c45efe2ec784e11","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2021-11-22 08:22:12.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"5ddaac134435a6238845f6d33c453f0e04593cc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 08:22:40.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"bd7c389c31c3a2be99ac4b0f22c888cbb5d080ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 09:57:06.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/dacbcdb1e9034942b84a642b0a471a97 : SUCCESS in 39m 45s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/754244152e58434a9380f506e5129ae8 : SUCCESS in 51m 49s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/be9c1f9262f54e35b1ea484691c5528a : SUCCESS in 1h 20m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f949995ca33544daa9ecdfb95d7da55e : SUCCESS in 31m 51s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8ba0652a0d034f68b50ce674fd714f27 : SUCCESS in 55m 00s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/57f5929c84e140deafafabc4106af2d4 : SUCCESS in 1h 27m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cae46fa73bb44867aa92eb1e7c44e207 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbdccc16c3fd4c33bc8b474a083fa7ad : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f9db3c76d4e4745a23adedec335e12d : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/635e821042e94c13908ffa90e543e2cc : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a421242345c54202ad2d20e78c7b050b : SUCCESS in 4m 46s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/713bcbddcfae416c8982778b5a4a33f7 : SUCCESS in 26m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d4daa5ff739d46d7ada24a21079cb200 : SUCCESS in 51m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a79acc5f6a0c4bba9379c886e7b87f1f : SUCCESS in 1h 12m 53s","accounts_in_message":[],"_revision_number":12},{"id":"d66f3f7860b2c510d0c0615b900686553a5f27fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 09:57:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"ef50f01f142f114d2f7137ab111baa4f42b986f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 09:58:44.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cafd25171e843889e22021ec4935e51 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82365a836cd8430a860753c83f6d37df : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"b8b181a0603ce799fa447c713afb0dfedfc3592a","revisions":{"6eab2c198e6ef246a5375cb3cd198035d42c1841":{"kind":"REWORK","_number":1,"created":"2021-09-25 19:47:31.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/15/811015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/811015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/811015/1"}}},"commit":{"parents":[{"commit":"404d076ec056a0b0d07434b8ed91f945ca2e6ec7","subject":"Move all distros to upstream MariaDB 10.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/404d076ec056a0b0d07434b8ed91f945ca2e6ec7"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-25 19:46:55.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-25 19:46:55.000000000","tz":0},"subject":"[DNM] MariaDB 10.6","message":"[DNM] MariaDB 10.6\n\nTo test compat early.\nWe will likely switch in Yoga.\n\nChange-Id: Ibcfbaa1903c7059e437bb2118bc4c9de4a9b854e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6eab2c198e6ef246a5375cb3cd198035d42c1841"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6eab2c198e6ef246a5375cb3cd198035d42c1841"}]},"branch":"refs/heads/master"},"14a169e831cdbeaa8e23b5d82a0d8fb1eb4bfdaf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-09-26 08:28:37.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/15/811015/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/811015/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/811015/2"}}},"commit":{"parents":[{"commit":"efba4db86dfba79808934447925b35ea7e1496df","subject":"Move all distros to upstream MariaDB 10.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efba4db86dfba79808934447925b35ea7e1496df"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-25 19:46:55.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-26 08:28:37.000000000","tz":0},"subject":"[DNM] MariaDB 10.6","message":"[DNM] MariaDB 10.6\n\nTo test compat early.\nWe will likely switch in Yoga.\n\nChange-Id: Ibcfbaa1903c7059e437bb2118bc4c9de4a9b854e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14a169e831cdbeaa8e23b5d82a0d8fb1eb4bfdaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14a169e831cdbeaa8e23b5d82a0d8fb1eb4bfdaf"}]},"branch":"refs/heads/master","description":"Rebase"},"7d8bb72c3fa14389b9a779c27cafc0e8882c2f89":{"kind":"NO_CHANGE","_number":3,"created":"2021-09-26 13:43:55.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/15/811015/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/811015/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/811015/3"}}},"commit":{"parents":[{"commit":"205f29a84319f3c34eefb23401f6a7ebdc8b8379","subject":"Move all distros to upstream MariaDB 10.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/205f29a84319f3c34eefb23401f6a7ebdc8b8379"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-25 19:46:55.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-26 13:43:55.000000000","tz":0},"subject":"[DNM] MariaDB 10.6","message":"[DNM] MariaDB 10.6\n\nTo test compat early.\nWe will likely switch in Yoga.\n\nChange-Id: Ibcfbaa1903c7059e437bb2118bc4c9de4a9b854e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d8bb72c3fa14389b9a779c27cafc0e8882c2f89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d8bb72c3fa14389b9a779c27cafc0e8882c2f89"}]},"branch":"refs/heads/master","description":"Rebase"},"95e0842b9989c2fa48f5cfc2849a09777a1ee00f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-09-26 16:39:21.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/15/811015/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/811015/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/811015/4"}}},"commit":{"parents":[{"commit":"b11b9e2e3d5c66ab546fe318d40d1f1bd34b9b9d","subject":"Move all distros to upstream MariaDB 10.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b11b9e2e3d5c66ab546fe318d40d1f1bd34b9b9d"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-25 19:46:55.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-26 16:39:21.000000000","tz":0},"subject":"[DNM] MariaDB 10.6","message":"[DNM] MariaDB 10.6\n\nTo test compat early.\nWe will likely switch in Yoga.\n\nChange-Id: Ibcfbaa1903c7059e437bb2118bc4c9de4a9b854e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95e0842b9989c2fa48f5cfc2849a09777a1ee00f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95e0842b9989c2fa48f5cfc2849a09777a1ee00f"}]},"branch":"refs/heads/master","description":"Rebase"},"e7ef51ab721df20a5380206c32455978a227ffab":{"kind":"NO_CHANGE","_number":5,"created":"2021-09-26 16:39:57.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/15/811015/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/811015/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/811015/5"}}},"commit":{"parents":[{"commit":"fc6c1a13b556bad685ec6f874dc6fae0a8538e51","subject":"Move all distros to upstream MariaDB 10.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc6c1a13b556bad685ec6f874dc6fae0a8538e51"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-25 19:46:55.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-26 16:39:57.000000000","tz":0},"subject":"[DNM] MariaDB 10.6","message":"[DNM] MariaDB 10.6\n\nTo test compat early.\nWe will likely switch in Yoga.\n\nChange-Id: Ibcfbaa1903c7059e437bb2118bc4c9de4a9b854e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7ef51ab721df20a5380206c32455978a227ffab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7ef51ab721df20a5380206c32455978a227ffab"}]},"branch":"refs/heads/master","description":"Rebase"},"0952aa1b369c4c2ed053038c7c81f92f7b935260":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-09-28 08:54:29.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/15/811015/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/811015/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/811015/6"}}},"commit":{"parents":[{"commit":"04b812b4003fdd9359f57ff7cf4f9a0c6efd647b","subject":"Move all distros to upstream MariaDB 10.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/04b812b4003fdd9359f57ff7cf4f9a0c6efd647b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-25 19:46:55.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-28 08:54:29.000000000","tz":0},"subject":"[DNM] MariaDB 10.6","message":"[DNM] MariaDB 10.6\n\nTo test compat early.\nWe will likely switch in Yoga.\n\nChange-Id: Ibcfbaa1903c7059e437bb2118bc4c9de4a9b854e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0952aa1b369c4c2ed053038c7c81f92f7b935260"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0952aa1b369c4c2ed053038c7c81f92f7b935260"}]},"branch":"refs/heads/master","description":"Rebase"},"a6322f44b3e36f298d095c47d737be87c1750e47":{"kind":"NO_CHANGE","_number":7,"created":"2021-09-28 11:41:25.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/15/811015/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/811015/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/811015/7"}}},"commit":{"parents":[{"commit":"97e356eda7409d36c4c20e79fba6d67dd3b7f1df","subject":"Move all distros to upstream MariaDB 10.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97e356eda7409d36c4c20e79fba6d67dd3b7f1df"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-25 19:46:55.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-28 11:41:25.000000000","tz":0},"subject":"[DNM] MariaDB 10.6","message":"[DNM] MariaDB 10.6\n\nTo test compat early.\nWe will likely switch in Yoga.\n\nChange-Id: Ibcfbaa1903c7059e437bb2118bc4c9de4a9b854e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6322f44b3e36f298d095c47d737be87c1750e47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6322f44b3e36f298d095c47d737be87c1750e47"}]},"branch":"refs/heads/master","description":"Rebase"},"7c53f6721948da1e5d9fa7b2d35f04154fbc3ad1":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-09-28 17:40:39.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/15/811015/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/811015/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/811015/8"}}},"commit":{"parents":[{"commit":"456397107362af71c722abf731e057ff637e608b","subject":"Move all distros to upstream MariaDB 10.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/456397107362af71c722abf731e057ff637e608b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-25 19:46:55.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-28 17:40:39.000000000","tz":0},"subject":"[DNM] MariaDB 10.6","message":"[DNM] MariaDB 10.6\n\nTo test compat early.\nWe will likely switch in Yoga.\n\nChange-Id: Ibcfbaa1903c7059e437bb2118bc4c9de4a9b854e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c53f6721948da1e5d9fa7b2d35f04154fbc3ad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c53f6721948da1e5d9fa7b2d35f04154fbc3ad1"}]},"branch":"refs/heads/master","description":"Rebase"},"ac5771b5cac98d265b68187f08b166cb1e6fa3fb":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-10-08 15:13:55.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/15/811015/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/811015/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/811015/9"}}},"commit":{"parents":[{"commit":"7652ef325e0e84fe92ccf104bfe751a4d3542227","subject":"Merge \"[bifrost] Drop workarounds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7652ef325e0e84fe92ccf104bfe751a4d3542227"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-25 19:46:55.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-08 15:13:55.000000000","tz":0},"subject":"[DNM] MariaDB 10.6","message":"[DNM] MariaDB 10.6\n\nTo test compat early.\nWe will likely switch in Yoga.\n\nChange-Id: Ibcfbaa1903c7059e437bb2118bc4c9de4a9b854e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac5771b5cac98d265b68187f08b166cb1e6fa3fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac5771b5cac98d265b68187f08b166cb1e6fa3fb"}]},"branch":"refs/heads/master","description":"Rebase"},"0f1294f33234f9328380c225ca93b90622fb8f28":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-10-17 07:06:41.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/15/811015/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/811015/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/811015/10"}}},"commit":{"parents":[{"commit":"6a3b1609c1527f7978d2e5545c9853d16038f1aa","subject":"Merge \"docs: Move Kayobe RC1 after Kolla \u0026 Kolla Ansible RC1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a3b1609c1527f7978d2e5545c9853d16038f1aa"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-25 19:46:55.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-17 07:06:41.000000000","tz":0},"subject":"[DNM] MariaDB 10.6","message":"[DNM] MariaDB 10.6\n\nTo test compat early.\nWe will likely switch in Yoga.\n\nChange-Id: Ibcfbaa1903c7059e437bb2118bc4c9de4a9b854e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f1294f33234f9328380c225ca93b90622fb8f28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f1294f33234f9328380c225ca93b90622fb8f28"}]},"branch":"refs/heads/master","description":"Rebase"},"1a883ca7973674c37c3c129d72a4e6d7344adbe2":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2021-10-17 07:07:34.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/15/811015/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/811015/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/811015/11"}}},"commit":{"parents":[{"commit":"6a3b1609c1527f7978d2e5545c9853d16038f1aa","subject":"Merge \"docs: Move Kayobe RC1 after Kolla \u0026 Kolla Ansible RC1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a3b1609c1527f7978d2e5545c9853d16038f1aa"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-25 19:46:55.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-17 07:07:34.000000000","tz":0},"subject":"[WIP] MariaDB 10.6","message":"[WIP] MariaDB 10.6\n\nTo test compat early.\nWe will likely switch in Yoga.\n\nChange-Id: Ibcfbaa1903c7059e437bb2118bc4c9de4a9b854e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a883ca7973674c37c3c129d72a4e6d7344adbe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a883ca7973674c37c3c129d72a4e6d7344adbe2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b8b181a0603ce799fa447c713afb0dfedfc3592a":{"kind":"REWORK","_number":12,"created":"2021-11-04 12:53:24.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/15/811015/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/811015/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/811015/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/811015/12"}}},"commit":{"parents":[{"commit":"ecb88c5d0c80d58708c5d391826c5da7d873eb6b","subject":"Merge \"Remove InfluxDB data source Grafana plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ecb88c5d0c80d58708c5d391826c5da7d873eb6b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-25 19:46:55.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-11-04 12:47:56.000000000","tz":0},"subject":"Use MariaDB 10.6","message":"Use MariaDB 10.6\n\nAs planned for Yoga.\n\nChange-Id: Ibcfbaa1903c7059e437bb2118bc4c9de4a9b854e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8b181a0603ce799fa447c713afb0dfedfc3592a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8b181a0603ce799fa447c713afb0dfedfc3592a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
