)]}'
{"id":"openstack%2Fkolla~811743","triplet_id":"openstack%2Fkolla~master~Ie10a41a94af95075a1b91feb2b74ade02b17117e","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ie10a41a94af95075a1b91feb2b74ade02b17117e","subject":"Build source type images by default","status":"MERGED","created":"2021-09-29 15:41:58.000000000","updated":"2021-09-30 11:04:12.000000000","submitted":"2021-09-30 11:02:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"811743","meta_rev_id":"37a812767594dcea95b92b02245f2fd2d2ef6290","_number":811743,"virtual_id_number":811743,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-30 11:02:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-09-30 08:40:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-09-30 09:40:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-09-30 09:40:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-29 15:43:40.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-09-29 15:43:44.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-09-29 17:19:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-30 08:40:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-09-30 09:40:02.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"4953e17133a603070196012dc7f83941806a9b66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-29 15:41:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a12d5c1eeeada5b308165a37751a7398bd9eec67","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-29 15:43:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c16a3225fbb932bb8d14c1856bf444a4e81a99e8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-09-29 15:43:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"23be4486dbc8f617f949f63b24fb767e95500f92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 17:19:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e2028ed6d9034cd1a7c188d14feac1f5 : SUCCESS in 24m 52s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/780797ccc1c0499183b13aeb6fb891a5 : SUCCESS in 43m 55s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d2cb8543b8de42f4a1c4cbcd0835d74f : SUCCESS in 1h 18m 40s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f0f143637578478dab1acd22bdd3f004 : SUCCESS in 1h 13m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5a0f88dce61b4161ab176bef23c60a7e : SUCCESS in 1h 36m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2383f1440f2f496aa6ebfc5f3c3e0934 : SUCCESS in 29m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d5a1853116d04a8da4dfca84fc99b17e : SUCCESS in 20m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0652ed5ad68e42c7838c508a8081804d : SUCCESS in 1h 01m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b41eef1eeb494543bfaa289debb1775d : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5556fb057c147f2875e8e4ac4b43ada : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7066c0315d140fdbae6f936c6dbec71 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdfb212ca621415bbc2b08f23fde213b : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0344c24ad51642a3ba22684de55d413e : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8dbdccaacda4f80be380b91850417b9 : SUCCESS in 4m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cc855fb4aaf04654a39a4c77ec19a9c9 : SUCCESS in 19m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d25efa180880406f98bcb169786533af : SUCCESS in 25m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/45b667dfb9fa483fa25c61031dd4c64e : SUCCESS in 50m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/72cea8dd22e94c13b57107a5a700fa25 : SUCCESS in 54m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4bcf0a7c5c0d475d90c21e3881049dc5 : SUCCESS in 1h 12m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a508b0731e17afc970e9659d6892746371eba0e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 17:54:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a1590f62e6e14464bf4df40b13709e94 : SUCCESS in 1h 58m 50s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9ecfdcd4d7944bcea6fd49a157178bd2 : SUCCESS in 1h 35m 27s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/738065d790b84adc9f3a691b4df527c1 : SUCCESS in 2h 10m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c7b4592b4ca4834ae83aecff8b222cb : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/895e51ff80d342d69a715670922e2f98 : SUCCESS in 6m 55s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/166189b508734f1d81d0681c47eb5cef : SUCCESS in 1h 25m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96bbf4c0b5b276482f47a6c00789d8c98af860d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-30 08:07:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"991f69b9b58fdf80f5bf26a02acd8ab32b3372b0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-30 08:40:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"aabfbc0e716bfa0d81f8c7e32da373e8b0567a12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 09:33:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f5dab250fb334a8c80e7b255ff22013c : SUCCESS in 28m 32s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/378acd1f3bca4083952d44c8d02be653 : SUCCESS in 36m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f1a4e6b20687490ca10e5cae2e993d9e : SUCCESS in 1h 17m 43s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7d8fdce744f5459489e8c930127544c0 : SUCCESS in 1h 06m 45s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/57ed71c469604405a6b161f09e08e63c : SUCCESS in 1h 21m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3321808dc8994830a94b219d1090d9a3 : SUCCESS in 31m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/57c08cb334c245af81b51c1cf8a1e254 : SUCCESS in 20m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/93544e10bcc5454188188fd2b7acf510 : SUCCESS in 56m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e9d344e663a4b72ad33830e29232002 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f72351c360a84646af9008ef6c855e5f : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba3a47724fc244f392f641f8b2b7ba56 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/647c827a9059449786847af44cc96ad8 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8f61b1b29b54f32b95b8859cfa140e3 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c2d90d381104c949b1d1f1c8468e3aa : SUCCESS in 4m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0086328ea84e4cdc8ad455d33eebd920 : SUCCESS in 19m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b8433acae6734afb81f8b05ca08349a2 : SUCCESS in 26m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/754ddf4c3ec14b13b0c48a6c03683862 : SUCCESS in 49m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0f979b9b363840048772a7ea67ae5b00 : SUCCESS in 47m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4c6fa9bcf8f0410ba3d4aab0fb567e72 : SUCCESS in 1h 07m 30s","accounts_in_message":[],"_revision_number":2},{"id":"a2f368fdfc43e1c798c67fcb3223be4129cf7e90","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-30 09:40:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nlovely, thanks","accounts_in_message":[],"_revision_number":2},{"id":"9a73f04be410d2dc0e0c455de7fd33bf22a31df8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 09:40:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5d2c702e8a08d66ad72d006816ba7888f1098213","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 10:38:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/53b536e2158b4aa8b3408c9a3ab52542 : SUCCESS in 54m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2169936069ab4e2fbab8996175c921c1 : SUCCESS in 1h 30m 58s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f6f45ab0fc714210a62f2ab733ce551c : SUCCESS in 2h 27m 56s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/485c0ee7d0cb402b88ff09288226765a : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/616fba77709f4d9d9c296186070d6271 : SUCCESS in 6m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9ca9b8cd886a4d3fbae6933eb774ed27 : SUCCESS in 1h 55m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6f25977e9a8b599a7301a16a88140dbfd2d24cd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 11:02:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7fc44d0beafd4e89a9d3c27e0eb1d112 : SUCCESS in 31m 48s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9a01082bdb664f1eb4ad4fcf7a6be648 : SUCCESS in 1h 03m 54s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9bbef7fa07264d71b474564be978f36e : SUCCESS in 1h 18m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18d6f83d55c54b4ba22d484351bbb777 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a81a78828c24985815c8d76dbe6d76a : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5cb5e1bf67f428ba86ebf700cce530a : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b8b41ad9ce1471797ee3fbf2ab3d233 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f956bfa8c4745c0aef4e7a0a42c7251 : SUCCESS in 4m 59s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d8ffcd15fd0d4a329285ec8ba33b101f : SUCCESS in 33m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/df701981c6864d75810201b44a54ec0f : SUCCESS in 49m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fdac43ec63c54058ae81cc36c78ec7b5 : SUCCESS in 1h 10m 36s","accounts_in_message":[],"_revision_number":2},{"id":"aa2be51a8afeee8998b76e819e273d3198ab8959","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 11:02:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"37a812767594dcea95b92b02245f2fd2d2ef6290","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 11:04:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aab3e5642eee485ea254ff4c895c2671 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d227c8fb26094576b02373a3edf4ce92 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e4f019a5ddfca68c7f0ac5cbfdeb47e88e170d20","revisions":{"2289ba87ed9ba7d4a84c1b3145b385855703192a":{"kind":"REWORK","_number":1,"created":"2021-09-29 15:41:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/43/811743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/811743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/811743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/811743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/811743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/811743/1"}}},"commit":{"parents":[{"commit":"e6c124f6c29e47e3365c0f395ea5618dcbb78280","subject":"Merge \"Debian: use Xena binary packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6c124f6c29e47e3365c0f395ea5618dcbb78280"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-29 15:39:40.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-29 15:40:08.000000000","tz":60},"subject":"Build source type images by default","message":"Build source type images by default\n\nSource images get the most test coverage, so it makes sense to build\nthese by default.\n\nChange-Id: Ie10a41a94af95075a1b91feb2b74ade02b17117e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2289ba87ed9ba7d4a84c1b3145b385855703192a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2289ba87ed9ba7d4a84c1b3145b385855703192a"}]},"branch":"refs/heads/master"},"e4f019a5ddfca68c7f0ac5cbfdeb47e88e170d20":{"kind":"REWORK","_number":2,"created":"2021-09-30 08:07:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/43/811743/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/811743/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/811743/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/811743/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/811743/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/811743/2"}}},"commit":{"parents":[{"commit":"e6c124f6c29e47e3365c0f395ea5618dcbb78280","subject":"Merge \"Debian: use Xena binary packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6c124f6c29e47e3365c0f395ea5618dcbb78280"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-29 15:39:40.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-30 08:07:10.000000000","tz":60},"subject":"Build source type images by default","message":"Build source type images by default\n\nSource images get the most test coverage, so it makes sense to build\nthese by default.\n\nChange-Id: Ie10a41a94af95075a1b91feb2b74ade02b17117e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4f019a5ddfca68c7f0ac5cbfdeb47e88e170d20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4f019a5ddfca68c7f0ac5cbfdeb47e88e170d20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
