)]}'
{"id":"openstack%2Fkolla~811977","triplet_id":"openstack%2Fkolla~master~I4ec4eb837c98f57964488769ab82ba8213ad582c","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I4ec4eb837c98f57964488769ab82ba8213ad582c","subject":"logstash: pin to 7.9.* for Elasticsearch OSS compatibility","status":"MERGED","created":"2021-09-30 14:12:15.000000000","updated":"2021-10-01 13:44:47.000000000","submitted":"2021-10-01 13:42:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"811977","meta_rev_id":"998d7a6a5be991c3faad5c24f3f01d04e00e5f3e","_number":811977,"virtual_id_number":811977,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-01 13:42:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-10-01 04:49:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-10-01 12:02:56.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-10-01 12:02:56.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-30 14:33:25.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2021-09-30 15:38:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-30 16:05:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2021-09-30 16:39:36.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-10-01 04:49:27.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"4edb7b6e3b17ffe4a0ddd3c28be51fac3c79204b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-30 14:12:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5021803652ea18967dffbcf7494084f332f061b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 15:38:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/13b75eedc92c4d1fb348696e95d5cbec : SUCCESS in 28m 42s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8ae808b11786410b9c78dded569f9278 : SUCCESS in 32m 51s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d404f36257e947f6884b11cfa105b368 : SUCCESS in 1h 09m 51s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/5f4dddae0e9f419a8fd8b3b29ac9b531 : SUCCESS in 1h 04m 51s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/14cdda9099e9403bba947fc07810d619 : SUCCESS in 1h 21m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/879c665d9afe47bdabd8f9f7f95b6042 : FAILURE in 32m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/16169146d1a44acf8454886a0d068151 : FAILURE in 22m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4a98186971e642eb988eaf2a14153c5c : SUCCESS in 59m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f05206ace7cf45d1ac85a690a04ec2cc : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21fccb9500f049ca81f38080e35baacd : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e64ce271c86e4a85b4a8c7c9a3153aa4 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c716ff39da2945348ea8d5a4719db68c : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c783a51302d46a08b05bbb0d3fb85a5 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ed6c9fc163a4ada959e74cc94863c8d : SUCCESS in 4m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0217dcaeaf3748ca9667a9ad14abe67c : FAILURE in 18m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8cf4b7b226b94fd8a097865e6f82f647 : FAILURE in 31m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d903afd214084d0b8e40f36aa94c879d : SUCCESS in 56m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3631094c4dbd40a9b953d389facf2899 : SUCCESS in 48m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9874fecf9e2140cc9d348ddfd86cd483 : SUCCESS in 1h 10m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a0496c528da3a69b6641767d7f58bd5cfeb0ee5e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-30 16:05:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aa8dc8d6b75aeb8b19c9fa2dffb705c502fadac1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 16:12:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b34d1dba522841d9a0568c7e739615f2 : SUCCESS in 1h 55m 19s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c96b8153748c4022b987d150e3d99985 : FAILURE in 1h 27m 11s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0483896ffb204fa6884d2bbd51d9acce : SUCCESS in 1h 19m 00s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/069f889d7aee44ba9f8f8fc445fb3a15 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/11fdda8ae9b5401e8d809528b5939314 : SUCCESS in 7m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4691c3dfb12e4f7eba804556d42c35e9 : FAILURE in 38m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4446ff6fc1b3694d88611a82fea3d6dfd4f04746","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-30 16:13:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6d64bb5500834b8ed9d140a21edecdfbce9cab28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-30 16:16:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4d970ec0deeca08b63d67a245c839a7dafe38b6c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-30 16:39:36.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"21175318fa2b0229fac819e395e0db6f40a2f9ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 18:44:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/da4fe73d91dd474e9ffc7e263194d401 : SUCCESS in 26m 44s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f7d11f0d61144f56b9a34d7c7a9dc78a : SUCCESS in 39m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/aa22c1195c0a4dffa29ca224c248d8e6 : SUCCESS in 1h 15m 20s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/57f0b27a7cd1474196e99169b6a5b5bb : SUCCESS in 1h 07m 11s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e099e9baef284970b4bba5486cf41809 : SUCCESS in 1h 28m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9b9ecd29d7e1445ba3435335d6ba3f7a : SUCCESS in 34m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2551b57167fc44b0852bb34ae92f2cc7 : SUCCESS in 20m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/197b63055d504035b5c5c12c196c3353 : SUCCESS in 51m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2d74da7d4c94c138da720328d28f231 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bde27490069d418a865987b22557f41b : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c9bb4043fcb4bfbbfd878f63e4fb0e2 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd9690faf4e8473e97574c9f402140da : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0461e665da4f4f1e8ec47612cffd9271 : SUCCESS in 8m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fd80c4a9008402e8a4c1c46077cdc5c : SUCCESS in 5m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ae9b7551d52d4873bd96d0426a866b09 : SUCCESS in 20m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1696c4802a8543b484470a43e6d66525 : SUCCESS in 33m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/31a390ff48734649873114ba50254e35 : SUCCESS in 54m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/13a8000d17f540efa153be8baf061ad5 : SUCCESS in 54m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1929a745549848789f2dc802e2f9cbfe : SUCCESS in 1h 01m 46s","accounts_in_message":[],"_revision_number":3},{"id":"28c734e654b9f1394af1dd615d62f262ea4c02d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 19:12:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b2ed7b6d7af44b89b9dae36b1ae85fc7 : SUCCESS in 39m 36s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fce23312dfd74983af230cd0c7c7d969 : SUCCESS in 43m 26s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/80f7715842a04211b8cd4f9829e3650c : SUCCESS in 1h 45m 12s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf3ec3adbce34687adde5f29e343099d : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d72835aa24c44b9abadfdf86887bb55 : SUCCESS in 6m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/43fa0f4870d34cbda3bedce9274ecf0e : SUCCESS in 59m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"66f44837b12696f95d71ad8b5f44256c87604424","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-30 19:22:21.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ad737c03f7fa1daf9a2da35f024eb2816e48cab4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-30 19:22:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6e297445773f02c37c53f053cbed14a298f88862","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-30 19:23:12.000000000","message":"Patch Set 3:\n\nI think I was trying to leave for the day when I pushed the last one :)","accounts_in_message":[],"_revision_number":3},{"id":"d02150cf30f20e7e10c7bb7d26ee7708195b3c98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 20:48:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6fc8f0ef78c645819e34bd640f265fa1 : SUCCESS in 22m 59s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/719363f93a214e3cad9c66f2b8933a0d : SUCCESS in 36m 48s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/11ea6b65c84d48dca39dc30da439558f : SUCCESS in 1h 02m 15s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/cb33500c96754f369bb75100008a6ab2 : SUCCESS in 1h 03m 14s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a06be37b5f5241e88a31293edde9967c : SUCCESS in 1h 24m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fbb32c20a5114544b8539ac5cc964627 : SUCCESS in 28m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/22589edc89954f47be1d59991569af0b : SUCCESS in 15m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4fe2e5874b3c4f25a1e36630920bbe2c : SUCCESS in 1h 03m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/097cc1afbc1b400eaf1554ab9d677264 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b203bc611184353a917d749f175ab0d : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5d9f7945ae34f5e9f724bf165a09687 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09dc34169be54c57bc1a01e41f5c2dc3 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e3c5c81562d4038a86947c833fef4da : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/266acdd7262c403895f64e5deecf185b : SUCCESS in 4m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/90a04530a2ad40b99015132ad336c7a6 : SUCCESS in 20m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8232d3463f6a41a8b6b570bbaaec3556 : SUCCESS in 30m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7465d764311648e489dd495b47a6abbc : SUCCESS in 53m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/69251f1a8fb74e82bd0cfefd189cf98a : SUCCESS in 51m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/57684ab0ee2b4f83b044df8f5291ba14 : SUCCESS in 59m 28s","accounts_in_message":[],"_revision_number":4},{"id":"a2284679e7a11dadbefea5185973456a3034dff1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 20:53:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7cede5c77f534a6b8dc3b4368b50a2b5 : SUCCESS in 38m 54s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/051bac9dd4b14ec59e2971247491cc62 : SUCCESS in 1h 17m 12s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/07745f3e80774bf7bf01a03212b3eeae : SUCCESS in 1h 19m 41s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e4c25a98dad41bb9d96d3c2a005caa7 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/609cde941fba45febd5e222f3580d3a2 : SUCCESS in 5m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/81cf4d8b22f5444dbb8b41e1bc002e8a : SUCCESS in 1h 05m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dab34d60093638421febc60caa46f442dc99ffea","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-01 04:49:27.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"29610c35a9134321a80b71133ef295f1c2172825","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-01 12:02:56.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)\n\n\u003e Patch Set 3:\n\u003e \n\u003e I think I was trying to leave for the day when I pushed the last one :)\n\n😂","accounts_in_message":[],"_revision_number":4},{"id":"7cedb31851b4e35530609eabb457059cef8cf9a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 12:03:19.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"45705a6ef9e92b7936c26b1d7e5856d3f2c92118","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 13:42:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3d556812254d43619d0cde3bee1f6f53 : SUCCESS in 1h 04m 55s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/10e7b0c46f9f4a64a73e355cbbace51d : SUCCESS in 1h 12m 39s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/78cee9791b4441149566ec402b953cf9 : SUCCESS in 1h 33m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28a148a73fa7475caf0996dee7d0eb33 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42b3f40702fc413dbfcba426513f9be3 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53fd9161a1884e7fad31847afe6c3c3a : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cfbfccf49714f339f6f0519688b063f : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ed06d198bce4bad83551a530771b376 : SUCCESS in 4m 48s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e853ee2c79b647969bfe4b52975c8f87 : SUCCESS in 32m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8fd7567a514d4845a4520884ef53d5bd : SUCCESS in 1h 20m 29s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1a40635f96dd4de7820da0b5894ff805 : SUCCESS in 1h 28m 32s","accounts_in_message":[],"_revision_number":4},{"id":"43b970f0a81b17a8d4765363429fedbb53532082","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 13:42:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"998d7a6a5be991c3faad5c24f3f01d04e00e5f3e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 13:44:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/856c8668abb246e3a932b7385ad283d2 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac7fc9f8c92948339081539d6bffae8d : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b32cf27cc93878bcc124bcc387749829b0997a66","revisions":{"a545186e9058050e0638c294adc150546fdc62df":{"kind":"REWORK","_number":1,"created":"2021-09-30 14:12:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/811977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/811977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/811977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/811977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/811977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/811977/1"}}},"commit":{"parents":[{"commit":"e4f019a5ddfca68c7f0ac5cbfdeb47e88e170d20","subject":"Build source type images by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4f019a5ddfca68c7f0ac5cbfdeb47e88e170d20"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-30 12:42:18.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-30 12:43:00.000000000","tz":60},"subject":"logstash: pin to 7.9.* for Elasticsearch OSS compatibility","message":"logstash: pin to 7.9.* for Elasticsearch OSS compatibility\n\nThe latest versions of logstash 7.x to not support using the OSS version\nof Elasticsearch, and fail to push logs if an OSS version is used. This\nprevents Monasca log API from working correctly.\n\nThis change fixes the issue by pinning logstash-oss to 7.9.x, the most\nrecent available package in the repo that still supports Elasticsearch\nOSS.\n\nNOTE: logstash 7.11 breaking introduced the change.\n\nChange-Id: I4ec4eb837c98f57964488769ab82ba8213ad582c\nCloses-Bug: #1941754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a545186e9058050e0638c294adc150546fdc62df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a545186e9058050e0638c294adc150546fdc62df"}]},"branch":"refs/heads/master"},"84a9bc3f06cb784f30bf60fa9513c45071058901":{"kind":"REWORK","_number":2,"created":"2021-09-30 16:13:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/811977/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/811977/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/811977/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/811977/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/811977/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/811977/2"}}},"commit":{"parents":[{"commit":"e4f019a5ddfca68c7f0ac5cbfdeb47e88e170d20","subject":"Build source type images by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4f019a5ddfca68c7f0ac5cbfdeb47e88e170d20"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-30 12:42:18.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-30 16:13:44.000000000","tz":60},"subject":"logstash: pin to 7.9.* for Elasticsearch OSS compatibility","message":"logstash: pin to 7.9.* for Elasticsearch OSS compatibility\n\nThe latest versions of logstash 7.x to not support using the OSS version\nof Elasticsearch, and fail to push logs if an OSS version is used. This\nprevents Monasca log API from working correctly.\n\nThis change fixes the issue by pinning logstash-oss to 7.9.x, the most\nrecent available package in the repo that still supports Elasticsearch\nOSS.\n\nNOTE: logstash 7.11 breaking introduced the change.\n\nChange-Id: I4ec4eb837c98f57964488769ab82ba8213ad582c\nCloses-Bug: #1941754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84a9bc3f06cb784f30bf60fa9513c45071058901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84a9bc3f06cb784f30bf60fa9513c45071058901"}]},"branch":"refs/heads/master"},"af4ba0b07710b2ba04831f16b954954727a8b39a":{"kind":"REWORK","_number":3,"created":"2021-09-30 16:16:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/811977/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/811977/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/811977/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/811977/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/811977/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/811977/3"}}},"commit":{"parents":[{"commit":"e4f019a5ddfca68c7f0ac5cbfdeb47e88e170d20","subject":"Build source type images by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4f019a5ddfca68c7f0ac5cbfdeb47e88e170d20"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-30 12:42:18.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-30 16:15:57.000000000","tz":60},"subject":"logstash: pin to 7.9.* for Elasticsearch OSS compatibility","message":"logstash: pin to 7.9.* for Elasticsearch OSS compatibility\n\nThe latest versions of logstash 7.x to not support using the OSS version\nof Elasticsearch, and fail to push logs if an OSS version is used. This\nprevents Monasca log API from working correctly.\n\nThis change fixes the issue by pinning logstash-oss to 7.9.x, the most\nrecent available package in the repo that still supports Elasticsearch\nOSS.\n\nNOTE: logstash 7.11 breaking introduced the change.\n\nChange-Id: I4ec4eb837c98f57964488769ab82ba8213ad582c\nCloses-Bug: #1941754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af4ba0b07710b2ba04831f16b954954727a8b39a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af4ba0b07710b2ba04831f16b954954727a8b39a"}]},"branch":"refs/heads/master"},"b32cf27cc93878bcc124bcc387749829b0997a66":{"kind":"REWORK","_number":4,"created":"2021-09-30 19:22:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/811977/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/811977/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/811977/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/811977/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/811977/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/811977/4"}}},"commit":{"parents":[{"commit":"e4f019a5ddfca68c7f0ac5cbfdeb47e88e170d20","subject":"Build source type images by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4f019a5ddfca68c7f0ac5cbfdeb47e88e170d20"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-30 12:42:18.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-30 19:22:34.000000000","tz":60},"subject":"logstash: pin to 7.9.* for Elasticsearch OSS compatibility","message":"logstash: pin to 7.9.* for Elasticsearch OSS compatibility\n\nThe latest versions of logstash 7.x do not support using the OSS version\nof Elasticsearch, and fail to push logs if an OSS version is used. This\nprevents Monasca log API from working correctly.\n\nThis change fixes the issue by pinning logstash-oss to 7.9.x, the most\nrecent available package in the repo that still supports Elasticsearch\nOSS.\n\nNOTE: logstash 7.11 introduced the breaking change, however 7.10 is\nunavailable in the package repos.\n\nChange-Id: I4ec4eb837c98f57964488769ab82ba8213ad582c\nCloses-Bug: #1941754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b32cf27cc93878bcc124bcc387749829b0997a66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b32cf27cc93878bcc124bcc387749829b0997a66"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
