)]}'
{"id":"openstack%2Fkolla~812140","triplet_id":"openstack%2Fkolla~master~Ic90cd8e3065fa629d5ad67abaf7c193fd845259a","project":"openstack/kolla","branch":"master","topic":"neutron-api-paste","hashtags":[],"change_id":"Ic90cd8e3065fa629d5ad67abaf7c193fd845259a","subject":"Remove neutron api-paste.ini from old location","status":"MERGED","created":"2021-10-01 14:17:02.000000000","updated":"2021-10-04 17:11:35.000000000","submitted":"2021-10-04 17:08:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812140-neutron-api-paste","meta_rev_id":"90f5836eb343276ba2f720807eed330a41eb4550","_number":812140,"virtual_id_number":812140,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-04 17:08:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-10-04 11:38:32.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-10-04 15:37:35.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-10-04 15:37:35.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-01 15:56:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-04 08:28:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-04 15:37:35.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"766b762c79e46df45514b71782f6fe5509222771","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-01 14:17:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4c1d0aba426a3a7931e14d8b32d40ec4a7cf362","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-01 14:25:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b1873e092f95503d7944a52dc45641669425c1dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 15:56:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1d2472b65b124598a0584d26e99f39c6 : SUCCESS in 34m 02s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/900889ede33143dc92b64071cb52a0f0 : SUCCESS in 42m 49s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/682ae77f1da7478fba826ee00d7c6509 : SUCCESS in 1h 11m 37s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/3bd99247bc65494291211431310700f7 : SUCCESS in 1h 13m 38s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/aba6c0d649d345b399cf2d269af55c31 : SUCCESS in 1h 26m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7037084aee69462fa8bbf576ca60c076 : SUCCESS in 36m 06s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a8456ec255a14c17a55b3c2ab69f4ad7 : SUCCESS in 19m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/efa40a09e12c4dfcbc0941b44c84080b : SUCCESS in 1h 00m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f7fc249a7ac4506bd28665bfd0b2f54 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fe3848ce0a64a4eb4cdc3a300969541 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56e469fd64574264adfe64e878eb1499 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2000b8b142e64e41a99fc54cf00009e9 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1058207daf5743559a779614eabedf73 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e237a5d9007f47b394701e120f4dd478 : SUCCESS in 5m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1210b2e68b1b409fbf72eacf6eeebf93 : SUCCESS in 32m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9cd8f9e1917f4f85a078491d09f70373 : SUCCESS in 36m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8e2088cc5b364782acb80ec7b5535f05 : SUCCESS in 58m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2df2e2ebcf77455fb09388e81c139c57 : SUCCESS in 54m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/406c922baa2b4b4a9d8a6f722edec53a : SUCCESS in 1h 11m 33s","accounts_in_message":[],"_revision_number":2},{"id":"03f8fb0e2c351b718bb4c70b4a0e68c8862063c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 16:50:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/01abb9f04c4e4f3bb645da80add133c4 : SUCCESS in 2h 19m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6738273c8c4443fd80f33ebe6103775f : SUCCESS in 41m 43s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/275b1e4f3d244723aa4d9a6e33d4f1d1 : SUCCESS in 2h 05m 12s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f1138ea751dc49ab881d1df5d4901ef7 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e17a45ce4e0422e8057810eb03b379e : SUCCESS in 8m 20s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f48d845d123e4cd5a58b8de645578d24 : SUCCESS in 2h 22m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"855f8f81cfc995544db89cbf792d00c7aa0cad90","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-04 08:28:46.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e080306b61cd0c75f4ca91cab609a66c3ae7ae3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-04 09:12:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7b0af5f76a289e5b16029f83a10e4f25db5dc42c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-04 09:12:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"11606463570dfffb353e5fffa99fddb5e033203c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-04 09:12:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4ca9952ae963d3076984674a3fe8a9ad4a6de629","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 10:34:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f6bc93d1dfca4176b76ad7a494bc746b : SUCCESS in 29m 55s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a551921d744549cf9c4d5b1e200ff4dd : SUCCESS in 41m 44s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2ba5f175ef7f4bd59fa64db0dffde8a3 : SUCCESS in 1h 08m 06s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/497bea721699490ba2cca6dc3565ba33 : SUCCESS in 1h 05m 03s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2bdbc9c29de744cf9f8b09047d1df110 : SUCCESS in 1h 20m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ec3e52bce8ac46efad1d1f6b77cef8ee : SUCCESS in 25m 26s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d229cdc51ab64c63b11b4854bfd1ba14 : SUCCESS in 20m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/914a1e24424741b9b4f77f050b09c6b5 : SUCCESS in 1h 01m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a66975d6a0b4195977f1e1217b8c7b7 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d119e1a0f43f4fa0a6f0affddf3c0a25 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a30831fdf6d6486fa24fe5d1cf391edf : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15d8262705644f76a8f5daa6f23aac8f : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/280f6434833441d8b52955093a792441 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf80bc467fc94d378cf4610f8acff6ee : SUCCESS in 4m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b46dccc18f6b4d23a7bf020282385cf0 : SUCCESS in 19m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0082d228fc48461fa24714bd3151f5ce : SUCCESS in 30m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/781cba890092439c920fde58536d4277 : SUCCESS in 52m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a2b058115b2343358318ce0f4427c317 : SUCCESS in 56m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e90449425db948ebba131bb208caf9b5 : SUCCESS in 1h 02m 22s","accounts_in_message":[],"_revision_number":3},{"id":"b0cbd0ef8785f682b800f8b88d18debc5337c211","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 11:32:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/394e83f51c0842bb93001b35280843a0 : SUCCESS in 49m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c4adf9156c214cb88b6688e3e42a9788 : SUCCESS in 1h 27m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/da7cb041028440d7af18c97bd4d6cae2 : SUCCESS in 1h 56m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8aa4e13d78984b99b12fde0ec4e9d949 : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7bee7119f45c4e59811f710f7b2627de : SUCCESS in 6m 33s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/87d71b0958494ade83cdb48efd19f555 : SUCCESS in 1h 24m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3c1ca3f1f9f2b7f73391510c644bbbcde23be3d3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-04 11:38:32.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"54cfa246464a28c48290acc18cbf4f2f26fa036f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-04 15:37:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e30f5fa5ab4ff9885d4b4a732d0162413c74ccb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 15:38:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"497c9190a0d19dbdb6623bb25c3d44503077525f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 17:08:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0e6bd68276504c55a67008acf7b1c059 : SUCCESS in 33m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6beea0307c2f446eb396691cc21d291d : SUCCESS in 1h 09m 46s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b8fc91458e0a4ce58d9034a63c3810e1 : SUCCESS in 1h 25m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/181c0341e3834d67b8a1c8c6bd12b422 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db3d92df31c547fe9505abf8b83cce19 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbfd1bcfdbb24fecba414f709a223e52 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acc180de956f4bdfa76b6cbe65d59178 : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cd2c006c805429081b367affbb6e025 : SUCCESS in 6m 27s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2db1e60da9d54212b15512bf4129f2a7 : SUCCESS in 32m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0e79c5759ceb4ece901aa7bf31094190 : SUCCESS in 44m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a49042ded9c246cea44bd95b17498549 : SUCCESS in 57m 21s","accounts_in_message":[],"_revision_number":3},{"id":"db679e7abfd80db4a06e4bde706fbc97fd42c649","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 17:08:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"90f5836eb343276ba2f720807eed330a41eb4550","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 17:11:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61f8567eefd547cf8ff6e2fec0476b3d : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/166f8a69251646c68dcf9c8f748a852a : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5a850592022450d9af85c37b0266fa9b7d317173","revisions":{"996ea0e977a24aae51522d1cb64ae7ba1c6b737f":{"kind":"REWORK","_number":1,"created":"2021-10-01 14:17:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/40/812140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/812140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/812140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/812140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/812140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/812140/1"}}},"commit":{"parents":[{"commit":"c729bc21e87e0fa55563b36d714db595104b45ac","subject":"Merge \"logstash: pin to 7.9.* for Elasticsearch OSS compatibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c729bc21e87e0fa55563b36d714db595104b45ac"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-01 14:15:58.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-01 14:15:58.000000000","tz":120},"subject":"Remove neutron api-paste.ini from old location","message":"Remove neutron api-paste.ini from old location\n\nWe have been configuring neutron to use /etc/neutron/api-paste.ini for\nmore than a month now. Remove this file from its old location before\nXena is released. Stop creating /usr/share/neutron which is now unused.\n\nChange-Id: Ic90cd8e3065fa629d5ad67abaf7c193fd845259a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/996ea0e977a24aae51522d1cb64ae7ba1c6b737f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/996ea0e977a24aae51522d1cb64ae7ba1c6b737f"}]},"branch":"refs/heads/master"},"03409af231ffc1384475041cdb2b9a4feb4f49a0":{"kind":"REWORK","_number":2,"created":"2021-10-01 14:25:42.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/40/812140/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/812140/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/812140/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/812140/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/812140/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/812140/2"}}},"commit":{"parents":[{"commit":"c729bc21e87e0fa55563b36d714db595104b45ac","subject":"Merge \"logstash: pin to 7.9.* for Elasticsearch OSS compatibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c729bc21e87e0fa55563b36d714db595104b45ac"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-01 14:25:30.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-01 14:25:30.000000000","tz":120},"subject":"Remove neutron api-paste.ini from old location","message":"Remove neutron api-paste.ini from old location\n\nWe have been configuring neutron to use /etc/neutron/api-paste.ini for\nmore than a month now. Remove this file from its old location before\nXena is released. Stop creating /usr/share/neutron which is now unused.\n\nChange-Id: Ic90cd8e3065fa629d5ad67abaf7c193fd845259a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03409af231ffc1384475041cdb2b9a4feb4f49a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03409af231ffc1384475041cdb2b9a4feb4f49a0"}]},"branch":"refs/heads/master"},"5a850592022450d9af85c37b0266fa9b7d317173":{"kind":"REWORK","_number":3,"created":"2021-10-04 09:12:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/40/812140/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/812140/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/812140/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/812140/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/812140/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/812140/3"}}},"commit":{"parents":[{"commit":"c729bc21e87e0fa55563b36d714db595104b45ac","subject":"Merge \"logstash: pin to 7.9.* for Elasticsearch OSS compatibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c729bc21e87e0fa55563b36d714db595104b45ac"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-04 09:12:24.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-04 09:12:24.000000000","tz":120},"subject":"Remove neutron api-paste.ini from old location","message":"Remove neutron api-paste.ini from old location\n\nWe have been configuring neutron to use /etc/neutron/api-paste.ini for\nmore than a month now. Remove this file from its old location before\nXena is released. Stop creating /usr/share/neutron which is now unused.\n\nChange-Id: Ic90cd8e3065fa629d5ad67abaf7c193fd845259a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a850592022450d9af85c37b0266fa9b7d317173"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a850592022450d9af85c37b0266fa9b7d317173"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
