)]}'
{"id":"openstack%2Fkolla~812203","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I4ec4eb837c98f57964488769ab82ba8213ad582c","project":"openstack/kolla","branch":"stable/wallaby","hashtags":[],"change_id":"I4ec4eb837c98f57964488769ab82ba8213ad582c","subject":"logstash: pin to 7.9.* for Elasticsearch OSS compatibility","status":"MERGED","created":"2021-10-04 07:53:02.000000000","updated":"2021-10-04 13:15:09.000000000","submitted":"2021-10-04 13:10:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812203","cherry_pick_of_change":811977,"cherry_pick_of_patch_set":4,"meta_rev_id":"642522794b249fc64ba350d87849f790b8fba2ed","_number":812203,"virtual_id_number":812203,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-04 13:10:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-10-04 07:59:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-10-04 07:54:45.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"date":"2021-10-04 08:04:27.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-10-04 11:50:34.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-04 07:54:45.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-10-04 07:59:24.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-10-04 09:15:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f2c01d966a79eb39057145fcff754ea89ac1d35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-04 07:53:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e0d6c47c1377f8f70562ac62da164b76548a87ad","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-04 07:54:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"09082eb83550f6f5efd86ffb856ccac40510dc01","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-04 07:59:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"28f7f268465fbabc4e6131daa7500a8492c96a7c","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-04 08:04:27.000000000","message":"Removed Workflow+1 by Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"1278ef75c79273c86817c26f63700337ae307431","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 09:15:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6b684a322968427bb53d5664cba6884c : SUCCESS in 25m 51s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ec2123abb20d47619ce0eecc54bc1b4a : SUCCESS in 40m 13s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/cb7588bd739a45018e08b5eda95a6a42 : SUCCESS in 1h 07m 40s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0d94f487eb03438285e8d768805822fb : SUCCESS in 1h 12m 55s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/19e15865c072422f9db5a278b793edcc : SUCCESS in 1h 17m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fb1e7b42f1524311b2c94ec100ae00fa : SUCCESS in 35m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/22f5fc758ce140bfb776b26763d90e96 : SUCCESS in 20m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9b61afda6e55465da16215450d15b8a4 : SUCCESS in 1h 02m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8e271a75baa43a48c88534a8dc13fe8 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a8bc9166fa448348223267f1347945b : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97de1f7adff64cf2b17017922e3bf9ee : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fde2064f7244ac4a39c2a568b73ef54 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e621d6587a449ce9572f88a83cf943d : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/787c6c8b9e5b48caac4e6f5c14214714 : SUCCESS in 4m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a40f917f2891457292945e0ef30b6b76 : SUCCESS in 22m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fb5af291a2a048f589242d4c038bf67f : SUCCESS in 37m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8fd2b608b20040baa10dadc78bec562a : SUCCESS in 52m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/addfc420b1704d5781e598054170b09b : SUCCESS in 52m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fb42f905a9754e9c94f490888bfce28f : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":1},{"id":"7fd1d5eae0dc641b2406a9dc9ff4310bbe06a7db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 10:22:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/cdc672d8a1914321942517b684ca4210 : SUCCESS in 2h 01m 02s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1c049e167cbb44f9a4f96eeeb32ca87f : SUCCESS in 1h 33m 18s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/388f281ff4964829b7cc1ab6d1ab6801 : SUCCESS in 2h 27m 38s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/dcb0f1b96aac46c7bfac060debae22b0 : SUCCESS in 1h 21m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c25f725474b5ee53c62dd1286497d69be51ce923","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-04 11:50:34.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"28ad3eee16865f3cc469e8140da9a27c97d8042a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 11:51:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f1bd96f53f3e421cd04b36e6e35327490e4f3a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 13:10:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d0f955a93d7944c6a5c683d98da162ca : SUCCESS in 30m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/624c5468bdd743f0be49a1c356abc77e : SUCCESS in 1h 13m 04s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/27bc6b06748e45afa3999ae0e9d92231 : SUCCESS in 1h 16m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aafd9e26c574de29e9d8648ba4e6b0b : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5da63e7bd6494c06bd6de6ad48c5ba04 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbc0b50507854268bf3681ccee9cd7cc : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c76d4a1a4b4c849f23fc16ceb9ced6 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ddd702ec0ff4c8e910f91577b8916e6 : SUCCESS in 4m 49s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9fd6d57005c1441b9644f52b57c417f0 : SUCCESS in 30m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/68be6333adef44aa8affe32f03a0a06d : SUCCESS in 57m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/31edf42cf52a4d3b898196098566084b : SUCCESS in 1h 12m 13s","accounts_in_message":[],"_revision_number":1},{"id":"edce37f2c84dca27de05e744417cc4c7ce0ce94d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 13:10:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"642522794b249fc64ba350d87849f790b8fba2ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 13:15:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a8e58906d1d4ee3a14933d76d16d60d : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08c2c69b2225470f8b3ccbbffb7c5898 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0366a31d268800c2f153c3d28cd84bea2119b505","revisions":{"0366a31d268800c2f153c3d28cd84bea2119b505":{"kind":"REWORK","_number":1,"created":"2021-10-04 07:53:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/03/812203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/812203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/812203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/812203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/812203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/812203/1"}}},"commit":{"parents":[{"commit":"29acd9aa309f639cbabd14280661b8acf8f1b063","subject":"Merge \"[debian] Use upstream MariaDB\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29acd9aa309f639cbabd14280661b8acf8f1b063"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-30 12:42:18.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-04 07:53:02.000000000","tz":0},"subject":"logstash: pin to 7.9.* for Elasticsearch OSS compatibility","message":"logstash: pin to 7.9.* for Elasticsearch OSS compatibility\n\nThe latest versions of logstash 7.x do not support using the OSS version\nof Elasticsearch, and fail to push logs if an OSS version is used. This\nprevents Monasca log API from working correctly.\n\nThis change fixes the issue by pinning logstash-oss to 7.9.x, the most\nrecent available package in the repo that still supports Elasticsearch\nOSS.\n\nNOTE: logstash 7.11 introduced the breaking change, however 7.10 is\nunavailable in the package repos.\n\nChange-Id: I4ec4eb837c98f57964488769ab82ba8213ad582c\nCloses-Bug: #1941754\n(cherry picked from commit b32cf27cc93878bcc124bcc387749829b0997a66)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0366a31d268800c2f153c3d28cd84bea2119b505"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0366a31d268800c2f153c3d28cd84bea2119b505"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
