)]}'
{"id":"openstack%2Fkolla~812249","triplet_id":"openstack%2Fkolla~master~I074d9d13f52c208c0f3d92ea060b3b9abb6a6b9e","project":"openstack/kolla","branch":"master","topic":"clean-mariadb-mysql","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-01-20 19:26:45.000000000","reason":"Michal Nasiadka replied on the change"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-01-20 20:51:22.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-01-12 09:42:00.000000000","reason":"Mark Goddard replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-01-20 20:51:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I074d9d13f52c208c0f3d92ea060b3b9abb6a6b9e","subject":"Clean MariaDB/MySQL","status":"MERGED","created":"2021-10-02 09:29:45.000000000","updated":"2022-01-20 20:53:47.000000000","submitted":"2022-01-20 20:51:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"812249-clean-mariadb-mysql","meta_rev_id":"c7058d51c9d7c74d5e5730061c8429f508c711f8","_number":812249,"virtual_id_number":812249,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-20 20:51:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-01-17 11:34:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":1,"date":"2021-10-03 11:39:42.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-01-20 19:26:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-01-12 09:42:00.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-15 06:32:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-01-20 19:26:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-02 10:55:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-03 11:39:42.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-10-04 08:17:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-01-07 17:41:04.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-01-07 17:41:04.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-01-07 17:41:04.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-01-15 06:32:28.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-01-17 11:34:45.000000000","updated_by":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"reviewer":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"state":"REVIEWER"}],"messages":[{"id":"96e5639182dac19c12e7d520dbbde61eb9f8f504","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-02 09:29:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b83084ef87f8834b6ae801e0f49a72fd0d0e28d7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-02 09:35:23.000000000","message":"Patch Set 1:\n\nadditional cleanup: https://review.opendev.org/c/openstack/kolla/+/812250","accounts_in_message":[],"_revision_number":1},{"id":"7dd647ebcd2ba3967b218b5523217d5e51e49c6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-02 10:55:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1ed9154f38a44b87826a6ce9a4897d27 : SUCCESS in 26m 06s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/85c205831ddf4fa3be2ef80725322aee : SUCCESS in 1h 22m 58s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0bd314e41a104041b1e50c00a5455d48 : FAILURE in 33m 30s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/64fef07d0cab4767b43a8faeea755a7f : FAILURE in 37m 18s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/bd6e2f34a7a449ab87c4ab2e4db43f17 : FAILURE in 1h 04m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5b58002d9910498992962dbe24ea0723 : SUCCESS in 28m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/53ecf572c9174282bf4ac8a4c901c1cf : SUCCESS in 17m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/42c33ac7828a425097bc6f7d8cdb35e3 : FAILURE in 39m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4745e42f41914aafa9637c6d54884088 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d5d2113e0c142018e93403279aea5e1 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0de141ecd174cbaafd280e6695b2996 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08af1ef1162b4c10a0b59b0b0c153de7 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41443ee1e4b04ff9ae447675c03095a5 : SUCCESS in 7m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2051d784f53b4d8ab04b1587aa22785d : SUCCESS in 20m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7d93e865e1a0439c8fc380f590a784ed : FAILURE in 27m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ead22ec09e3c487f9ba0d57558e89677 : FAILURE in 25m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e607b5b683f64845b65985fe584135e6 : FAILURE in 26m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/37fcc868123949ceaf03c69da791b672 : FAILURE in 52m 57s","accounts_in_message":[],"_revision_number":1},{"id":"529ab2500dcf8e53e2ff5464dbfedd8a9b4e985d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-02 11:21:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2308358f28e049149110670159651913 : SUCCESS in 1h 49m 20s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ae9a653dd044d9896834b6197fa5fbf : SUCCESS in 1h 34m 29s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b5780cb893214251acee8274fb84de1c : FAILURE in 1h 09m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72c2f005012e4bc4bf7e44f6b2e348c9 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d55c42c575e74232a134116e1c66034a : SUCCESS in 6m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6e4d27cd2c274d929e12bb23d3a07438 : SUCCESS in 1h 28m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"989436d5d5abb5799859567803d432b44a0c7854","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-02 14:07:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b7fc63726497e43aa820242d80bb20d4e96efde8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-02 14:52:14.000000000","message":"Patch Set 2:\n\nmore k-a testing in https://review.opendev.org/c/openstack/kolla-ansible/+/812252","accounts_in_message":[],"_revision_number":2},{"id":"d9220a58fa14b901e7859775b6610033b3c49d58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-02 15:48:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/086f35f4260e41dbb4b199a543092747 : SUCCESS in 1h 17m 29s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d83acc3909a642eca43cb56e1a0519f6 : SUCCESS in 36m 44s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f9a22a27769d4af09b78eb7b5bb60ff5 : SUCCESS in 1h 07m 45s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c01284556e374320ada76ec82bf996cf : SUCCESS in 1h 03m 04s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6554c94f114547f99e52df6b1c680d09 : SUCCESS in 1h 21m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5412ebe7ced0431f8342ebee133b9f67 : SUCCESS in 27m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1db1d308d3624378a61bd1f5aa404500 : SUCCESS in 15m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ca54be6cd03543da96a495e00f4170d6 : SUCCESS in 53m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/befd9cdd286c48e4be3a585391f9db81 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1fec035e9224cc2b2ad3ac61109c4c2 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2ae5040f0b04c618f55d9628f427b0c : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03a90e52ac584aeda2585a263da31086 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1cde113894a40c0a235b38df7e5dc58 : SUCCESS in 8m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f69a2992967b4e39b66254448c74a8ac : SUCCESS in 21m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/84c841e1caf84ac5b5119d2fd4fa80b3 : FAILURE in 26m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/eb21369fd15e4dc0bf15fd49e7089898 : SUCCESS in 54m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9003389dc89c48fc988830fb813f8a10 : SUCCESS in 1h 34m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/37556a92b4d34320b7e469684b2c3c84 : SUCCESS in 1h 03m 27s","accounts_in_message":[],"_revision_number":2},{"id":"5480327e3cc07240b5fd8573d10a79bcf27576d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-02 16:18:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b783d68d6c02486a845ee687e05015bc : SUCCESS in 38m 39s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ce4778eca75b4faf8bfaf0a76f68f930 : SUCCESS in 1h 19m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/84d5d2fc86e146bba8f58f77251e2b78 : SUCCESS in 1h 53m 18s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cba5f2fc1af543688d465554a4eb92f5 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f8142fb7d961400c980c65db8beba301 : SUCCESS in 6m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/59f23313e36a483da9ec14cdba136c89 : SUCCESS in 1h 52m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5851d33f8fa80e3f6dab7b2ee480141e00f3d934","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-02 17:41:09.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c85d1594fe8a4b5ce0d5ea9c2cbc1337781ac5f3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-02 17:41:36.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"9be78af550b83b9d812e62eee9582341d2c21bd6","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-02 17:41:38.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"d852586a08d4d50bce2d29278a40d0bc37304ce1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-02 19:10:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6f33ecde63804d288f8a952bb63d98af : SUCCESS in 27m 45s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1a61387a801649f4a77d7887f30d7667 : SUCCESS in 29m 34s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/52bfbb042c274cf3b3da0297dc4d8f37 : SUCCESS in 1h 01m 30s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d75184cc7c1d4b33b3d6bbcd3e453989 : SUCCESS in 1h 07m 41s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/55698e65a64e431287b9a4fcb5224c6f : SUCCESS in 1h 27m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a880dedb6f354a4a82664d50050ea589 : SUCCESS in 26m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cac2fa5fe4b549138dd7a890372d8460 : SUCCESS in 22m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2534d955086348ec98ea225dd2318101 : SUCCESS in 56m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f347ca5545e74eb1a6c0d6ecc8600d18 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e4885f96fac43a393e83c3db2344e3b : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c97385b37e7949c1965e008ef27f68a9 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/780b7aad227441a3bbff343ef14841c8 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/843ded29e62b4f4292ba616a1186985e : SUCCESS in 7m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/246af19d746647b3a0b26d1cc264bcb1 : SUCCESS in 17m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7a53d4f30a5f4786b9d722ea974245cb : SUCCESS in 29m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/00ef62ba3d2447ed97c656bd507a68f1 : SUCCESS in 47m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3978110124dc458cac3aa6193fe2e63d : SUCCESS in 47m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5bd6a576fe264f9e9a4fca064ec07fe4 : SUCCESS in 1h 12m 20s","accounts_in_message":[],"_revision_number":4},{"id":"755b6e4e9c2e2280c7348ad1f3c3e13236a3b3c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-02 19:36:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4868c90066e24760bb0dc9f4fd9d6d18 : SUCCESS in 1h 47m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5d525009f2c54ecb950a1613f38ee405 : SUCCESS in 1h 20m 01s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/775b2de094714c89ae137e5dc3d550dc : SUCCESS in 1h 49m 12s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9fdebde7e6ff4477875dd7f8921fe46e : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bfee8c918d9a4cd2baa35c4de00824c4 : SUCCESS in 5m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/58e4b015de8c43f18a6c6aaac82eacfc : SUCCESS in 1h 52m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"16dc1ae924683905273f3a3d36850f63a76e0a04","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-03 11:39:42.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"96ac6b31a6cd46705af7e57f404df97d4a09a6a4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-04 08:17:36.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6cee9d908ceacdfbe44d36873b3a532b1bb55808","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-04 08:20:16.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cadd83fb00eee9a405ba09251c0955920e2b1b8c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-07 17:41:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2c7716b27a8df8eba6f16667b28fb8af40ddcde1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-12 09:42:00.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"005e9b9327b38895c070109e5a0817afb37e7b5d","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-01-15 06:32:28.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5c13ddb2bdd8c4ebead91232240e3cec462b983e","author":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"date":"2022-01-17 11:34:45.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5756df63b0dfd5e4b70a7085cae8245e8cec93b7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-20 19:26:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"358cb3e17fec3227002694d8cfd18e82bce9240b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 19:28:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8cfbf3bd384897348b3bd944ad49456aafe4c953","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 20:51:22.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2a1e64f63cc84a7cb93427ebf95cba3d : SUCCESS in 37m 39s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/97266e2eafe24c4cbad26df9ec95e865 : SUCCESS in 1h 10m 11s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/80c62d7826294c1bb9142b75b3e042bd : SUCCESS in 1h 17m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/882bf1b0078e4aa78e59f56483e22a8f : SUCCESS in 28m 46s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2cd34e32ca1a4dfda8e570d40a622dcd : SUCCESS in 1h 01m 15s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e648a09f635749a793d14299fe94edcb : SUCCESS in 1h 16m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a63adb7d9ec4a638e7247a2ec303722 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/174ac2d858e24725a1af4da86423bda3 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6b2dfb402114804829a874230a6a442 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89ea60abf6d948f6845946b16e29d76c : SUCCESS in 6m 25s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7ba65255d2ea40c9aa947037119d2c0d : SUCCESS in 20m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4e8793b85eb94b37a36ce3cad50de5c5 : SUCCESS in 54m 25s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b5a9bb2df3b244358ea813a438afd481 : SUCCESS in 1h 03m 01s","accounts_in_message":[],"_revision_number":4},{"id":"ff46147eba686b42ff5f32f2cf99455ee82e5918","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 20:51:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c7058d51c9d7c74d5e5730061c8429f508c711f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 20:53:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/957f78ad590142538983d5c18da3b919 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6b203efde7e60ae6ab63f230b4e9c944eb83c5a2","revisions":{"049410fa1a63a13daf198ea4a5fc7ff50e01e5cd":{"kind":"REWORK","_number":1,"created":"2021-10-02 09:29:45.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/49/812249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/812249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/812249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/812249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/812249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/812249/1"}}},"commit":{"parents":[{"commit":"c729bc21e87e0fa55563b36d714db595104b45ac","subject":"Merge \"logstash: pin to 7.9.* for Elasticsearch OSS compatibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c729bc21e87e0fa55563b36d714db595104b45ac"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-02 09:28:56.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-02 09:28:56.000000000","tz":0},"subject":"[WIP] Clean MariaDB/MySQL","message":"[WIP] Clean MariaDB/MySQL\n\nWe are likely overinstalling these.\nPyMySQL should be enough.\n\nChange-Id: I074d9d13f52c208c0f3d92ea060b3b9abb6a6b9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/049410fa1a63a13daf198ea4a5fc7ff50e01e5cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/049410fa1a63a13daf198ea4a5fc7ff50e01e5cd"}]},"branch":"refs/heads/master"},"4cab65f4d6e4c2ff43e20f88ccef3a185a2921c9":{"kind":"REWORK","_number":2,"created":"2021-10-02 14:07:24.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/49/812249/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/812249/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/812249/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/812249/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/812249/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/812249/2"}}},"commit":{"parents":[{"commit":"c729bc21e87e0fa55563b36d714db595104b45ac","subject":"Merge \"logstash: pin to 7.9.* for Elasticsearch OSS compatibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c729bc21e87e0fa55563b36d714db595104b45ac"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-02 09:28:56.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-02 14:06:44.000000000","tz":0},"subject":"[WIP] Clean MariaDB/MySQL","message":"[WIP] Clean MariaDB/MySQL\n\nWe are likely overinstalling these.\nPyMySQL should be enough.\n\nChange-Id: I074d9d13f52c208c0f3d92ea060b3b9abb6a6b9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4cab65f4d6e4c2ff43e20f88ccef3a185a2921c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4cab65f4d6e4c2ff43e20f88ccef3a185a2921c9"}]},"branch":"refs/heads/master"},"32995731dcf4af2d1a33acfb8747ab89f59f35c3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-10-02 17:41:09.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/49/812249/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/812249/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/812249/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/812249/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/812249/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/812249/3"}}},"commit":{"parents":[{"commit":"c4d227cddca59694e992d9054940bc6057d1b1b3","subject":"[bifrost] Drop workarounds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4d227cddca59694e992d9054940bc6057d1b1b3"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-02 09:28:56.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-02 17:41:09.000000000","tz":0},"subject":"[WIP] Clean MariaDB/MySQL","message":"[WIP] Clean MariaDB/MySQL\n\nWe are likely overinstalling these.\nPyMySQL should be enough.\n\nChange-Id: I074d9d13f52c208c0f3d92ea060b3b9abb6a6b9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32995731dcf4af2d1a33acfb8747ab89f59f35c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32995731dcf4af2d1a33acfb8747ab89f59f35c3"}]},"branch":"refs/heads/master","description":"Rebase"},"6b203efde7e60ae6ab63f230b4e9c944eb83c5a2":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-10-02 17:41:36.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/49/812249/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/812249/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/812249/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/812249/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/812249/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/812249/4"}}},"commit":{"parents":[{"commit":"c4d227cddca59694e992d9054940bc6057d1b1b3","subject":"[bifrost] Drop workarounds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4d227cddca59694e992d9054940bc6057d1b1b3"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-02 09:28:56.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-02 17:41:33.000000000","tz":0},"subject":"Clean MariaDB/MySQL","message":"Clean MariaDB/MySQL\n\nWe are just overinstalling these.\nPyMySQL is enough nowadays.\n\nChange-Id: I074d9d13f52c208c0f3d92ea060b3b9abb6a6b9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b203efde7e60ae6ab63f230b4e9c944eb83c5a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b203efde7e60ae6ab63f230b4e9c944eb83c5a2"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
