)]}'
{"id":"openstack%2Fkolla~812253","triplet_id":"openstack%2Fkolla~master~I13bc63447da878890730e9458b48267a8530d5af","project":"openstack/kolla","branch":"master","topic":"drop-bifrost-workarounds","hashtags":[],"change_id":"I13bc63447da878890730e9458b48267a8530d5af","subject":"[bifrost] Drop workarounds","status":"MERGED","created":"2021-10-02 15:07:29.000000000","updated":"2021-10-06 14:57:25.000000000","submitted":"2021-10-06 14:55:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812253-drop-bifrost-workarounds","meta_rev_id":"86f6b4f0ba71425cc12e76f507f3b7c51affd92c","_number":812253,"virtual_id_number":812253,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-06 14:54:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-10-06 13:16:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-10-04 07:57:52.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-10-06 13:16:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-02 16:27:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-04 07:57:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-06 13:16:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"f26193f0a9f0175fe17356b45aea7634dd7f1c82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-02 15:07:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa97e7575c98e44ef271e7f8d0ecc8dfba9ba612","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-02 15:09:13.000000000","message":"Patch Set 1:\n\nproper testing in https://review.opendev.org/c/openstack/kolla-ansible/+/812254","accounts_in_message":[],"_revision_number":1},{"id":"d7704ee2d95ac8cc209db5e97c220d537de525fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-02 16:27:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fe80ed9558b54656b0b7b17f9b28401e : SUCCESS in 36m 31s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a6c9c9e371b447fe8dbbbb1eca3ff91d : SUCCESS in 27m 38s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/98be67e755c347eda690da9e206a8e47 : SUCCESS in 1h 09m 56s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/3d1f3205c8534942ba8f03210f0a0279 : SUCCESS in 1h 07m 09s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3fb712d4200e49e58bb15966f3905705 : SUCCESS in 1h 17m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/753fd5c778e946ff9d79e08163c80b1b : SUCCESS in 23m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e80b6aaa127f4f068f577d516ebcd0c5 : SUCCESS in 13m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ed8b4ecec8794d7dace2ead729d2a98e : SUCCESS in 54m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f6f122acb3448109fbf7e8c3b175c2b : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75296bf9e62343998a344235d842c26a : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82cd7949bb484494b7c202f6ba5e56e8 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eeab112a96f54c1a9655d164ad330ba4 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cbc836f29e04a4b9e6ad411a4e18828 : SUCCESS in 5m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4078eb0ddb1944bc9a7fc8709acd23cc : SUCCESS in 19m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/98e259b967d14f9f9bf3b975eec7aef6 : SUCCESS in 26m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0f63055e328b4061b9a1fa575cf11db5 : SUCCESS in 44m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ca6fd6034b17402396fe817b7dca8bc2 : SUCCESS in 58m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cee79191a513473da78a5fc539ce2dc5 : SUCCESS in 1h 00m 19s","accounts_in_message":[],"_revision_number":1},{"id":"957cae3a24618596c3965e0b7202ed9d8b7e36e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-02 17:02:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7be63345f4ac4f2589a1948a24e0c6c6 : SUCCESS in 1h 37m 50s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2c6d286416cc4e339e1c992265cc5fd3 : SUCCESS in 1h 27m 48s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b4dcc3bed0484ba393d455771eba5d33 : SUCCESS in 1h 52m 22s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/79fcfc99e9f44f949cc7d3354863edbe : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/741e917f1e9a45dc979d1e03e5ec724f : SUCCESS in 6m 02s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2c173a4933204b3d9ee9f2f2e156796a : SUCCESS in 1h 24m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ac376262611d87e7d290fc6215197811dd3e38ba","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-04 07:57:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c504e7eddc283c3cb7156d24b76566d3f0e9e5a9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-06 13:16:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9e531d799e70db50a0ee2096921d71b3d29d1a13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 13:20:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"28248e74a30098a4bd5ba8e897b1fb289744cc6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 14:54:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/90c552b20a7446aa83f674cfbf111a6f : SUCCESS in 47m 48s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e16b73bcff8148fba2fd178507766186 : SUCCESS in 1h 04m 51s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/68257d2766d64bcd857325886e566601 : SUCCESS in 1h 30m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2dc4c0105cd479d980efd36ee81b3d3 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/959024d01ea64205b7281a2ce7895b59 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38e89b364cdd4588930522e012e6c330 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bb75a002aea4094bde0223790de9c46 : SUCCESS in 7m 15s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/40063020b6224c5ca9bd9ed1e67a8936 : SUCCESS in 36m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/84cd9d1fb7ab4cb4807046c853a759ab : SUCCESS in 58m 55s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b240a0c5e9bb4b9993ab72901dfb6861 : SUCCESS in 1h 02m 08s","accounts_in_message":[],"_revision_number":1},{"id":"30e6bc35fd64adca9aaab40646cdf7129e3d6ab9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 14:55:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"86f6b4f0ba71425cc12e76f507f3b7c51affd92c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 14:57:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c1de0e12333458ab7302f160bc27456 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4d227cddca59694e992d9054940bc6057d1b1b3","revisions":{"c4d227cddca59694e992d9054940bc6057d1b1b3":{"kind":"REWORK","_number":1,"created":"2021-10-02 15:07:29.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/53/812253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/812253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/812253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/812253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/812253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/812253/1"}}},"commit":{"parents":[{"commit":"c729bc21e87e0fa55563b36d714db595104b45ac","subject":"Merge \"logstash: pin to 7.9.* for Elasticsearch OSS compatibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c729bc21e87e0fa55563b36d714db595104b45ac"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-02 15:06:40.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-02 15:06:40.000000000","tz":0},"subject":"[bifrost] Drop workarounds","message":"[bifrost] Drop workarounds\n\nThese are no longer required but block other cleanups.\n\nChange-Id: I13bc63447da878890730e9458b48267a8530d5af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4d227cddca59694e992d9054940bc6057d1b1b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4d227cddca59694e992d9054940bc6057d1b1b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
