)]}'
{"id":"openstack%2Fkolla~812479","triplet_id":"openstack%2Fkolla~master~I05cc853ab76aaf1b97ae7caad4705dc9da6a8d98","project":"openstack/kolla","branch":"master","topic":"change-812479-1","attention_set":{},"removed_from_attention_set":{"28752":{"account":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"last_update":"2022-06-02 11:20:06.000000000","reason":"Change was abandoned"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2022-06-02 11:20:06.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I05cc853ab76aaf1b97ae7caad4705dc9da6a8d98","subject":"[DNM] Remove default policy files","status":"ABANDONED","created":"2021-10-05 10:56:44.000000000","updated":"2022-06-02 11:20:06.000000000","total_comment_count":7,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"9535a059da6485e6c2a2bcf3df82a872eae48142","_number":812479,"virtual_id_number":812479,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-12 15:34:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2022-03-08 17:15:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":-1,"date":"2021-11-23 12:08:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"}],"CC":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-05 11:39:32.000000000","updated_by":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"reviewer":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"state":"CC"},{"updated":"2021-10-05 12:26:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-23 12:08:55.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-03-08 17:15:58.000000000","updated_by":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"reviewer":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"state":"REVIEWER"},{"updated":"2022-03-08 19:23:11.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"}],"messages":[{"id":"4719da6688ac77bd0ea4c3fd700ed8a69ec2cd2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-05 10:56:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9b0cec0815a8907289c77c33b732a3f2abc2fa0","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2021-10-05 11:39:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d589dda018a5007000f36d1a7e9d78fcb33b259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 12:26:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8b57f0907646405e9f760f8b3252eb5a : SUCCESS in 19m 58s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f69a8198bddc4a789f772c676d3097b5 : SUCCESS in 44m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/01e7730bc389498d8c4d6a9a0ae423ac : SUCCESS in 59m 52s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/5c9beeb88fd748f0838d9678c0450c6a : SUCCESS in 1h 10m 06s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/76006490ac8744e6aedfa4227e450ca4 : SUCCESS in 1h 24m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/413e41be18e940c0ac528f3a096a6112 : SUCCESS in 23m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/76bd3f08052146778d06276c8bbc8f80 : SUCCESS in 16m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d788bf455a934e38b2e01c21efcbbaef : SUCCESS in 55m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7b310b556034c64975ac7e3f40d3828 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35859821acab424abc8a3eecbf44d2ab : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/998d8330495044e39826a0978bcc52d7 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c134e8839a124695b52f46bd15ff3abb : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8635d5251eaa4d9b83ddd583ba813755 : SUCCESS in 5m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f87f91944b214a17a6baa59456ed3e35 : SUCCESS in 20m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/64b97fa5bceb413d9a7bbfcbf53c66e1 : SUCCESS in 29m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/621c6b0296224f2780282d204f0412ed : SUCCESS in 49m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ee9fbd38487146f5bd5bd2ed4aec616b : SUCCESS in 1h 01m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e66a00b9057a4cb997e156805d0ee911 : SUCCESS in 1h 05m 03s","accounts_in_message":[],"_revision_number":1},{"id":"a195d258e2d711f4d202fc9720b94d8d8fec7750","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 14:18:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/30abf44bfd0e4642a5f04e645b06b7b9 : SUCCESS in 1h 59m 29s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/86289c3e38d240ee8e48bc648217aa77 : SUCCESS in 1h 27m 39s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a6ae7319b43148b3adb0bfd3871f495f : SUCCESS in 2h 19m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/965206e565df4bff8576737e0ea82133 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f68cd0f085114036b057137a131f63eb : SUCCESS in 7m 01s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/364dedb29356404492db2ebeb70bb844 : SUCCESS in 1h 24m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eebf3b4e24cc5c856635c0476035dda3b6a80bbd","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-06 10:38:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b68e254f6efb308fa112337958dc6dba7fe7360a","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2021-10-06 12:09:29.000000000","message":"Patch Set 1:\n\nMy comment was more about the distribution vs sources models than what\u0027s the command do :p\n\nAll in all here is my point:\n\nDistributors such as Debian/Centos/etc are by nature opinionated.\nThey choose to put (for whatever reasons, this is not the point to discuss here) to put some policies by default within their packages.\n\nSo I\u0027m thinking we shouldn\u0027t question the distribution opinion for binary based build as a tool provider BUT let the final user to choose to override/erase those addition by itself within Kolla-Ansible rather than on Kolla itself.\n\nThat would avoid us having to delete the macro if ever Debian or others changed their mind.","accounts_in_message":[],"_revision_number":1},{"id":"1bc119e7bb2efb3bf70b472f8aa5b06d9b280d7d","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-06 14:17:53.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e My comment was more about the distribution vs sources models than what\u0027s the command do :p\n\u003e \n\u003e All in all here is my point:\n\u003e \n\u003e Distributors such as Debian/Centos/etc are by nature opinionated.\n\u003e They choose to put (for whatever reasons, this is not the point to discuss here) to put some policies by default within their packages.\n\u003e \n\u003e So I\u0027m thinking we shouldn\u0027t question the distribution opinion for binary based build as a tool provider BUT let the final user to choose to override/erase those addition by itself within Kolla-Ansible rather than on Kolla itself.\n\u003e \n\u003e That would avoid us having to delete the macro if ever Debian or others changed their mind.\n\nHmm, for example horizon Dockerfile totally ignoring distro packages and do his own work:\n- deleting files\n- deleting symlinks\n- etc\n\nBut whatewer, horizon is not on discuss now 😊\nSo, my question is, what about to use policy.d also for source,centos ..etc ? And rework kolla-ansible to copy policy to that folder, so everywhere it will be same (moreover, kolla can set some default if wants)..","accounts_in_message":[],"_revision_number":1},{"id":"6d754ea2ea05427503a1cb761934a899cbdd92eb","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2021-10-06 14:45:26.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e My comment was more about the distribution vs sources models than what\u0027s the command do :p\n\u003e \u003e \n\u003e \u003e All in all here is my point:\n\u003e \u003e \n\u003e \u003e Distributors such as Debian/Centos/etc are by nature opinionated.\n\u003e \u003e They choose to put (for whatever reasons, this is not the point to discuss here) to put some policies by default within their packages.\n\u003e \u003e \n\u003e \u003e So I\u0027m thinking we shouldn\u0027t question the distribution opinion for binary based build as a tool provider BUT let the final user to choose to override/erase those addition by itself within Kolla-Ansible rather than on Kolla itself.\n\u003e \u003e \n\u003e \u003e That would avoid us having to delete the macro if ever Debian or others changed their mind.\n\u003e \n\u003e Hmm, for example horizon Dockerfile totally ignoring distro packages and do his own work:\n\u003e - deleting files\n\u003e - deleting symlinks\n\u003e - etc\n\u003e \n\u003e But whatewer, horizon is not on discuss now 😊\n\u003e So, my question is, what about to use policy.d also for source,centos ..etc ? And rework kolla-ansible to copy policy to that folder, so everywhere it will be same (moreover, kolla can set some default if wants)..\n\nYes exactly that!","accounts_in_message":[],"_revision_number":1},{"id":"87699411640b450e916b1d81c692351391c8efbe","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-06 14:46:35.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \n\u003e \u003e \u003e My comment was more about the distribution vs sources models than what\u0027s the command do :p\n\u003e \u003e \u003e \n\u003e \u003e \u003e All in all here is my point:\n\u003e \u003e \u003e \n\u003e \u003e \u003e Distributors such as Debian/Centos/etc are by nature opinionated.\n\u003e \u003e \u003e They choose to put (for whatever reasons, this is not the point to discuss here) to put some policies by default within their packages.\n\u003e \u003e \u003e \n\u003e \u003e \u003e So I\u0027m thinking we shouldn\u0027t question the distribution opinion for binary based build as a tool provider BUT let the final user to choose to override/erase those addition by itself within Kolla-Ansible rather than on Kolla itself.\n\u003e \u003e \u003e \n\u003e \u003e \u003e That would avoid us having to delete the macro if ever Debian or others changed their mind.\n\u003e \u003e \n\u003e \u003e Hmm, for example horizon Dockerfile totally ignoring distro packages and do his own work:\n\u003e \u003e - deleting files\n\u003e \u003e - deleting symlinks\n\u003e \u003e - etc\n\u003e \u003e \n\u003e \u003e But whatewer, horizon is not on discuss now 😊\n\u003e \u003e So, my question is, what about to use policy.d also for source,centos ..etc ? And rework kolla-ansible to copy policy to that folder, so everywhere it will be same (moreover, kolla can set some default if wants)..\n\u003e \n\u003e Yes exactly that!\n\nThank you Fl1nt :) , i will propose changes","accounts_in_message":[],"_revision_number":1},{"id":"bb4ac65726721e10c74649561754bbe6f3767131","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-06 17:03:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7c6a659a97bcec9f8d77b8afdecda20b8649218f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-06 17:09:56.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"665998367bdc25d41af50a479652e8c13307c3e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 18:24:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9cacded900664053a24bb033b65b5f04 : SUCCESS in 34m 09s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2d38540ad32a4edba790e24b5b40d880 : SUCCESS in 36m 02s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ec03f897ee4b41bc9ce8f1ffce5c574a : SUCCESS in 55m 37s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4bcbf9b88dfb441a94df3beab67af825 : SUCCESS in 54m 03s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/dd9d7aa6f6fc479a840028a367a4304f : SUCCESS in 1h 12m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fcdb183fe0b3417b83eef2c3ba971748 : SUCCESS in 30m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ec8916689f4c496ebb575f33517d22f6 : SUCCESS in 21m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/80b8447cbfec4b48b579036d39eda4f9 : SUCCESS in 45m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65a9b1fd140247ba93b1e985e328bf1f : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91524ad26a854d1c8e9c118779adc2d4 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/386d01bae13e49e0967213a303edd143 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa2168fd95c84dcb9aa86cd063df633e : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1285ec90cbe94859bf39db2a8db52575 : SUCCESS in 6m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6127aaa853ea479b80b404c3943ad4f5 : SUCCESS in 23m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7c9e18c8563044da918db27226237b22 : SUCCESS in 31m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/998da152121d4d029ab210006eb292c5 : SUCCESS in 43m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/46d4be502946462eae8914f2140c31d1 : SUCCESS in 49m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/53bb34e45ae94111be50c59ce8945ff1 : SUCCESS in 1h 09m 27s","accounts_in_message":[],"_revision_number":3},{"id":"8745fd2e8134605e3e8d0b7abc4379fc811d8410","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 23:39:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/302ff1db7bbc4ccf82db6869d7a3a489 : TIMED_OUT in 6h 03m 57s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4da4e96cd18a47e4a8c897c44ea7d029 : SUCCESS in 2h 10m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/626923aefb194f7ca9398f04e4cc9e08 : SUCCESS in 2h 03m 20s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/50820da20a8e47aeb0f6f4b466509abe : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9cae38d59a884c67a26c4441b387beea : SUCCESS in 6m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7b43db1b5b04457690f079962d8b3bc8 : SUCCESS in 1h 21m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dd222c3c1a9ca13c915422ba93099fb06b4bebc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-12 14:03:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"007ec9ee8849e051e03c0bec716b09d543caa3ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 15:34:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/2b69f6b987094589b5d39fdf648428fc : SUCCESS in 27m 43s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/08ef5afe3cc342228bac3ffad1fa2220 : SUCCESS in 36m 00s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/595fad904ef047a9b53429e5c64184c5 : SUCCESS in 1h 11m 46s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/85c38427907344418ee56107c888a19e : SUCCESS in 58m 03s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3461e71f08174c9a8758e61bcb8e4bc1 : SUCCESS in 1h 28m 36s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f5b451f14bdd44e78f8543d3a676200d : SUCCESS in 27m 36s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/acfeff6d773d4b528588a50eba470f0e : FAILURE in 21m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/534b7a9674944008a95f157caec66532 : SUCCESS in 54m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c81173e49ec04db4ada4b391ce49cc9d : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe1f023e50b947c3a07aaede9cddfa18 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aec225a2f5cd4511811cc3244ef5ae3e : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0403b0ddd3a2467486afefdde0bf431a : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7334b27742f64951b570475c363df8f4 : SUCCESS in 7m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7e22169714c6492ea6745f002aa2240c : SUCCESS in 37m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4815e568f4fa4503a02c02f9b94e63e8 : SUCCESS in 31m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2948b1a681d44034854e99f97d43e7c8 : SUCCESS in 48m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/591794f51a764709b4c4de3bbec89437 : SUCCESS in 52m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2991e39a2ca64a22940639f09c8f42a6 : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":4},{"id":"d90b1c3170218a840e4399a1761bf37c94b6a185","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 16:02:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/54859fba7ef4448a8ff4a18ef8f2fd31 : SUCCESS in 1h 55m 10s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/64aafe2109f24313b8376c306841d239 : SUCCESS in 1h 33m 39s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/40f35d36333a4824a1a2b7161e0904df : SUCCESS in 1h 56m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f1bdcea07f9a4e019c589093d7b17a33 : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/302b15814e254eed9b7bb359cb9dae6c : SUCCESS in 9m 42s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/98b8c16059a34d0cbee538defa3e99db : SUCCESS in 1h 38m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b1f2b752b2ecb52236c94fa3028903c526baa249","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-23 12:08:55.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"8e58f4c9fa1fb0591c1e44d9d4918d2b0bca9a8e","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2022-03-08 17:15:58.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0e4ab981bad71ccd3981c05dfdc929d6d3d2ee18","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-08 19:23:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9535a059da6485e6c2a2bcf3df82a872eae48142","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-06-02 11:20:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"051dcbe5a23f65c5827f41f2246671bb857c76a2","revisions":{"cd94d386e950588f07b9da9de526af2e8a22d660":{"kind":"REWORK","_number":1,"created":"2021-10-05 10:56:44.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/79/812479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/812479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/812479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/812479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/812479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/812479/1"}}},"commit":{"parents":[{"commit":"1255eec797d324d68ce0028e7a9f74d11bca0fe9","subject":"Merge \"Remove neutron api-paste.ini from old location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1255eec797d324d68ce0028e7a9f74d11bca0fe9"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-05 10:49:59.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-05 10:56:41.000000000","tz":120},"subject":"Remove default policy files","message":"Remove default policy files\n\nDebian added policy.d folder where default\npolicy files are stored. This is problem as\nthese policy files are overwriting policy yaml\npotentionally injected by kolla-ansible or\ndefault in-code policy.\n\nThis patch is adding macro which is used to remove\ndefault policy files from openstack images.\n\nChange-Id: I05cc853ab76aaf1b97ae7caad4705dc9da6a8d98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd94d386e950588f07b9da9de526af2e8a22d660"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd94d386e950588f07b9da9de526af2e8a22d660"}]},"branch":"refs/heads/master"},"271833cc46f627e816a531d937b93324e16df4e5":{"kind":"REWORK","_number":2,"created":"2021-10-06 17:03:40.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/79/812479/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/812479/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/812479/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/812479/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/812479/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/812479/2"}}},"commit":{"parents":[{"commit":"1255eec797d324d68ce0028e7a9f74d11bca0fe9","subject":"Merge \"Remove neutron api-paste.ini from old location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1255eec797d324d68ce0028e7a9f74d11bca0fe9"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-05 10:49:59.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-06 17:03:37.000000000","tz":120},"subject":"Remove default policy files","message":"Remove default policy files\n\nDebian added policy.d folder where default\npolicy files are stored. This is problem as\nthese policy files are overwriting policy yaml\npotentionally injected by kolla-ansible or\ndefault in-code policy.\n\nThis patch is adding macro which is used to remove\ndefault policy files from openstack images.\n\nChange-Id: I05cc853ab76aaf1b97ae7caad4705dc9da6a8d98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/271833cc46f627e816a531d937b93324e16df4e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/271833cc46f627e816a531d937b93324e16df4e5"}]},"branch":"refs/heads/master"},"9d7a220fa266d21d7253decc26debf2e57101685":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-10-06 17:09:56.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/79/812479/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/812479/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/812479/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/812479/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/812479/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/812479/3"}}},"commit":{"parents":[{"commit":"1255eec797d324d68ce0028e7a9f74d11bca0fe9","subject":"Merge \"Remove neutron api-paste.ini from old location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1255eec797d324d68ce0028e7a9f74d11bca0fe9"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-05 10:49:59.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-06 17:03:43.000000000","tz":120},"subject":"Switch openstack services to use policy.d","message":"Switch openstack services to use policy.d\n\nDebian binary packages already providing policy.d\nfolder where can be potentionally some default\npolicy (default commented). This dir is preferred,\nor better said - files inside that dir, overwrites\nthe global one which is injected by kolla-ansible\nif user defined custom policy.\n\nThis patch is step needed to kolla-ansible\ncan implement copying policy files into that folder.\n\nThis approach has several pros:\n  - User can split several policies logically\n  - Kolla can provide own default policy (if want)\n\nChange-Id: I05cc853ab76aaf1b97ae7caad4705dc9da6a8d98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d7a220fa266d21d7253decc26debf2e57101685"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d7a220fa266d21d7253decc26debf2e57101685"}]},"branch":"refs/heads/master"},"051dcbe5a23f65c5827f41f2246671bb857c76a2":{"kind":"REWORK","_number":4,"created":"2021-10-12 14:03:27.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/79/812479/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/812479/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/812479/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/812479/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/812479/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/812479/4"}}},"commit":{"parents":[{"commit":"529e7add036e4333fdb33ac4d272d3f65d482e0c","subject":"Merge \"Add missing reno\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/529e7add036e4333fdb33ac4d272d3f65d482e0c"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-05 10:49:59.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-12 14:03:26.000000000","tz":120},"subject":"[DNM] Remove default policy files","message":"[DNM] Remove default policy files\n\nDebian added policy.d folder where default\npolicy files are stored. This is problem as\nthese policy files are overwriting policy yaml\npotentionally injected by kolla-ansible or\ndefault in-code policy.\n\nThis patch is adding macro which is used to remove\ndefault policy files from openstack images.\n\nChange-Id: I05cc853ab76aaf1b97ae7caad4705dc9da6a8d98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/051dcbe5a23f65c5827f41f2246671bb857c76a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/051dcbe5a23f65c5827f41f2246671bb857c76a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
