)]}'
{"id":"openstack%2Fkolla~812505","triplet_id":"openstack%2Fkolla~master~I461f598bca6606f1d2939e385bb7befe2b792aa6","project":"openstack/kolla","branch":"master","attention_set":{"32657":{"account":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"last_update":"2021-11-03 19:30:26.000000000","reason":"Radosław Piliszek replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I461f598bca6606f1d2939e385bb7befe2b792aa6","subject":"Remove InfluxDB data source Grafana plugin","status":"MERGED","created":"2021-10-05 13:55:38.000000000","updated":"2021-11-03 19:30:26.000000000","submitted":"2021-11-03 11:55:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"812505","meta_rev_id":"295d03675b362deff29d81a5e4c734a6c38a13b7","_number":812505,"virtual_id_number":812505,"owner":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-03 11:55:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-11-03 10:23:46.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-10-06 09:54:22.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-11-03 10:23:46.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"CC":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-05 15:47:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-10-05 15:52:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-06 09:54:22.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-11-03 10:23:46.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-11-03 19:30:26.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"}],"messages":[{"id":"200071277096926f507591948c6801c293851875","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-10-05 13:55:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fc472e5de47226adbf27ecc98a2b3c668d0bef7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-05 15:47:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bc343fb3795e8a082c2d06c5cf4de698d13bc048","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 15:52:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/963a0271e88246a0a77ec6340eb1bebf : SUCCESS in 26m 22s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/99c1649469fc40d5a364047ad01c9252 : SUCCESS in 47m 12s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a3daee2d7891426896103407345f580a : SUCCESS in 58m 00s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/121684f6036e40da8227bab2b519bc05 : SUCCESS in 1h 06m 56s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7e81241c4baa4abcb85ac902fe2e20d1 : SUCCESS in 1h 29m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5093f274e78e426eadeeb9fc91a2529e : SUCCESS in 32m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ac84b8c05a014f27b6a14f3116e1b085 : SUCCESS in 17m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fdd7cd8a6c9d441cbe2c9acc6280795e : SUCCESS in 56m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d76d2540598c4e60b7f9fbdb04e455e6 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cae061da9fb34315b89f64f35efebebb : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d26024ff84a84a709a4cee45bd557a00 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7df1a24a2204270982313798e80c831 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93fca63ef4ee4876bb45aa714051409f : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7ba90af1cf749f19825fba85ac07a3e : SUCCESS in 4m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d159195144ab48beac17f2b65c81f1fe : SUCCESS in 19m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fb5b5734dc2d408283ea2ca106e2796c : SUCCESS in 35m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/70cf5b9758474f3fb96056ec1ce12592 : SUCCESS in 50m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/84b190200d244660afd604a5e5063bc0 : SUCCESS in 46m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f50a71d5b7154a03ad2b2b4964de14a7 : SUCCESS in 1h 03m 59s","accounts_in_message":[],"_revision_number":1},{"id":"789266817b132b71a182b33d86530e903cb41a15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 16:49:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1241753b94914602ba108d4b9a568a75 : SUCCESS in 2h 41m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/20979f65e6b14c84976b469e269325fa : SUCCESS in 1h 49m 48s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2b16f628fff04adda3341e5638d424e4 : SUCCESS in 2h 08m 41s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/33106e11694b4ce69466f0ff9705a8bf : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/411623ccd8eb493dbd57b275df363ce4 : SUCCESS in 6m 41s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1a73904de5da4739b48b70368cd27421 : SUCCESS in 1h 51m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cdc9bf4ed13ddef743ea8cbd557563f6a360c685","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-10-05 18:00:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f4798138c8546f1c6ef28a5c31fd4f2ac3d4197a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 19:31:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/c3cf79f92f4941aab2aeb0ebc33355d1 : SUCCESS in 27m 57s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/07b1310739174b72b75d4e11140ab634 : SUCCESS in 42m 45s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1f8e3af887bf4ffd9ce501e67b3207a0 : SUCCESS in 1h 20m 41s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0e9626145d49439994a512edf26b9244 : SUCCESS in 1h 08m 58s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/47d84e6b51d6427598e18d8c9031090a : SUCCESS in 1h 25m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c10df6f072864fada5bae36242a2f43f : SUCCESS in 23m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c7629e2a4a7b427cb5f4c36dc8751e39 : SUCCESS in 20m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/63eb6d9786f243a48c59ba6079ec2dfa : SUCCESS in 1h 03m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1c5e49cae844ea48fd22f4fecc9ba86 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe1ab02336094f4fb81bc3b2e49b8f18 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2c7e93416e34e4d9101b7fc5a029b95 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4d8de28b84c42299188f106de9d2e76 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe31edd7bf854bafb781525dc033b622 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce4e58425318485cbdf0edcbb2f24784 : SUCCESS in 4m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1fe78bd8eebc4ce69613253ab8a0a9f3 : SUCCESS in 19m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0d9b1ba7d24142d49e22380870f852f0 : SUCCESS in 30m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0eb64465ad2f4fefa25b8f5e16411c84 : SUCCESS in 50m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a71451bd15d64170994cc39a251d07dc : SUCCESS in 57m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/432ee214cdba473faf90c62db2f2f7bd : SUCCESS in 1h 03m 48s","accounts_in_message":[],"_revision_number":2},{"id":"2414cda89a7163e495eb0d921bb21d27f70689cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 20:07:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7f90f7fdbc324cda9ca1e66026e4f966 : SUCCESS in 1h 57m 15s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e3639f78c1f24d309b42bb0a649af8fa : SUCCESS in 1h 14m 29s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/81576aafde754655b1e87a9d37034765 : SUCCESS in 2h 04m 31s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a3bdaf9ab6474629b7ca54a00e77eb25 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0fe4bc74be264d2db0262212ce7d251f : SUCCESS in 6m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2e3b8c0a688543ddaa0ff2461031a612 : SUCCESS in 1h 13m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5d3778e42a2317fb08e7e49266f6997d6380f150","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-06 09:54:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d2c29ece149134baed5b68b744685965ba267c59","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-03 10:23:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"eb948f3e690b9faaee37a74c81f39eae5a978dfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 10:24:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a51804a345021c05ebd103eda13c6b8760d93918","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 11:55:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/cb44245afc774759a9b14d80e395579c : SUCCESS in 38m 55s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a9ee3b36eba940e68c89f84e99114c5d : SUCCESS in 1h 14m 43s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e73ede81ca8446858c6c30947d53aea5 : SUCCESS in 1h 29m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e67925a270d04fa1830344dd8817094b : SUCCESS in 28m 50s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7e76d2dd9dfc4c9eac7372a2a96759a2 : SUCCESS in 1h 08m 19s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/de1847b700064baaa5a1f268073c7f8f : SUCCESS in 1h 06m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/858065adbd4a43b88b01260e3130b4c2 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a35b0de3e92e41d69740cd97561319b7 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c33d54f87dbb411db95f5595132b14ee : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07e06cd5a69242648c5db79155246160 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f8bd21d5dd94d04a87d34a8f4f69a70 : SUCCESS in 4m 59s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2fd41446d6f14355b74dd6af59ab8eaf : SUCCESS in 31m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/50579192ec5047f0936c1f935c416a88 : SUCCESS in 54m 12s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5bc9ca35426548889c41639a01f6b13a : SUCCESS in 1h 08m 50s","accounts_in_message":[],"_revision_number":2},{"id":"e76e7b7482a4cbb3f033f310af0b13f449ef163f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 11:55:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"427d235ea8e153a04b4c9fb3e110a9885be89d61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 11:57:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72048af56e8049e48cb3d8596e6f5aba : SUCCESS in 1m 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b4a2db37d0a4323824266ff3a9f1cdd : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":2},{"id":"295d03675b362deff29d81a5e4c734a6c38a13b7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-03 19:30:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8c092ac5c6273f400d8b488f0fcce756ca15eaa3","revisions":{"29e10a783fe71f8d318b238f8ae1816dcddb45d9":{"kind":"REWORK","_number":1,"created":"2021-10-05 13:55:38.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/05/812505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/812505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/812505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/812505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/812505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/812505/1"}}},"commit":{"parents":[{"commit":"1255eec797d324d68ce0028e7a9f74d11bca0fe9","subject":"Merge \"Remove neutron api-paste.ini from old location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1255eec797d324d68ce0028e7a9f74d11bca0fe9"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-10-05 13:55:26.000000000","tz":120},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-10-05 13:55:26.000000000","tz":120},"subject":"Remove Influxdb data source Grafana plugin","message":"Remove Influxdb data source Grafana plugin\n\nGrafana has now native support for configuring Influxdb data sources.\nPlugin has been defunct anyway since Grafana requieres it to be signed.\n\nChange-Id: I461f598bca6606f1d2939e385bb7befe2b792aa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29e10a783fe71f8d318b238f8ae1816dcddb45d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29e10a783fe71f8d318b238f8ae1816dcddb45d9"}]},"branch":"refs/heads/master"},"8c092ac5c6273f400d8b488f0fcce756ca15eaa3":{"kind":"REWORK","_number":2,"created":"2021-10-05 18:00:49.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/05/812505/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/812505/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/812505/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/812505/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/812505/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/812505/2"}}},"commit":{"parents":[{"commit":"1255eec797d324d68ce0028e7a9f74d11bca0fe9","subject":"Merge \"Remove neutron api-paste.ini from old location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1255eec797d324d68ce0028e7a9f74d11bca0fe9"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-10-05 13:55:26.000000000","tz":120},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-10-05 17:58:38.000000000","tz":120},"subject":"Remove InfluxDB data source Grafana plugin","message":"Remove InfluxDB data source Grafana plugin\n\nGrafana has now native support for configuring Influxdb data sources.\nPlugin has been defunct anyway since Grafana requires it to be signed.\n\nChange-Id: I461f598bca6606f1d2939e385bb7befe2b792aa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c092ac5c6273f400d8b488f0fcce756ca15eaa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c092ac5c6273f400d8b488f0fcce756ca15eaa3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
