)]}'
{"id":"openstack%2Fkolla~812671","triplet_id":"openstack%2Fkolla~master~I42546cfba2be5b94fa57728a60f2925ddf0e7901","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I42546cfba2be5b94fa57728a60f2925ddf0e7901","subject":"DNM: mariabackup - test pinning mysql collection","status":"ABANDONED","created":"2021-10-06 12:18:33.000000000","updated":"2021-10-06 13:53:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3e4a803c5e381ca88b531d4555111053e497aabd","_number":812671,"virtual_id_number":812671,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-06 13:40:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-06 13:40:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"939840dcd7eece81e4cc5d1afb3955b0d21c93d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-06 12:18:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f55a280ab4d0d39185219b06d71ca22749f89d2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 13:40:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/01ea9a93da1547d6bbf0cb6645d6cff4 : SUCCESS in 31m 40s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/35c0e4cb2bd24b6f86d0d6fea2f57c70 : SUCCESS in 37m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/cea78bfbb35740ff9909c74771831de2 : SUCCESS in 1h 05m 14s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2fc5ab067e6247319ee04c2fd191df4b : SUCCESS in 1h 08m 44s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/835cc1eb45bf4e8bb62db6eafd330d67 : SUCCESS in 1h 16m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/51d708870cb64b7ca745d96a334a2825 : SUCCESS in 27m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5e5188fc611543e4bef6433e584f509a : SUCCESS in 18m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ef5b79c68b104a4a897a1fe181de415d : SUCCESS in 56m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00ea49c68dd9409b86d2a2c1c76a8165 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e5d88ab8dc64d5ab65ffdc0f8dc11db : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/618b1786e32c4d2b9593a8c7d5d93ed2 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e2a8788218549fb9f9b72658fee1da0 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86ac90c64caf48a58d2517052fab99cd : SUCCESS in 7m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5c36272e08bf4a6184d04a2e8f5f3262 : SUCCESS in 19m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5eb743122a6b41c39da081edee6776dd : SUCCESS in 30m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9acaf097dbd04cc18d026c44d3ab23bc : SUCCESS in 53m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f306f962f1284d3b8bf86df3b59ba2f5 : SUCCESS in 53m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3cd179f7b7334224ae192dbfd6e79faa : SUCCESS in 1h 11m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3e4a803c5e381ca88b531d4555111053e497aabd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-06 13:53:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ae3c5030bc352bf36c2af761b01f2a1a7d6679e","revisions":{"4ae3c5030bc352bf36c2af761b01f2a1a7d6679e":{"kind":"REWORK","_number":1,"created":"2021-10-06 12:18:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/71/812671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/812671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/812671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/812671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/812671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/812671/1"}}},"commit":{"parents":[{"commit":"1255eec797d324d68ce0028e7a9f74d11bca0fe9","subject":"Merge \"Remove neutron api-paste.ini from old location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1255eec797d324d68ce0028e7a9f74d11bca0fe9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-06 12:18:25.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-06 12:18:25.000000000","tz":120},"subject":"DNM: mariabackup - test pinning mysql collection","message":"DNM: mariabackup - test pinning mysql collection\n\nChange-Id: I42546cfba2be5b94fa57728a60f2925ddf0e7901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ae3c5030bc352bf36c2af761b01f2a1a7d6679e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ae3c5030bc352bf36c2af761b01f2a1a7d6679e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
