)]}'
{"id":"openstack%2Fkolla~812730","triplet_id":"openstack%2Fkolla~master~I7c192bb906ade6cc600a3d0f7ae91b1d3111e802","project":"openstack/kolla","branch":"master","topic":"kolla-xena-release","hashtags":[],"change_id":"I7c192bb906ade6cc600a3d0f7ae91b1d3111e802","subject":"Use RDO\u0027s Xena Delorean packages","status":"MERGED","created":"2021-10-06 16:49:49.000000000","updated":"2021-10-08 20:25:31.000000000","submitted":"2021-10-08 20:23:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812730-kolla-xena-release","meta_rev_id":"b1105dfc8ffbe22850db90baba7a962dfecf4936","_number":812730,"virtual_id_number":812730,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-08 20:23:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-10-07 09:28:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-10-08 11:48:03.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2021-10-08 11:50:45.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-10-08 11:48:03.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-06 18:22:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-07 09:28:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-08 11:48:03.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-10-08 11:50:45.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"25b947f0d3b94cceb8b8583a62d0c5977b4612c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-06 16:49:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c5e5bf4632362dbac6f92560c9dbf5a5ce0609c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-06 16:50:06.000000000","message":"Topic set to kolla-xena-release","accounts_in_message":[],"_revision_number":1},{"id":"ac0ef2718366b6fe43222d51ca6b40688ae33078","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 18:22:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fa9ef23636fa44c7865761d56f4744ea : SUCCESS in 32m 16s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/463822cffe7044029ba64f36070878b6 : SUCCESS in 33m 00s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3da970e0f6c0492dbe211fd166bff7ce : SUCCESS in 1h 18m 18s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/54a122eb9e0c42c2af08fa5936845e4a : SUCCESS in 1h 09m 48s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/51c7e8536c354fa1a8a258d772620d05 : SUCCESS in 1h 30m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/76c30bc5bf4e46349bf039a02c839080 : SUCCESS in 37m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0413ca4ba3974a398231dcd1a20d02fb : SUCCESS in 19m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/14cbf62369f64946805a93ccf41399d7 : SUCCESS in 47m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f99b6f5237a4da7878cd454e5618b30 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4c1d6b7ed6f448097dbe01fa6c84b3d : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77085296912b4953843aff9108f081fb : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/206867a324bc4d598ae8eca3f6766cd4 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24c3810f756347cba792b7668889f87a : SUCCESS in 6m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6750f53d50a443f5a12fef075b473af0 : SUCCESS in 19m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e649059dfde1402b82806330836b17b6 : SUCCESS in 32m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e6e6c9d4ff2b48e59d7955ebf18219cd : SUCCESS in 55m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/61cb207ede98462cae4ffef83d2baa6e : SUCCESS in 50m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8018131f771443ccaa794e82edf64993 : SUCCESS in 1h 07m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ff46eb2e67196c39157b09afa32c7dddbc7479b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 19:51:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6a4a5181c6b24794a3a19b71908379f5 : SUCCESS in 2h 32m 30s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c173ac354ef34d54b6f69b1867cb48b0 : SUCCESS in 2h 17m 14s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7cc73187094f46629077e8129ad70399 : SUCCESS in 2h 19m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3cf596090c442588735ee51967e31c0 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/44f23265422f4c40a4e316973f21996e : SUCCESS in 6m 55s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d565ed9f47ae4d05a91ca5eb0ff8e3be : SUCCESS in 1h 54m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2035b6409b38e3c019fde27c281750cf1c98d4af","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-07 09:28:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"98e7598667f18ff127448b9c07ad0e4e01eaec13","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-08 11:48:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"278ac1e0b747382d7bc1c123c9bd67745fa72fe6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 11:48:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"afc053cdc87a9eeefa662508bc99e186bcc034d2","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-08 11:50:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ad43dfafe15434f54566d106c8fac342b7a0e76c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 15:25:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/50f2bfe08580490299c10055a8915a7f : SUCCESS in 52m 00s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/dabb6edfdf0a413db55a4805a1f0fdf6 : SUCCESS in 1h 19m 38s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4cc05e18b522456f9cf87278ccd6822d : FAILURE in 1h 14m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd644c17470e4c73878df7666594422d : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f8a84a9ffdc4c9f86f02627b9957065 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04278b5420d94d27b07afa3e16162cb6 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9433e5c03aa44940b705d33e55c96b4f : SUCCESS in 10m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c1f5f9008eb48cfa60c87eeae87f97c : SUCCESS in 30m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/059cfcb7dd524f2eb854ccb7cecdd495 : SUCCESS in 58m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8e3ea8dc71a6415ebb783baa958be9e2 : FAILURE in 54m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e3dd0957bd95ebe069d3fdb2158a5e1f9c1588bd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-08 17:36:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b82bc027f6935d63ecdface0005f450f9c5ba74b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 18:55:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1a1f1e8494de4b7db58c82b5119de46a : SUCCESS in 41m 44s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f164210dd4f844299276b0b8c2c3616e : SUCCESS in 43m 39s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6d0f7d5462d44dd3a64b8e2ebd431e2b : SUCCESS in 1h 13m 26s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/5a1071523561480b9c85e8cfca92243c : SUCCESS in 1h 06m 03s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b5267d2fc3f64eb7ac26b4ebadaa66d8 : SUCCESS in 1h 16m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bd5beb5a8f164e6db20811a7c016a9a6 : SUCCESS in 22m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0492787c7f73435d88b75c098a7cf908 : SUCCESS in 17m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6708e6baac094b66990b6ba7add5867f : SUCCESS in 1h 04m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe4a574db53d46a6befa45359a47add4 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1a6a9b4cabe4738bde44decd441c7ef : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e0703094b2643379b14bcf8d13aa179 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c02d84850feb406dbde84c06f9d7fa13 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92abc9cede824b20aa081422f9ea5601 : SUCCESS in 6m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f65485b3c03542f39f8c2ad22b8888de : FAILURE in 20m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d93078f02aae4e4da47bd76fe2f9c108 : SUCCESS in 28m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6252465b887f4c16882645093346aa95 : SUCCESS in 59m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/79a692fea8dd4fd0bfca749aebac3578 : SUCCESS in 53m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bdc1da1407fc42289bf5e6425116bdf7 : SUCCESS in 1h 05m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2531b2318c4c0417eac824a888aa1703fa879717","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 18:56:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73457283d8538dbfa9981455890a693588c25d9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 20:16:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3a8824867c374543bf7d419b650f333f : SUCCESS in 2h 06m 26s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/044c5454cabc46528913026aebe33b19 : SUCCESS in 1h 24m 04s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c5e043bbb78f434194e0714f5e2c3182 : SUCCESS in 2h 18m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed7300bac3964f948eae2e6f7453fe13 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e3824bb5c8804dc6a379045c958c2d0f : SUCCESS in 6m 20s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d869e4ad0f5c428abf778dec7cc4fbc9 : SUCCESS in 1h 21m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab27dbcf4aa91a0427a8b4a8c4ff0fb2beb93282","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 20:23:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/998a9913cedb4e3b8e78db221e35bb93 : SUCCESS in 37m 51s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c7c2ebd8e9ee4ec59be6a79ac98a4b84 : SUCCESS in 1h 03m 00s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ddc37159ac634f59aea3c15d966f3103 : SUCCESS in 1h 25m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4856bab5e72e404ab3d690a4fbd918c5 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd33b7860699493fb054e5420173c4a4 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb8e377928e94c06a41b94fba678b636 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad7787a8c7a74f02b530c84076dd8f9a : SUCCESS in 7m 02s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/85c31d55562f4b42aa7826c41f8ffecb : SUCCESS in 28m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/89153951135a4d69a20a082d44fb4ea8 : SUCCESS in 54m 56s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4fb555dabf0d458fa172ed0955b209f7 : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":1},{"id":"6c4531599f69a2b773c7b832cf25138af7277f0b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 20:23:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b1105dfc8ffbe22850db90baba7a962dfecf4936","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 20:25:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f1d62b23a784355b8aa8ebffe572b3d : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d640980e07c6bfd30611914cb305c19ab282b546","revisions":{"d640980e07c6bfd30611914cb305c19ab282b546":{"kind":"REWORK","_number":1,"created":"2021-10-06 16:49:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/30/812730/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/812730/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/812730/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/812730/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/812730/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/812730/1"}}},"commit":{"parents":[{"commit":"7652ef325e0e84fe92ccf104bfe751a4d3542227","subject":"Merge \"[bifrost] Drop workarounds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7652ef325e0e84fe92ccf104bfe751a4d3542227"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-06 16:49:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-06 16:49:47.000000000","tz":120},"subject":"Use RDO\u0027s Xena Delorean packages","message":"Use RDO\u0027s Xena Delorean packages\n\nThis is a temporary solution until centos-release-openstack-xena is\navailable.\n\nChange-Id: I7c192bb906ade6cc600a3d0f7ae91b1d3111e802\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d640980e07c6bfd30611914cb305c19ab282b546"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d640980e07c6bfd30611914cb305c19ab282b546"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
