)]}'
{"id":"openstack%2Fkolla~813008","triplet_id":"openstack%2Fkolla~master~If31b08fcf45bafc0ab25498ece9208a576aae41b","project":"openstack/kolla","branch":"master","topic":"logrotate-schedule","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2021-12-02 06:44:49.000000000","reason":"Removed by Dr. Jens Harbott (frickler) by clicking the attention icon"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-12-20 12:40:02.000000000","reason":"Mark Goddard replied on the change"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2021-12-21 18:46:42.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-12-21 18:46:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If31b08fcf45bafc0ab25498ece9208a576aae41b","subject":"Make cron\u0027s logrotate schedule configurable","status":"MERGED","created":"2021-10-07 13:45:55.000000000","updated":"2021-12-21 18:48:53.000000000","submitted":"2021-12-21 18:46:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813008-logrotate-schedule","meta_rev_id":"9cde64fb7f108859c7e4fa9a1636d2d3824df270","_number":813008,"virtual_id_number":813008,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-21 18:46:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2021-12-08 07:23:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":1,"date":"2021-12-20 16:29:22.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2021-12-21 16:25:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-12-13 11:50:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2021-12-20 12:40:02.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-03 02:30:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2021-12-21 16:25:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-07 14:52:18.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-10-07 14:52:18.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-10-07 14:52:18.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-10-07 16:14:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-12 09:21:09.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-11-12 12:16:32.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2021-11-12 14:13:21.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-12-03 02:30:19.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2021-12-08 07:23:59.000000000","updated_by":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"reviewer":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"state":"REVIEWER"},{"updated":"2021-12-13 11:50:54.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"bd2275731ac5e796e628086227ad199694df02e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-07 13:45:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9dfa75f9d894f00eefa1a2adb18933a71594829","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-07 13:50:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"49baa8c0c0a4fa9a8347c3830af6a4ddcea96385","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-07 13:56:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"89348df84e774dd212695a77205b51b83ebe9796","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-07 14:46:31.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"0c4edde284eef9760c45e11d1b221e3e00cf27db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 16:14:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3136d052199f42d787f581240be8bba5 : SUCCESS in 57m 06s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9292232e9a8040c783d43fa868f4f328 : SUCCESS in 1h 05m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/dbc75a79499d459c90a0932e7dd20596 : FAILURE in 1h 09m 21s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a555a3f92ec54472b9874892d3e4e431 : FAILURE in 56m 24s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5ffc168ea5f94d2988e51a1def13e191 : FAILURE in 1h 25m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/777e459e608b45d89a8978a0e1f549c4 : SUCCESS in 30m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4a8563c8c08c4c83b5c15b632658c738 : SUCCESS in 21m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f1dba303b82d485794a6b4bc2d24c4a8 : FAILURE in 49m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dc04341acb7409bb91ced9985e0f6b2 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/560bad485b314d07906accecee875c50 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b71bf61408a74e0d8af4b6bf95b55a6f : SUCCESS in 9m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5968bf12c3444638b4ce46050ab797d : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2653d04a07c4d4a98b9cf5352db3565 : SUCCESS in 8m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b3ea2a865dd24175a0f3b99b421d3c28 : SUCCESS in 22m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/11479e91b5af47888e3279022b38e35f : SUCCESS in 36m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ab7d8b24ffa44e01af3448dd58e3f62a : FAILURE in 42m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/11ae9ded7b5d437d8acf14faa5bae0bb : FAILURE in 42m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/315aa087de9243668ec17755dcc47d7e : FAILURE in 1h 13m 33s","accounts_in_message":[],"_revision_number":4},{"id":"06597b60f8b78141d4c185cbccd724f967984902","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 17:13:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/416d54660d864507ad91416344ad2b47 : SUCCESS in 2h 07m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/609ee43088a3490e8236ff976e30298f : SUCCESS in 1h 57m 10s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8156e8f1073d4865ab01e5bdb41a37b6 : FAILURE in 1h 47m 40s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7d9d8015a864183a5a69c9846e296e7 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9268e79f713429d9b0e6ce59471299a : SUCCESS in 6m 27s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ec15f8063b744cbca0942ed20da889b4 : FAILURE in 1h 25m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"66750e94bfe7861bfa44b839db8918802e22e7d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-07 17:49:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e83893127a71864d43e1366c69b4c12392823046","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 19:38:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7c0e0a416ae44ba1a8a6a984ece73fb5 : SUCCESS in 32m 24s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e803526cbe714961af24b13fb4de8aa4 : SUCCESS in 45m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/cd31ccd23e87449ab784986ff109376e : FAILURE in 1h 07m 51s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/377bd55e2b12461dbb2c556d2e9c7293 : FAILURE in 57m 05s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f4ceb2af2eeb4b3bb22d891c7988f1c2 : FAILURE in 1h 40m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7e2f5bf8f52c48eaabdabcf4bfb18748 : SUCCESS in 21m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bf444a3ea9c54e3f9ae966106a1a432d : SUCCESS in 19m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/36e3965493494a82a881a56200bbc875 : FAILURE in 42m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03a968c7eaae45fab27e979c18ba691a : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdab2398eb744134b417a4f03d05229d : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e161e5eb5a24cb895b54a4fe8b0f7ad : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ea83f669d1e47b2875712cc0c8cd2ba : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ab379c54c844e49982d021ef28dfde3 : SUCCESS in 5m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/238e3ffdab174785a726d00098847c5b : SUCCESS in 20m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b2274e9a383244f2ae235a9f34d1015d : SUCCESS in 32m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/07c6b505bd354e62ab2abbbacaca8832 : FAILURE in 40m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8e72ec85bc104a4f8f510e17d94a5657 : FAILURE in 34m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a8b90a70ecf64951ac5d53e078a79ac3 : FAILURE in 1h 07m 44s","accounts_in_message":[],"_revision_number":5},{"id":"1d0ba08758f0535172e6494befc4b31ab0d284f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 20:08:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/32a5a4a6b3e945288454db6c480022b6 : SUCCESS in 2h 16m 02s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/62a71bf987d94981be231dc45c2f6c92 : SUCCESS in 1h 51m 58s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/44a73a1f0b5c4b71bdf65b09137ceaf6 : FAILURE in 1h 42m 01s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/200f86bca2794f5b9d71a5b7e4700803 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/38d63a50b8d441df9570349115d227f9 : SUCCESS in 6m 21s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7202d788058a48b39aa86d3975d28736 : FAILURE in 1h 17m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d3cb0bdcfc483f921ca11d167acbc419ca47b27f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-07 22:32:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e6a201ccd6ed40d345a37a4905d5f69a0698fdd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 00:04:22.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d97e09d04dca40a1acd6d99c396fd27d : SUCCESS in 42m 16s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4ddec78872f74199b6bae28886c48bcf : SUCCESS in 49m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6a6f0ea6375a4bc6ba6899450a171695 : FAILURE in 1h 11m 49s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/39328ad612fa400b9f288b09e7489f2c : FAILURE in 52m 56s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/942a40f796f948cbad4b0babe9edc72f : FAILURE in 1h 30m 23s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d0ceb6db34a64c29b03e356a3a76203e : SUCCESS in 28m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0d16a93d98854ae98a99c50b7b72eaf6 : SUCCESS in 17m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/db763fd949cd43158fd1ac600ff2ca32 : FAILURE in 46m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87d6797ce8c34120bae2f6b28c66a200 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4803ac9cd118451990742fe7b0de841c : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f004b2befbc24f9e94bd05976656b839 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8bec2f589d8041ba8bd0c408bc7a716b : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/816cdea958744d04870a3807fe00b1c8 : SUCCESS in 6m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/94f70ba0ece84dfba67afcf7faf4c15b : SUCCESS in 16m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dc04cdff917c4a869cfb83b56eca8895 : SUCCESS in 25m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3a2653a9a01d4dbca393a3e380b5e256 : FAILURE in 39m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cbe08f33374d4c88aeeaca90b454c88c : FAILURE in 33m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/10addb7863ab4520af15a581ad3b140c : FAILURE in 1h 04m 01s","accounts_in_message":[],"_revision_number":6},{"id":"35f2fc4b188f2ef93176c930c78397132cc60009","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 00:34:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/90be7438b03946e8b82da5d00aba9a22 : SUCCESS in 1h 59m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/67582535dde54c2e8a6d74f7ac85be8f : SUCCESS in 1h 26m 51s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1bca0dc1f11143d8abb43c8be261d2fe : FAILURE in 1h 50m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e25699337514a90b0c6499f312f0d05 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9770f684db91436682aac02559518e42 : SUCCESS in 6m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3c711b6177134b02a5c2bc1494a90707 : SUCCESS in 1h 32m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ce696339925c2e30b3f88c6057d0199ae0dc0224","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-08 11:05:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"cf17e3d69c116889f668863aa5d57358965a4cf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 12:41:39.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3eaf321b7a52408ea55214f59b6c4369 : SUCCESS in 27m 33s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/25d906a812504feb86382841de917f21 : SUCCESS in 35m 22s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4729f8192dbe4877bbf22c1c1088892e : SUCCESS in 1h 10m 20s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/1bcf93c622f84d6bbc0070644fba4342 : SUCCESS in 1h 12m 16s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/bdce526cb0f045d68cfd988ba4c9d5a2 : FAILURE in 1h 29m 54s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f1fa6f94fcdd41cb93827d0c5e734816 : SUCCESS in 28m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/046e6fe05bf54efaba202bbd57206a02 : SUCCESS in 19m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ba43d4b5ced04a48af8625df37a1d39a : SUCCESS in 1h 01m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4328132c58b4c0b9ad2e22f716e9664 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60dc68e1855749dbb1e92f3fceb0b598 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cca08bca0f3b44a8b62e788632b38602 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/003561d48706473c88f2ed841b401f26 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/212a3dd3808643759d07706cb68b7648 : SUCCESS in 6m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d5c94b0e4557401da0aa7c2b06d6fc2a : FAILURE in 20m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d243141f3dc5456aba3d1ccf9b649f66 : SUCCESS in 29m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/21954aa0731c4835b466b589274db7af : SUCCESS in 56m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e86beb3aba694be59858db44f2030ebf : SUCCESS in 56m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b0d9e71bc2c14629b39d82c08d16d893 : FAILURE in 57m 57s","accounts_in_message":[],"_revision_number":7},{"id":"b9905e3ba5b01f7f7f45d5b00798c49e0f7bfad1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 13:26:44.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a30c39368b974e9e861eb09532f50c69 : SUCCESS in 2h 01m 11s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4acadce563ee43ec90f51c533b683436 : SUCCESS in 2h 10m 54s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/21a5a6f8ff8e465689a6fd84b719c043 : SUCCESS in 2h 18m 10s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5e7f178f13a4bbab974901b77917b3e : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4a93304906404d8ba815c06727322f2a : SUCCESS in 7m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7d69948235fc4a478141f1cbc86e9c37 : SUCCESS in 1h 31m 31s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4d46293075da69b030389243eedf86bd4f573c94","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-10 11:55:23.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"5dd5a2352f465846440b0db3be9ac31ee9166b9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-10 13:18:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/45733bfbe10247c296450c2ae6e360ae : SUCCESS in 24m 13s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/46a4153e5833466da252631b8c21000b : SUCCESS in 38m 09s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3c5446866f914dbeb38b83e99c305a2e : SUCCESS in 1h 03m 29s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e4e5e13208d74ca4a91679ed0536bae6 : SUCCESS in 1h 02m 22s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4e885c60cb7e48e39f80fc313266db64 : SUCCESS in 1h 13m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/37aec9bdb4154b139f551ba18217ace2 : SUCCESS in 26m 26s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c9971a138c13427187f4769a60032ae7 : FAILURE in 17m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b9285100cd1543409870a98d938f8402 : SUCCESS in 53m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43bf50336b8545329552446482baa012 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/989e685403d54f8dba8f631bf08c26a1 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08b76c16c71343d1992f3de420b9d9d2 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8c7fb05b75f409a84273d248403933f : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44b5b97557c0408fbf1601f766b576c9 : SUCCESS in 7m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ddc5ddb1e8f842a9941160765bd53e8a : SUCCESS in 30m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ae6d11b505ce49e7acf2b866aa1e966d : SUCCESS in 27m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/87ebf2cc1907465794121146d023d625 : SUCCESS in 47m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf595862b6574f61af0e8c22866d36e0 : SUCCESS in 54m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7a7c1e035662411a93f4580d179f6bc7 : SUCCESS in 50m 06s","accounts_in_message":[],"_revision_number":7},{"id":"eaf64436e9c616af62d6ee9d285d8c36f4bb9ca2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-10 13:43:46.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7b7b094eb0b64f159bdf51b387113fba : SUCCESS in 1h 38m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/367b64b6ae9741a4b7faa2aa1840aa03 : SUCCESS in 1h 27m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/08eed0c6f95046019cff394af450b1f5 : SUCCESS in 1h 46m 08s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67cfa991e2064707abfcb5f5845fd116 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/afbc872cc36541efa4ac59bb9178b70b : SUCCESS in 5m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6c099cfba5fb405891f6b2fd43016a4f : SUCCESS in 1h 35m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2f6402a25d328c6692eb1f6bb0a24ca0dd1f086e","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-11-12 09:12:04.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"32d7c514f039f1016e750d4307cf0b00941f06ea","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-12 09:48:04.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"640f5ab0f478d8349015eb3285627d29175af40a","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-11-12 10:23:33.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"119ff9435c5cecc9ef7307cedeadc417ada6e18c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-12 12:16:32.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"1bb76f3b7b22e943ef0db56cf81f5d27e41d4d17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-11-12 13:47:22.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7bc28d51e0a508b0f07aef2e3429ae8cdf1155cb","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-11-12 13:50:17.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b8efbca130b4683c8896338df83d7884d5fc6675","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-12 14:13:21.000000000","message":"Patch Set 8: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"0488289369701d4647a31209d2c1f0520378487c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 15:10:52.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/eeb8c717a466454abbf5269d0e8cfdb9 : SUCCESS in 32m 34s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2a91ae0d010a487aadc12d5e1274d21f : SUCCESS in 40m 26s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/67fb9be86292455d8878f3af9b0db126 : SUCCESS in 1h 09m 00s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/464791646be74236af37c989125429bf : SUCCESS in 1h 02m 32s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1978665c89e14bb49484abd149bc0bce : SUCCESS in 1h 17m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/640ff682616d420c84cb799cfce0c238 : SUCCESS in 32m 19s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a61d6364c57c4b1c9e2cb98dd75510cb : SUCCESS in 16m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e34fdd9fc04c4e77bfe120d876db2789 : SUCCESS in 58m 42s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d122b07904274dccb7bc5f4873a929e1 : SUCCESS in 1h 03m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7103d05ac9d4657b04fcf5e8ea4fe23 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3dec55f758043b695bc055b06212163 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af0ff3b20f0244dab1e48363a5cf6125 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9649e6489eaf499ab82ec604ac594fd7 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1685b301dc20444eb9fa14bb4cfa59e6 : SUCCESS in 6m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/152c129d510641c8ae3f186c700d85f4 : SUCCESS in 17m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/049402d61566473da4124ee46f7e9dc5 : SUCCESS in 34m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a2465c59154b4e648aaa7fdcdd3167dd : SUCCESS in 48m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4b25b06fd5344a8b941691e261df45cc : SUCCESS in 58m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/508a98a10db14ad0a5f46e7043371708 : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":8},{"id":"8ea3b69e23334bb38028e5b92c66878b0d6036bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 15:28:57.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0c3232ac21034429a378ade476083d7a : SUCCESS in 1h 24m 37s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bdd0ccb73adc4dc0af0ede89fa96edb8 : SUCCESS in 59m 24s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9932462813304777a1ca459fe388df5c : SUCCESS in 1h 37m 50s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/58d03c37953740ccad17b6e0b2642f12 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c59b2d0733f14e5c9bbb48cef7960bda : SUCCESS in 6m 21s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b989bc9b5dd44d59bf3b5ea485e17154 : SUCCESS in 1h 11m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"087f96f52db738a1fd3709a1decd6a4197280f38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-11-12 16:30:30.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ab45314c0744b996fa1aae0195e46dcaeb135c2b","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-11-12 16:30:42.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"43cec314dcb44613fd19e87b8f75ac115d2289ec","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-11-12 16:33:52.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d83a7bcc1b9e031b8528a2f8ae3fe5d4812e4e94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 18:02:21.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4e06e6cc66f2405fa0133ee755aa0e75 : SUCCESS in 58m 15s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/856a9164c8d34e3ca03e474e17f43738 : SUCCESS in 1h 11m 17s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a1e5d435740a476dbaff6268e1204a76 : SUCCESS in 1h 19m 04s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e68f8e3f43843648c495529a842187f : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f514977528394648946258d610a7f493 : SUCCESS in 7m 42s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/88292fdb4ffa4388bf6f6b22541bd963 : SUCCESS in 55m 02s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2b510ed7fc6ddfceb2f1c3f421ef18254a5da582","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 18:07:29.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/445633ceea1b4cc69c0b8666fc4ce6ae : SUCCESS in 28m 06s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fedd5108a88e4e84b849bc596b781c6f : SUCCESS in 51m 45s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8f5a2089c0fb45ffb71f65b438338034 : SUCCESS in 1h 04m 21s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/91916e3520394166b3fe5493006e7975 : SUCCESS in 1h 19m 29s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d7aaad1139444fdda50bf85fb8469e52 : SUCCESS in 1h 31m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/685448a1ac51467999e2fd6ca69d579d : SUCCESS in 35m 44s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4ebaa7567dfd454ab63fa5d8827673a3 : SUCCESS in 22m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e8b3d7b1c0544ab589219d9e903d80d7 : SUCCESS in 54m 31s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d40cb392ad34492484677e5e74b3adf3 : SUCCESS in 1h 08m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d27ce8d7284641499ce9ec1bbf3109d2 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ed792e8d88540d2a37c699b8f91e124 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9771c68661364bf4933ad36d7dd40e87 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/690ce92585604860bd3fbd20f9beef61 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09b9f062fad545218c1e022a28fece1b : SUCCESS in 8m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6c533a1642dc4f31b991d95691919a6a : SUCCESS in 24m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bbf717bce3854caf8870e8132e41e50b : SUCCESS in 33m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9160c8649a6840659b93c257c5e001e4 : SUCCESS in 1h 01m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/db75246eff5647cba77955ec3145cbbd : SUCCESS in 56m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a0cfb78e2797411eaa759cfa12538160 : SUCCESS in 1h 07m 40s","accounts_in_message":[],"_revision_number":9},{"id":"203aefa562e2f5b8916998c8fa560ab8cb47a8bf","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-13 07:05:02.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"052c44b2bba4dff3722eca12616c4d8c826d61ea","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-11-13 08:06:47.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5e8134b7ccf239dd2a4d66cf3bda67026d0b4d58","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-11-21 09:40:32.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"781d33884d150531103f9cb5a3dd4603d3f188cf","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-12-03 02:30:19.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"a06e5141edbc0114e55ee2470bcff4c5fc18e0f3","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-12-06 14:30:29.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"65a3a7f68d1540fcb60f5c2bd33d1acc8364e1a9","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2021-12-08 07:23:59.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1f09cd5a3ab78a0e8def1eab6d2d978795faa4ed","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2021-12-13 11:50:54.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"fb23794b6ab1b5cafae208cfe87e6afa1736ed30","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-20 12:40:02.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"cf85ebb9dc68968a9893cc5ebb54643c7bfcf49c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-20 16:29:22.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9b1411bdb5ffdf1b786ba5301c10e7fab9664e4e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-21 16:25:27.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"e83aebe5decc9afdda309222b60b62c1de80d617","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 16:26:09.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"817ecc3f706f541f87a32233ae5d1b303e8c7a14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 18:46:42.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b7ba4c2bc4aa49d78cd6b4f1a37fa0c3 : SUCCESS in 44m 33s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/68dc544b77c443bf8a4308a4aedbb8fe : SUCCESS in 1h 16m 01s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/19dfc46b857f4518b0d736cd758d6756 : SUCCESS in 1h 14m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cec50a39a3884230909cc7f213610da0 : SUCCESS in 35m 29s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/509d0335fd3a449ab6f35ba45675eaae : SUCCESS in 1h 00m 12s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9b0353b2042741afa5f5ebe7082baa41 : SUCCESS in 1h 11m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1ec6fe62d8a41959e01c5e626441c7b : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59d34b2367334e4abe08a89e8b9bf5c6 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b2097f69dae4635a4f05eee5f55dece : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0579cd324b264af9bc794b2fc65cd00f : SUCCESS in 7m 21s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a7d98d7c47a84b50a1a6f501cd7fe169 : SUCCESS in 31m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3922a05e96fc4778aafbaaf7ec3cc042 : SUCCESS in 55m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aa2ef47d04564eb69dc4cbe168681e72 : SUCCESS in 58m 32s","accounts_in_message":[],"_revision_number":9},{"id":"0fccec66963717b7b992cd567c2e8610901bc4bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 18:46:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"9cde64fb7f108859c7e4fa9a1636d2d3824df270","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 18:48:53.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a7bd397ffb44eabac8d78e62d306ec8 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"f229eb7e06173649974376cf689d676c673f4988","revisions":{"8ad5a15c692d8170412ab814931ed3f2feba9a2a":{"kind":"REWORK","_number":1,"created":"2021-10-07 13:45:55.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/08/813008/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/813008/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/813008/1"}}},"commit":{"parents":[{"commit":"7652ef325e0e84fe92ccf104bfe751a4d3542227","subject":"Merge \"[bifrost] Drop workarounds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7652ef325e0e84fe92ccf104bfe751a4d3542227"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-07 13:45:02.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-07 13:45:53.000000000","tz":120},"subject":"Make cron\u0027s logrotate schedule configurable","message":"Make cron\u0027s logrotate schedule configurable\n\nKolla-ansible allows user to change logrotate.conf,\nproblem is that logrotate cron file is running\ndaily by default in all distros, this is sometimes\nnot enough. So, this patch is adding script and\nway how to reconfigure schedule of logrotate job\n(cron file).\n\nChange-Id: If31b08fcf45bafc0ab25498ece9208a576aae41b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ad5a15c692d8170412ab814931ed3f2feba9a2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ad5a15c692d8170412ab814931ed3f2feba9a2a"}]},"branch":"refs/heads/master"},"c7588da30ebddf5c25327ef4e5b9c1623d5ef4e3":{"kind":"REWORK","_number":2,"created":"2021-10-07 13:50:31.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/08/813008/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/813008/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/813008/2"}}},"commit":{"parents":[{"commit":"7652ef325e0e84fe92ccf104bfe751a4d3542227","subject":"Merge \"[bifrost] Drop workarounds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7652ef325e0e84fe92ccf104bfe751a4d3542227"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-07 13:45:02.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-07 13:50:14.000000000","tz":120},"subject":"Make cron\u0027s logrotate schedule configurable","message":"Make cron\u0027s logrotate schedule configurable\n\nKolla-ansible allows user to change logrotate.conf,\nproblem is that logrotate cron file is running\ndaily by default in all distros, this is sometimes\nnot enough. So, this patch is adding script and\nway how to reconfigure schedule of logrotate job\n(cron file).\n\nChange-Id: If31b08fcf45bafc0ab25498ece9208a576aae41b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c7588da30ebddf5c25327ef4e5b9c1623d5ef4e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c7588da30ebddf5c25327ef4e5b9c1623d5ef4e3"}]},"branch":"refs/heads/master"},"7685ed6f849e70520faec40fbea43da6cee93fcf":{"kind":"REWORK","_number":3,"created":"2021-10-07 13:56:15.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/08/813008/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/813008/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/813008/3"}}},"commit":{"parents":[{"commit":"7652ef325e0e84fe92ccf104bfe751a4d3542227","subject":"Merge \"[bifrost] Drop workarounds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7652ef325e0e84fe92ccf104bfe751a4d3542227"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-07 13:45:02.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-07 13:56:08.000000000","tz":120},"subject":"Make cron\u0027s logrotate schedule configurable","message":"Make cron\u0027s logrotate schedule configurable\n\nKolla-ansible allows user to change logrotate.conf,\nproblem is that logrotate cron file is running\ndaily by default in all distros, this is sometimes\nnot enough. So, this patch is adding script and\nway how to reconfigure schedule of logrotate job\n(cron file).\n\nChange-Id: If31b08fcf45bafc0ab25498ece9208a576aae41b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7685ed6f849e70520faec40fbea43da6cee93fcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7685ed6f849e70520faec40fbea43da6cee93fcf"}]},"branch":"refs/heads/master"},"656d7eaa8ca0e51da5942641e5b4cda70dcf9206":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-10-07 14:46:31.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/08/813008/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/813008/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/813008/4"}}},"commit":{"parents":[{"commit":"7652ef325e0e84fe92ccf104bfe751a4d3542227","subject":"Merge \"[bifrost] Drop workarounds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7652ef325e0e84fe92ccf104bfe751a4d3542227"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-07 13:45:02.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-07 14:45:50.000000000","tz":120},"subject":"Make cron\u0027s logrotate schedule configurable","message":"Make cron\u0027s logrotate schedule configurable\n\nKolla-ansible allows user to change logrotate.conf,\nproblem is that logrotate cron file is running\ndaily by default in all distros, this is sometimes\nnot enough. So, this patch is adding script and\nway how to reconfigure schedule of logrotate job\n(cron file).\n\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/813039\n\nChange-Id: If31b08fcf45bafc0ab25498ece9208a576aae41b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/656d7eaa8ca0e51da5942641e5b4cda70dcf9206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/656d7eaa8ca0e51da5942641e5b4cda70dcf9206"}]},"branch":"refs/heads/master"},"c66a00cbe1e10b1e48f430b02c2ac47348a63d83":{"kind":"REWORK","_number":5,"created":"2021-10-07 17:49:04.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/08/813008/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/813008/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/813008/5"}}},"commit":{"parents":[{"commit":"7652ef325e0e84fe92ccf104bfe751a4d3542227","subject":"Merge \"[bifrost] Drop workarounds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7652ef325e0e84fe92ccf104bfe751a4d3542227"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-07 13:45:02.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-07 17:48:57.000000000","tz":120},"subject":"Make cron\u0027s logrotate schedule configurable","message":"Make cron\u0027s logrotate schedule configurable\n\nKolla-ansible allows user to change logrotate.conf,\nproblem is that logrotate cron file is running\ndaily by default in all distros, this is sometimes\nnot enough. So, this patch is adding script and\nway how to reconfigure schedule of logrotate job\n(cron file).\n\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/813039\n\nChange-Id: If31b08fcf45bafc0ab25498ece9208a576aae41b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c66a00cbe1e10b1e48f430b02c2ac47348a63d83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c66a00cbe1e10b1e48f430b02c2ac47348a63d83"}]},"branch":"refs/heads/master"},"50ad4283b4d5c3f4a54fbfc99faec1b870104afd":{"kind":"REWORK","_number":6,"created":"2021-10-07 22:32:28.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/08/813008/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/813008/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/813008/6"}}},"commit":{"parents":[{"commit":"7652ef325e0e84fe92ccf104bfe751a4d3542227","subject":"Merge \"[bifrost] Drop workarounds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7652ef325e0e84fe92ccf104bfe751a4d3542227"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-07 13:45:02.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-07 22:32:06.000000000","tz":120},"subject":"Make cron\u0027s logrotate schedule configurable","message":"Make cron\u0027s logrotate schedule configurable\n\nKolla-ansible allows user to change logrotate.conf,\nproblem is that logrotate cron file is running\ndaily by default in all distros, this is sometimes\nnot enough. So, this patch is adding script and\nway how to reconfigure schedule of logrotate job\n(cron file).\n\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/813039\n\nChange-Id: If31b08fcf45bafc0ab25498ece9208a576aae41b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50ad4283b4d5c3f4a54fbfc99faec1b870104afd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50ad4283b4d5c3f4a54fbfc99faec1b870104afd"}]},"branch":"refs/heads/master"},"4bcd95dbf7507bbfc6582d63eb8854d1dca94a50":{"kind":"REWORK","_number":7,"created":"2021-10-08 11:05:49.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/08/813008/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/813008/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/813008/7"}}},"commit":{"parents":[{"commit":"7652ef325e0e84fe92ccf104bfe751a4d3542227","subject":"Merge \"[bifrost] Drop workarounds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7652ef325e0e84fe92ccf104bfe751a4d3542227"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-07 13:45:02.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-08 11:05:48.000000000","tz":120},"subject":"Make cron\u0027s logrotate schedule configurable","message":"Make cron\u0027s logrotate schedule configurable\n\nKolla-ansible allows user to change logrotate.conf,\nproblem is that logrotate cron file is running\ndaily by default in all distros, this is sometimes\nnot enough. So, this patch is adding script and\nway how to reconfigure schedule of logrotate job\n(cron file).\n\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/813039\n\nChange-Id: If31b08fcf45bafc0ab25498ece9208a576aae41b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4bcd95dbf7507bbfc6582d63eb8854d1dca94a50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4bcd95dbf7507bbfc6582d63eb8854d1dca94a50"}]},"branch":"refs/heads/master"},"8c062e467987411e13d1a96b7846d8824e400b88":{"kind":"REWORK","_number":8,"created":"2021-11-12 13:47:22.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/08/813008/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/813008/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/813008/8"}}},"commit":{"parents":[{"commit":"7652ef325e0e84fe92ccf104bfe751a4d3542227","subject":"Merge \"[bifrost] Drop workarounds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7652ef325e0e84fe92ccf104bfe751a4d3542227"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-07 13:45:02.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-11-12 13:47:20.000000000","tz":60},"subject":"Make cron\u0027s logrotate schedule configurable","message":"Make cron\u0027s logrotate schedule configurable\n\nKolla-ansible allows user to change logrotate.conf,\nproblem is that logrotate cron file is running\ndaily by default in all distros, this is sometimes\nnot enough. So, this patch is adding script and\nway how to reconfigure schedule of logrotate job\n(cron file).\n\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/813039\n\nChange-Id: If31b08fcf45bafc0ab25498ece9208a576aae41b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c062e467987411e13d1a96b7846d8824e400b88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c062e467987411e13d1a96b7846d8824e400b88"}]},"branch":"refs/heads/master"},"f229eb7e06173649974376cf689d676c673f4988":{"kind":"REWORK","_number":9,"created":"2021-11-12 16:30:30.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/08/813008/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/813008/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/813008/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/813008/9"}}},"commit":{"parents":[{"commit":"7652ef325e0e84fe92ccf104bfe751a4d3542227","subject":"Merge \"[bifrost] Drop workarounds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7652ef325e0e84fe92ccf104bfe751a4d3542227"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-07 13:45:02.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-11-12 16:30:22.000000000","tz":60},"subject":"Make cron\u0027s logrotate schedule configurable","message":"Make cron\u0027s logrotate schedule configurable\n\nKolla-ansible allows user to change logrotate.conf,\nproblem is that logrotate cron file is running\ndaily by default in all distros, this is sometimes\nnot enough. So, this patch is adding script and\nway how to reconfigure schedule of logrotate job\n(cron file).\n\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/813039\n\nChange-Id: If31b08fcf45bafc0ab25498ece9208a576aae41b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f229eb7e06173649974376cf689d676c673f4988"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f229eb7e06173649974376cf689d676c673f4988"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
