)]}'
{"id":"openstack%2Fkolla~813082","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I44bcccd7aa7eb90e1d189acf2ecd1066455634e6","project":"openstack/kolla","branch":"stable/train","topic":"train-ci","hashtags":[],"change_id":"I44bcccd7aa7eb90e1d189acf2ecd1066455634e6","subject":"Fix multiple issues for stable/train","status":"ABANDONED","created":"2021-10-07 20:25:32.000000000","updated":"2022-01-25 13:51:39.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"a9025c91dae20f41674022955d1a1dbf9db29e9f","_number":813082,"virtual_id_number":813082,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-10-08 21:20:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-07 22:19:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-08 07:23:09.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-10-08 07:49:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-10-08 11:52:21.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"c45390a12944adbe6390e6a4f48b66bf46958bc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-07 20:25:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f296c07c4e24201cbd8c1cffd488984c6fa70593","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 22:19:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b23e64ffbced46b3b8d70d804ce2cd9c : RETRY_LIMIT in 3m 23s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b69521e7e7174d0d8bb4d07341e58f13 : RETRY_LIMIT in 3m 33s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/afe156519ac641c2a913d1d07ac493aa : RETRY_LIMIT in 3m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5ab94ef560a749219862dfc7a2757b56 : RETRY_LIMIT in 3m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/72b308499a1141b6becd367ac60e78ab : FAILURE in 22m 58s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b6ab6d74b1644a10b42443799f2110e8 : SUCCESS in 1h 15m 57s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ff573f3ec6eb483a852484e92680e222 : FAILURE in 27m 37s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/452647c030b147e8bdaaf1e8fbb1e146 : SUCCESS in 1h 06m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/fd87a05c65964e80a8caee4aba47cb15 : FAILURE in 24m 58s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/120354c4628f4231b504cdac1bc32e9e : SUCCESS in 47m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3af9bdb0e28d40b58a4a38860c425837 : SUCCESS in 22m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/28b0ee6e19b5491b9e402a763aaef557 : SUCCESS in 24m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ad1c7034c69c49d39a8e5524dd66b288 : FAILURE in 8m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0acce5395d384fadb3706921947fa277 : SUCCESS in 7m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/720c8bf9058b4786bb86370d413d46a7 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b36259b4ffeb47299e0a6eb9d176c5b3 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22f25784e2344a809bab9eb378137420 : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05d73385810944aa86179454d95a9de6 : SUCCESS in 7m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/08ce92a23d5f45999fc4718ed502fe3d : SUCCESS in 23m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/17113d29d2d64de698196fa06e3c1c2c : SUCCESS in 24m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/903c06d302df4ab0b48cf75aab78d637 : SUCCESS in 57m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/92ae957c8df044038909d1b6d3096818 : SUCCESS in 57m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6d9da971652b4022bd643194f0daa959 : SUCCESS in 1h 03m 02s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"884fa1750f7d89fe5a2df09db68aaccc6f33d3cc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-08 07:23:09.000000000","message":"Patch Set 1:\n\nthe CI itself has to be fixed too","accounts_in_message":[],"_revision_number":1},{"id":"bc74e48535569969f3b798a2388568eb746b5894","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-08 07:49:52.000000000","message":"Patch Set 1:\n\nhttps://bugs.launchpad.net/kolla/+bug/1943617","accounts_in_message":[],"_revision_number":1},{"id":"8bcad6256c029b73acd502829605fd1c9cce5144","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-08 10:36:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"81383902417eae1d9ae29473f65bd8ed10018655","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-08 11:52:21.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nfine otherwise and I approve of the cinderlib statement","accounts_in_message":[],"_revision_number":2},{"id":"6d07ce383fefcefccc7436dae46952ae3473428e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 11:53:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/930849f4ac1841de976bb2c10a37f821 : FAILURE in 41m 08s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/961de34baa7f4b04a5617f93221872a6 : FAILURE in 51m 59s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f15a760b782b4f26a1b5e01aa1afad37 : SUCCESS in 38m 02s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8a116e38aa94485c9ccee003ac3c5073 : FAILURE in 30m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cfd3f830afd243e497a97224564721de : FAILURE in 21m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9e5b36c4733b4271a8776f7c00c1c395 : SUCCESS in 59m 31s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4b3fc7625cbe44b3b9bf4e685e67347a : FAILURE in 23m 18s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/addcfe2c073e4038aafb66018f8a0cc8 : SUCCESS in 58m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/cd4c8b91665a46aa9963a7ec4f21a0a7 : FAILURE in 22m 20s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/576acdb272cd4187834a4a5ad293b14c : SUCCESS in 48m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9e740cc985ba4de88a10d9a4b4a758f9 : SUCCESS in 25m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bcf8a8881da4458fb74b3ba501d17ccb : SUCCESS in 24m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5826bb4e0f674fd79dd1d40073a22c9a : FAILURE in 6m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4a3a4bc70ab402eb415ef6fe094f28d : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4b9069f8cbfa4a77b22d350cce4f72f8 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cbc4ce2eb4642fe9e97c6ba2ad817cf : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/48cd78e3f73940cbab94c907218d4efc : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1f43f5dbacd400b90bb51c2b9fa3fef : SUCCESS in 7m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/adcca5792ff14f2187bb196a902d749a : SUCCESS in 24m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/25652276d85c49a78e2315ef1fbaaaac : SUCCESS in 29m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5d137bcb96944145b9805d12d1defc36 : SUCCESS in 47m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3f650fc4d53a4c88ba093b9f6b15368e : SUCCESS in 54m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b8931c7e2294480db339aea89d0e3355 : SUCCESS in 1h 11m 33s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"f00586dbdd536a7c66c873fec62b7677885c3acf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-08 12:49:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"465532489757309f933b86736f0d7e0f377996c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 13:03:19.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"3fbd4aa8c0838c196f84c2257656c8ab6dad6505","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-08 13:54:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e18bb85b292562fd8024fb9401b51645370aeb6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 15:21:05.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f30c909430a14a6c97dfbf8ad84df333 : FAILURE in 50m 57s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c1d05727218847aa9ee60cb40a98e3c1 : FAILURE in 1h 08m 41s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/95144a5088a746f69987a2b7b69ec92d : SUCCESS in 28m 16s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3b25d48312424571bca9664610622ed3 : FAILURE in 56m 10s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/94a0e9c219764ca885890de712d71b86 : FAILURE in 24m 18s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1fba38d48de84a9f9b8765b9b306fc50 : SUCCESS in 55m 23s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b1f16bd0d8f54a19bb9d24492342c455 : FAILURE in 24m 10s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6a6e7ce319944c068922b1574cb657c2 : SUCCESS in 1h 03m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4719c38f33a543e4b86f63895d1f5c87 : FAILURE in 25m 40s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/14fd014986c34271825c050a7b68aa2c : SUCCESS in 56m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1d01f7bf967a4aea898910c4a4ae102e : FAILURE in 21m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/23efd9bd5bdc4213a8dc22a5dd47ec47 : SUCCESS in 20m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3385ade1011b4dcb82f7725388527e0e : FAILURE in 6m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8a4e5e077424fc5833397dc64907cb7 : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6eb394e3934c41cca9787c49c9ef4cd5 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a4f879af4ef4f14bde1dc2ff45ee677 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6bc271db9ec7425c8f0f6d31310da6b1 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a2b189ca2bc4ab88ca03dfe7388cfa0 : SUCCESS in 6m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3c08c5b36203497aa03be934b5b1a066 : SUCCESS in 27m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/77b65a7c3134423894efb607727b8862 : FAILURE in 29m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/584fe65a6f0c46dc8d19eb895f35b351 : SUCCESS in 47m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8909e62bf97c4c4f81ed5bcf51e383da : SUCCESS in 47m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/61ed4db357e44720a36b2fdc3c40b398 : SUCCESS in 58m 16s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"fa522650ccb3c2ae16fdc33b8def0b33f27bd17d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-08 15:33:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ec65a6d0f7278fd38b6a7f461b39abf2f2e029e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 16:55:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4aa8e0c8866b4ac99d334a281fb7aa83 : FAILURE in 51m 52s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/662931d268d741b0a41832c48348acba : FAILURE in 54m 37s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/109f0d22888f45859981e01d0229c834 : SUCCESS in 32m 40s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/908b05b44f0d4b9686e3470cc3cc7f79 : FAILURE in 33m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7b0e0dc86e9d4267b7e43e74bc0ca4ba : FAILURE in 21m 56s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/41ef8ef38a074b0698c368ac72eb0a1c : SUCCESS in 53m 45s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a01478cb6db4495ca5e294e35b50b567 : FAILURE in 22m 54s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/00ccf72e5e4f49b2a4ac276eda4f1daf : SUCCESS in 1h 10m 55s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3ce39b6eeb894227909e27bf2517fd40 : FAILURE in 22m 48s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6a246f13b4e44cb0be523850141cfca0 : SUCCESS in 45m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/91ffd48373c24624a8ba4eb6e1f827d1 : SUCCESS in 23m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e77b655947ae491997c853be514dd099 : SUCCESS in 26m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7f0be784d851455b8eaf858c26521c3c : FAILURE in 7m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1acda6db04544629ac4df40337cd52c2 : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/133b81e2c82d4472860bff3a60d6c587 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2838729d2604c828e203ef4e6c5e854 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e70e07b86714678bc93af3e392590e9 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb0d648bd1b14017a51298ec4b05ba6b : SUCCESS in 7m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9ae7bc9b2f914820b5cee791c14814c0 : SUCCESS in 23m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/10d545352ede4c9ebb6bd5995f9c981b : SUCCESS in 23m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e0434dfc5055417d83c636d4fc354379 : SUCCESS in 47m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3faef7a992524321bc3e556d643dcdf4 : SUCCESS in 56m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a1a6fae545864dd0af95a0a7d0386d25 : SUCCESS in 1h 02m 50s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":4},{"id":"a73b503ec4870a3dbca2dcd1924d0906714b1dfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-08 20:15:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"479bcbfc72a2e9db484065d86f7cd671a6ac6d51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 21:20:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ff238cb75d024bfcbb070a7967904256 : FAILURE in 47m 08s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/bdc52d6fe5d5469b977655f7f97991b3 : SUCCESS in 53m 51s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/782b0cb87d0e4dbcac37bd5194bdbd7c : SUCCESS in 30m 00s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/995a0bf3edd34fa29fc6cf092a4fed8a : FAILURE in 30m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/74c2f014919544c1bd198b0cea9fd207 : FAILURE in 21m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7e6a844bddaf49a8982f55140376032d : SUCCESS in 59m 33s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/49b92d2b386244279450ec2cd9f523fd : FAILURE in 20m 18s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a17df73bf40241a38050d08920716a15 : SUCCESS in 58m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9cc5f81caee84724b2a262d44d9633ac : FAILURE in 20m 11s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/45c1559d87d5439b916a75910365d1a8 : SUCCESS in 47m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4902fe949cb149758f87596e4d015f8e : SUCCESS in 38m 30s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/833667470dbf46978f46b2168d18f90b : SUCCESS in 20m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f0329c7ae7464d57a24e30ed949638ad : FAILURE in 7m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f9f543ef9d481db8dd91b387af1ee0 : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/320a7dab17c249d682b0343190cc95e3 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afed7ed5c13a4c8c86f9be67a0dffc80 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f538d15f4f844a69f20b99cecd8795e : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/133203c021464ccca78aa67488298c69 : SUCCESS in 10m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/84b0ea65a0b840b38388bed571964192 : SUCCESS in 27m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f905cfff568436ca421bba4080de7da : SUCCESS in 25m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f2972fe0bdab46f08c206f30849c5536 : SUCCESS in 54m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fea212a1451742958695a6f1867c4e2d : SUCCESS in 49m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/242a8cad30b04fcdb8084a88e89a663d : SUCCESS in 59m 48s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":5},{"id":"d02607e2186794aef3f6a03e30404fd07615e382","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-09 12:00:51.000000000","message":"Patch Set 5:\n\nneed to disable bifrost building and also fix the ssl issue with k building on k-a side:\nSSLError: HTTPSConnectionPool(host\u003d\u0027tarballs.openstack.org\u0027, port\u003d443): Max retries exceeded with url: /requirements/requirements-stable-train.tar.gz (Caused by SSLError(SSLError(1, u\u0027[SSL: CERTIFICATE_VERIFY_FAILED] certificate verify failed (_ssl.c:618)\u0027),))","accounts_in_message":[],"_revision_number":5},{"id":"a9025c91dae20f41674022955d1a1dbf9db29e9f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-25 13:51:39.000000000","message":"Abandoned\n\nSee https://review.opendev.org/c/openstack/kolla/+/826033 instead.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"bcbdd62bbbece992c081e55bfc27ecb0d3bd3293","revisions":{"4dde997a4112745d2887c232fa9654e95c834652":{"kind":"REWORK","_number":1,"created":"2021-10-07 20:25:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/82/813082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/813082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/813082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/813082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/813082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/813082/1"}}},"commit":{"parents":[{"commit":"fa53a60c149026ff6641e1e1d9bae240dd888f3d","subject":"Fix _get_images_dir when pip-installed with --user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa53a60c149026ff6641e1e1d9bae240dd888f3d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-07 20:24:43.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-07 20:24:43.000000000","tz":120},"subject":"Fix 2to3 issue with source image build","message":"Fix 2to3 issue with source image build\n\nsetuptools 58.0.0 removed support for 2to3, which causes the\nopenstack-base build to fail when installing anyjson 0.3.3.\n\nStop installing setuptools via virtualenv, which would get the latest\nversion. Install an older release separately instead.\n\nChange-Id: I44bcccd7aa7eb90e1d189acf2ecd1066455634e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4dde997a4112745d2887c232fa9654e95c834652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4dde997a4112745d2887c232fa9654e95c834652"}]},"branch":"refs/heads/stable/train"},"3732fb6b260a6e095493fcfded78125538f27063":{"kind":"REWORK","_number":2,"created":"2021-10-08 10:36:39.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/82/813082/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/813082/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/813082/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/813082/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/813082/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/813082/2"}}},"commit":{"parents":[{"commit":"fa53a60c149026ff6641e1e1d9bae240dd888f3d","subject":"Fix _get_images_dir when pip-installed with --user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa53a60c149026ff6641e1e1d9bae240dd888f3d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-08 10:36:20.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-08 10:36:20.000000000","tz":120},"subject":"CI: Fix multiple issues for stable/train","message":"CI: Fix multiple issues for stable/train\n\n1) Require pip version compatible with Python 2\n\nUpgrading from pip 8.1.2 as packaged in CentOS 7 fails with:\n\nCollecting pip\n  Downloading https://files.pythonhosted.org/packages/52/e1/06c018197d8151383f66ebf6979d951995cf495629fc54149491f5d157d0/pip-21.2.4.tar.gz (1.6MB)\n    100% |████████████████████████████████| 1.6MB 444kB/s\n    Complete output from command python setup.py egg_info:\n    Traceback (most recent call last):\n      File \"\u003cstring\u003e\", line 1, in \u003cmodule\u003e\n      File \"/tmp/pip-build-66n_Zb/pip/setup.py\", line 7\n        def read(rel_path: str) -\u003e str:\n                         ^\n    SyntaxError: invalid syntax\n\n    ----------------------------------------\nCommand \"python setup.py egg_info\" failed with error code 1 in /tmp/pip-build-66n_Zb/pip/\nYou are using pip version 8.1.2, however version 21.2.4 is available.\nYou should consider upgrading via the \u0027pip install --upgrade pip\u0027 command.\n\nSee https://github.com/pypa/pip/issues/10214 for more details.\n\n2) Use local upper-constraints file to avoid SSL verification issues\n\nThe R3 Let\u0027s Encrypt intermediate CA used by releases.openstack.org\nexpired, which causes SSL verification issues.\n\n3) Fix 2to3 issue with source image build\n\nsetuptools 58.0.0 removed support for 2to3, which causes the\nopenstack-base build to fail when installing anyjson 0.3.3.\n\nStop installing setuptools via virtualenv, which would get the latest\nversion. Install an older release separately instead.\n\nCloses-Bug: #1943617\n\n4) Drop cinderlib from cinder-volume image\n\nThis package cannot be installed on CentOS 7 anymore due to package\ndependency issues in EPEL. According to yoctozepto, cinderlib is\nactually an implementation for users wishing to use cinder functionality\nwithout OpenStack.\n\nChange-Id: I44bcccd7aa7eb90e1d189acf2ecd1066455634e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3732fb6b260a6e095493fcfded78125538f27063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3732fb6b260a6e095493fcfded78125538f27063"}]},"branch":"refs/heads/stable/train"},"b92fb62633479115bc3b7d411cef1ab826ab5cf6":{"kind":"REWORK","_number":3,"created":"2021-10-08 12:49:05.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/82/813082/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/813082/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/813082/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/813082/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/813082/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/813082/3"}}},"commit":{"parents":[{"commit":"fa53a60c149026ff6641e1e1d9bae240dd888f3d","subject":"Fix _get_images_dir when pip-installed with --user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa53a60c149026ff6641e1e1d9bae240dd888f3d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-08 12:48:27.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-08 12:48:27.000000000","tz":120},"subject":"Fix multiple issues for stable/train","message":"Fix multiple issues for stable/train\n\n1) CI: Require pip version compatible with Python 2\n\nUpgrading from pip 8.1.2 as packaged in CentOS 7 fails with:\n\nCollecting pip\n  Downloading https://files.pythonhosted.org/packages/52/e1/06c018197d8151383f66ebf6979d951995cf495629fc54149491f5d157d0/pip-21.2.4.tar.gz (1.6MB)\n    100% |████████████████████████████████| 1.6MB 444kB/s\n    Complete output from command python setup.py egg_info:\n    Traceback (most recent call last):\n      File \"\u003cstring\u003e\", line 1, in \u003cmodule\u003e\n      File \"/tmp/pip-build-66n_Zb/pip/setup.py\", line 7\n        def read(rel_path: str) -\u003e str:\n                         ^\n    SyntaxError: invalid syntax\n\n    ----------------------------------------\nCommand \"python setup.py egg_info\" failed with error code 1 in /tmp/pip-build-66n_Zb/pip/\nYou are using pip version 8.1.2, however version 21.2.4 is available.\nYou should consider upgrading via the \u0027pip install --upgrade pip\u0027 command.\n\nSee https://github.com/pypa/pip/issues/10214 for more details.\n\n2) CI: Use local upper-constraints file to avoid SSL verification issues\n\nThe R3 Let\u0027s Encrypt intermediate CA used by releases.openstack.org\nexpired, which causes SSL verification issues.\n\n3) Fix 2to3 issue with source image build\n\nsetuptools 58.0.0 removed support for 2to3, which causes the\nopenstack-base build to fail when installing anyjson 0.3.3.\n\nStop installing setuptools via virtualenv, which would get the latest\nversion. Install an older release separately instead. For more details\nsee https://bugs.launchpad.net/kolla/+bug/1943617.\n\n4) Drop cinderlib from cinder-volume image\n\nThis package cannot be installed on CentOS 7 anymore due to package\ndependency issues in EPEL. According to yoctozepto, cinderlib is\nactually an implementation for users wishing to use cinder functionality\nwithout OpenStack.\n\n5) Mark several images as unbuildable\n\nMark gnocchi-base, monasca-grafana and sensu-base as unbuildable. This\nmay be reverted if they can be fixed.\n\nChange-Id: I44bcccd7aa7eb90e1d189acf2ecd1066455634e6\nCloses-Bug: #1943617\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/813171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b92fb62633479115bc3b7d411cef1ab826ab5cf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b92fb62633479115bc3b7d411cef1ab826ab5cf6"}]},"branch":"refs/heads/stable/train"},"c0fd78fb5ce16ee6202cb30223e10cd41478280f":{"kind":"REWORK","_number":4,"created":"2021-10-08 15:33:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/82/813082/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/813082/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/813082/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/813082/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/813082/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/813082/4"}}},"commit":{"parents":[{"commit":"fa53a60c149026ff6641e1e1d9bae240dd888f3d","subject":"Fix _get_images_dir when pip-installed with --user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa53a60c149026ff6641e1e1d9bae240dd888f3d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-08 15:33:49.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-08 15:33:49.000000000","tz":120},"subject":"Fix multiple issues for stable/train","message":"Fix multiple issues for stable/train\n\n1) CI: Require pip version compatible with Python 2\n\nUpgrading from pip 8.1.2 as packaged in CentOS 7 fails with:\n\nCollecting pip\n  Downloading https://files.pythonhosted.org/packages/52/e1/06c018197d8151383f66ebf6979d951995cf495629fc54149491f5d157d0/pip-21.2.4.tar.gz (1.6MB)\n    100% |████████████████████████████████| 1.6MB 444kB/s\n    Complete output from command python setup.py egg_info:\n    Traceback (most recent call last):\n      File \"\u003cstring\u003e\", line 1, in \u003cmodule\u003e\n      File \"/tmp/pip-build-66n_Zb/pip/setup.py\", line 7\n        def read(rel_path: str) -\u003e str:\n                         ^\n    SyntaxError: invalid syntax\n\n    ----------------------------------------\nCommand \"python setup.py egg_info\" failed with error code 1 in /tmp/pip-build-66n_Zb/pip/\nYou are using pip version 8.1.2, however version 21.2.4 is available.\nYou should consider upgrading via the \u0027pip install --upgrade pip\u0027 command.\n\nSee https://github.com/pypa/pip/issues/10214 for more details.\n\n2) CI: Use local upper-constraints file to avoid SSL verification issues\n\nThe R3 Let\u0027s Encrypt intermediate CA used by releases.openstack.org\nexpired, which causes SSL verification issues.\n\n3) Fix 2to3 issue with source image build\n\nsetuptools 58.0.0 removed support for 2to3, which causes the\nopenstack-base build to fail when installing anyjson 0.3.3.\n\nStop installing setuptools via virtualenv, which would get the latest\nversion. Install an older release separately instead. For more details\nsee https://bugs.launchpad.net/kolla/+bug/1943617.\n\n4) Drop cinderlib from cinder-volume image\n\nThis package cannot be installed on CentOS 7 anymore due to package\ndependency issues in EPEL. According to yoctozepto, cinderlib is\nactually an implementation for users wishing to use cinder functionality\nwithout OpenStack.\n\n5) Mark several images as unbuildable\n\nMark gnocchi-base, monasca-grafana and sensu-base as unbuildable. This\nmay be reverted if they can be fixed.\n\nChange-Id: I44bcccd7aa7eb90e1d189acf2ecd1066455634e6\nCloses-Bug: #1943617\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/813171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0fd78fb5ce16ee6202cb30223e10cd41478280f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0fd78fb5ce16ee6202cb30223e10cd41478280f"}]},"branch":"refs/heads/stable/train"},"bcbdd62bbbece992c081e55bfc27ecb0d3bd3293":{"kind":"REWORK","_number":5,"created":"2021-10-08 20:15:38.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/82/813082/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/813082/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/813082/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/813082/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/813082/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/813082/5"}}},"commit":{"parents":[{"commit":"fa53a60c149026ff6641e1e1d9bae240dd888f3d","subject":"Fix _get_images_dir when pip-installed with --user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa53a60c149026ff6641e1e1d9bae240dd888f3d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-08 20:15:31.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-08 20:15:31.000000000","tz":120},"subject":"Fix multiple issues for stable/train","message":"Fix multiple issues for stable/train\n\n1) CI: Require pip version compatible with Python 2\n\nUpgrading from pip 8.1.2 as packaged in CentOS 7 fails with:\n\nCollecting pip\n  Downloading https://files.pythonhosted.org/packages/52/e1/06c018197d8151383f66ebf6979d951995cf495629fc54149491f5d157d0/pip-21.2.4.tar.gz (1.6MB)\n    100% |████████████████████████████████| 1.6MB 444kB/s\n    Complete output from command python setup.py egg_info:\n    Traceback (most recent call last):\n      File \"\u003cstring\u003e\", line 1, in \u003cmodule\u003e\n      File \"/tmp/pip-build-66n_Zb/pip/setup.py\", line 7\n        def read(rel_path: str) -\u003e str:\n                         ^\n    SyntaxError: invalid syntax\n\n    ----------------------------------------\nCommand \"python setup.py egg_info\" failed with error code 1 in /tmp/pip-build-66n_Zb/pip/\nYou are using pip version 8.1.2, however version 21.2.4 is available.\nYou should consider upgrading via the \u0027pip install --upgrade pip\u0027 command.\n\nSee https://github.com/pypa/pip/issues/10214 for more details.\n\n2) CI: Use local upper-constraints file to avoid SSL verification issues\n\nThe R3 Let\u0027s Encrypt intermediate CA used by releases.openstack.org\nexpired, which causes SSL verification issues.\n\n3) Fix 2to3 issue with source image build\n\nsetuptools 58.0.0 removed support for 2to3, which causes the\nopenstack-base build to fail when installing anyjson 0.3.3.\n\nStop installing setuptools via virtualenv, which would get the latest\nversion. Install an older release separately instead. For more details\nsee https://bugs.launchpad.net/kolla/+bug/1943617.\n\n4) Drop cinderlib from cinder-volume image\n\nThis package cannot be installed on CentOS 7 anymore due to package\ndependency issues in EPEL. According to yoctozepto, cinderlib is\nactually an implementation for users wishing to use cinder functionality\nwithout OpenStack.\n\n5) Mark several images as unbuildable\n\nMark gnocchi-base, monasca-grafana and sensu-base as unbuildable. This\nmay be reverted if they can be fixed.\n\nChange-Id: I44bcccd7aa7eb90e1d189acf2ecd1066455634e6\nCloses-Bug: #1943617\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/813171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bcbdd62bbbece992c081e55bfc27ecb0d3bd3293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bcbdd62bbbece992c081e55bfc27ecb0d3bd3293"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
