)]}'
{"id":"openstack%2Fkolla~813144","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I179b5aa346e0b4f38ae1a6bbf93d6a23e36c271f","project":"openstack/kolla","branch":"stable/train","topic":"pip-train","hashtags":[],"change_id":"I179b5aa346e0b4f38ae1a6bbf93d6a23e36c271f","subject":"Require pip version compatible with Python 2","status":"ABANDONED","created":"2021-10-08 08:24:17.000000000","updated":"2021-10-08 10:48:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b4a6d1f4fa02a4a7cb2029a3562d4ad24eb495ee","_number":813144,"virtual_id_number":813144,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-10-08 09:38:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-08 09:38:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"380e3f4a0e53921b319721466a15f48b7be04f53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-08 08:24:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eccc7bc710c1cd321b015260731cb4f5e84a0e17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-08 08:24:53.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d3b558878f0909c4d9950f1143d3413e7491dace","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 09:38:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b4af4c8d2068460fa54e5adef6c7c41c : FAILURE in 5m 04s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1476efe231114cf2b9571e056caa9930 : FAILURE in 5m 03s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/52ca707ec7b14508ac05f5bfe147aff8 : FAILURE in 5m 21s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3e6e92d5efb5405e8d19e37c2e70a3ef : FAILURE in 5m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4b21e62f024a4c558c8b64bdb5197024 : FAILURE in 24m 23s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b27497baecf04306a043bbfe78be1e87 : SUCCESS in 1h 01m 47s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d97d48ba22be40dcb11f3bcc5ef873fe : FAILURE in 24m 59s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/cfef6d2666e541cfa7fefcf0e4f92233 : SUCCESS in 1h 05m 14s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3f9c2f52bc8f46e9abc79ea9bffacbdd : FAILURE in 28m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dafc0be3b2854e7c85b42def751a63e2 : SUCCESS in 24m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c956ee508f5b4912b45b8eac38b71e6b : SUCCESS in 20m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aac7d64b3e6749e7a06f726d78627c7a : FAILURE in 8m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab906a4b56d44d52b2a7496902543a54 : SUCCESS in 6m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/64c65db41571491b86eaaa8c4840d55d : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3eda5c2fbc754b4ba9917d701c379eb0 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c72a6e732f794615aad50a8d3ce13dbe : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f1e86bcd6634e829bb9bc5c632b0306 : SUCCESS in 7m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fbf5aad7bb85482ba113adf811766a84 : SUCCESS in 26m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0f2c6cffb66444388ac724e9810f30d3 : SUCCESS in 24m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/eec2c78a26e34812a7cdcd7ef1d11040 : SUCCESS in 50m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/270f51ef2aef4446b9e2cbabe9511d3b : SUCCESS in 50m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aea37ddb9a524d4ba26c6a12826c110b : SUCCESS in 1h 04m 49s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"b4a6d1f4fa02a4a7cb2029a3562d4ad24eb495ee","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-08 10:48:08.000000000","message":"Abandoned\n\nFolded into https://review.opendev.org/c/openstack/kolla/+/813082","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"394a0181bc691ec3d02d718b63d09af0a03368be","revisions":{"9f8f169ed5e781da8f167067d52f3f0b9d63b2dd":{"kind":"REWORK","_number":1,"created":"2021-10-08 08:24:17.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/813144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/813144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/813144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/813144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/813144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/813144/1"}}},"commit":{"parents":[{"commit":"fa53a60c149026ff6641e1e1d9bae240dd888f3d","subject":"Fix _get_images_dir when pip-installed with --user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa53a60c149026ff6641e1e1d9bae240dd888f3d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-08 08:21:34.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-08 08:21:34.000000000","tz":120},"subject":"Require pip version compatible with Python 2","message":"Require pip version compatible with Python 2\n\nUpgrading from pip 8.1.2 as packaged in CentOS 7 fails with:\n\nCollecting pip\n  Downloading https://files.pythonhosted.org/packages/52/e1/06c018197d8151383f66ebf6979d951995cf495629fc54149491f5d157d0/pip-21.2.4.tar.gz (1.6MB)\n    100% |████████████████████████████████| 1.6MB 444kB/s\n    Complete output from command python setup.py egg_info:\n    Traceback (most recent call last):\n      File \"\u003cstring\u003e\", line 1, in \u003cmodule\u003e\n      File \"/tmp/pip-build-66n_Zb/pip/setup.py\", line 7\n        def read(rel_path: str) -\u003e str:\n                         ^\n    SyntaxError: invalid syntax\n\n    ----------------------------------------\nCommand \"python setup.py egg_info\" failed with error code 1 in /tmp/pip-build-66n_Zb/pip/\nYou are using pip version 8.1.2, however version 21.2.4 is available.\nYou should consider upgrading via the \u0027pip install --upgrade pip\u0027 command.\n\nChange-Id: I179b5aa346e0b4f38ae1a6bbf93d6a23e36c271f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f8f169ed5e781da8f167067d52f3f0b9d63b2dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f8f169ed5e781da8f167067d52f3f0b9d63b2dd"}]},"branch":"refs/heads/stable/train"},"394a0181bc691ec3d02d718b63d09af0a03368be":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-08 08:24:53.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/813144/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/813144/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/813144/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/813144/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/813144/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/813144/2"}}},"commit":{"parents":[{"commit":"fa53a60c149026ff6641e1e1d9bae240dd888f3d","subject":"Fix _get_images_dir when pip-installed with --user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa53a60c149026ff6641e1e1d9bae240dd888f3d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-08 08:24:36.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-08 08:24:36.000000000","tz":120},"subject":"Require pip version compatible with Python 2","message":"Require pip version compatible with Python 2\n\nUpgrading from pip 8.1.2 as packaged in CentOS 7 fails with:\n\nCollecting pip\n  Downloading https://files.pythonhosted.org/packages/52/e1/06c018197d8151383f66ebf6979d951995cf495629fc54149491f5d157d0/pip-21.2.4.tar.gz (1.6MB)\n    100% |████████████████████████████████| 1.6MB 444kB/s\n    Complete output from command python setup.py egg_info:\n    Traceback (most recent call last):\n      File \"\u003cstring\u003e\", line 1, in \u003cmodule\u003e\n      File \"/tmp/pip-build-66n_Zb/pip/setup.py\", line 7\n        def read(rel_path: str) -\u003e str:\n                         ^\n    SyntaxError: invalid syntax\n\n    ----------------------------------------\nCommand \"python setup.py egg_info\" failed with error code 1 in /tmp/pip-build-66n_Zb/pip/\nYou are using pip version 8.1.2, however version 21.2.4 is available.\nYou should consider upgrading via the \u0027pip install --upgrade pip\u0027 command.\n\nSee https://github.com/pypa/pip/issues/10214 for more details.\n\nChange-Id: I179b5aa346e0b4f38ae1a6bbf93d6a23e36c271f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/394a0181bc691ec3d02d718b63d09af0a03368be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/394a0181bc691ec3d02d718b63d09af0a03368be"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
