)]}'
{"id":"openstack%2Fkolla~813252","triplet_id":"openstack%2Fkolla~master~Ie388938fc207c982c5650221baee354a5bb9be30","project":"openstack/kolla","branch":"master","topic":"kolla-xena-release","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2021-10-11 13:09:58.000000000","reason":"Michal Nasiadka replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-10-11 14:27:15.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-10-11 14:27:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie388938fc207c982c5650221baee354a5bb9be30","subject":"Fix GitHub tarball URL for monasca-thresh","status":"MERGED","created":"2021-10-08 20:36:39.000000000","updated":"2021-10-11 14:29:13.000000000","submitted":"2021-10-11 14:27:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"813252-kolla-xena-release","meta_rev_id":"5b477034743fc84694931d5b63de13dffadb8e31","_number":813252,"virtual_id_number":813252,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-11 14:27:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-10-11 13:09:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-10-11 09:19:21.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-10-11 13:09:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-10-11 14:29:13.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-08 22:13:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-09 12:03:59.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-10-09 12:28:26.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-10-09 12:28:26.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-09 12:28:26.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"1bef0fb2c0268b4aae9afc2d009b7f5a54f1658d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-08 20:36:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aac0b0530cd8e33e53f175d1b1dfa00715dff435","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-08 20:38:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ee55722912f8700c6f5d682076358ff3a83aff4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 22:13:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b497684fb182442ab630ee32c400810b : SUCCESS in 30m 27s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c1d4bb93983b45b0b3043e8f530c9a8f : SUCCESS in 42m 55s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/340f6139403b46c5bee35cf95aead18a : SUCCESS in 1h 15m 54s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9e4ee5b6a4a74edb983a8a93baabac9a : SUCCESS in 1h 10m 12s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/490c5708e7be4d56ac165b35ae686c32 : SUCCESS in 1h 27m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6213659c1ef54ee4a97629f1d5dd811d : SUCCESS in 23m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/93ee727aaff74c1683105e906f64e274 : SUCCESS in 16m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a3471bcb06a842918729209c405ddd2c : SUCCESS in 1h 01m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddfe800b980642dc9f31f758ef8a67dd : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09af8e11feb8447d8f37a734add94acf : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35010e8d29894fda852b0a3784d0b87d : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21335e6cc40d4b3c99d159a44c8b200c : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eb4aa5c6311447b85fdcb15ea12ed15 : SUCCESS in 6m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/db65189bcfdc439398dc2190ac26cbb6 : SUCCESS in 21m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fc4224a04d824c4ba70aa1d4d37c0d66 : SUCCESS in 26m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/01f6af31309b46c9a77bcba4d4641941 : SUCCESS in 56m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/797e0957fe4641a7b3571c0dc8e97472 : SUCCESS in 53m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/584bc08b29a14d4bad97aee7344ff763 : SUCCESS in 1h 12m 26s","accounts_in_message":[],"_revision_number":2},{"id":"17b28f126a42d2a8836ff7a779aec4e4331f6467","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 22:43:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/8094c97fa25648af8374cc2769642284 : SUCCESS in 2h 02m 50s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/572a524f173f4195bf6d66ed5795e0ca : SUCCESS in 1h 08m 28s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/50c096fac6674683a01c576741117464 : SUCCESS in 2h 02m 08s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bb281936f3b745a5b3a7963c381c03b4 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f12302aac990416bb1914185a9f1a88d : SUCCESS in 7m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9ad734d6d7cb4d1b92d6002a33c06ce0 : SUCCESS in 1h 10m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c4c8abd8b25f22120c9161b1d64fc8c8ea98d3b6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-09 12:03:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d48b9cf113e89b23a38883fc770dcbce4ec24eae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-09 12:27:57.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c71732ae10599e5e558fa5421e8d977dccf5e7a7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-09 12:28:26.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"a9b1872796646abc4fc64bdf3f37971697bd05e1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-09 12:28:31.000000000","message":"Topic set to kolla-xena-release","accounts_in_message":[],"_revision_number":3},{"id":"5a8f39ae38162a8f2dd71a6ef02bdcd295b53eda","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-09 12:34:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9ef602d7e85b2855065d7efdc6b4247e9cd31422","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 12:47:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7eab18767d8e49648e1b31790957478f : FAILURE in 8m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b110f965f9f4c1e84428d87063d2c29 : FAILURE in 6m 08s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d61ae0a6a4cd46779785d86bc5ed3017 : FAILURE in 17m 31s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ea79d62a1b744c7965fcae797544165 : FAILURE in 6m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba9f2b13bb7841d4a5872886a4a1dffa : FAILURE in 6m 31s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/42117a1bd4a443ac9d8b95217f1cc5a8 : FAILURE in 6m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"25ad6db2d279c35ac68facd606b4a9e848146b75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 12:54:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d88f69860c834f0b95246c4fa81943ac : FAILURE in 5m 22s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c03bde7577f44204b55d2a9254d8dd91 : FAILURE in 5m 02s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9b6bc754b55b490b87cd9a8f12cbd19d : FAILURE in 10m 19s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8201e8629abb40c48e234876e3c643c3 : FAILURE in 10m 38s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3c42c804a0354bce8b4bf374049fced1 : FAILURE in 10m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5e1e45a1b3b844ef96cdef380f9a2e80 : FAILURE in 3m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a385e7117f1843d5b1d8d15278af7a67 : FAILURE in 7m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ea2c8c3278af481d8644fb2f0e596e5b : FAILURE in 13m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39bb002e94f3489f904b0776b9d4e8e8 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d211bb99065f4e68bc89b2fe4926cba4 : FAILURE in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9fa3ab44fe7409c9b4a13753f02cb13 : FAILURE in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42ddbc0e49694640bddf34cb51f7cd8a : FAILURE in 3m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f399beef75ce4f0c97329b7611e28555 : SUCCESS in 5m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7ab8060a116a4eac8402e4cc3d29f4f5 : FAILURE in 3m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/da4986ff4cd14fdf8b7b00881f5b9a27 : FAILURE in 3m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f6341818aa4401a80a89edbbce0bdd6 : FAILURE in 9m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/10c4d2fff9ba48cf8620c326724141cd : FAILURE in 9m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/136ca3dae95d4eaa8b602c1094ae93d4 : FAILURE in 8m 30s","accounts_in_message":[],"_revision_number":3},{"id":"e5702df05edc3102b6d99cc8b8a328232aeaf61a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-09 14:51:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"029660d340de5666eed3289d0926132774f2d073","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-09 16:04:22.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"d3cee8feaae27391c27ec086c8f2e6e6b1cee0be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 17:36:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/dcbcf33fee8442fe9da5be326338fda3 : SUCCESS in 25m 20s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8a3e7d76053c4c61a52e3b8fe7192c77 : SUCCESS in 37m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/293b4d91f33b4349a371e320d43e79b7 : SUCCESS in 1h 04m 53s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/1ac081937af54545895740cc4f5f0844 : SUCCESS in 54m 08s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5641142063cc479c89f701369d994959 : SUCCESS in 1h 11m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/eea0b9f0e8dc42aa931aee6c289a7fd7 : SUCCESS in 21m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b5dcf628fa104b1dbff4e79bb7eaee26 : FAILURE in 18m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9ba139454df343a9a243a5197361206e : SUCCESS in 1h 10m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6e201391fab4cfd9c585838fadd9154 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/384a4d5ba8b34985a87a4c341d53219c : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd1e672e95ae4a9f87dd7fb6eec0e127 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72d7531949f84a8d8e2c29c113d76fb7 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcd202eea01844d9b98b1e772a07422c : SUCCESS in 5m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f6dbf016dfbf441ea6db979c250d1de7 : SUCCESS in 20m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/31c5d7d6dc264e7292c8742331dd36a1 : SUCCESS in 35m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/22b9472eeac64b6e843f1f7318669baf : SUCCESS in 53m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4473ff5436e64e69a0ef2f2b2c2767ee : SUCCESS in 37m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b35aeaba7463407280b97b5c0b02f25c : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":5},{"id":"9f676751903288f8d9ff855d736f4cb9ec54c74b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 18:08:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1a7e8f970800404586d2dcf094ab583e : SUCCESS in 1h 46m 42s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/696a8ec47e154cbea28391cd9f4577b2 : SUCCESS in 1h 26m 54s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/293b0f1964b645fe83b3fce4daaba564 : SUCCESS in 2h 01m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fff26fcebd5e46b7b185bd8f88ec2a53 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/282fde29c1494d22b7f2a0b81b3b0ead : SUCCESS in 5m 44s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a4727790164b4dfc8356c803b04c21b0 : SUCCESS in 1h 25m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1b6e1a34c6c5b2ba2fccb7658c844c990b7e05e0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-11 09:19:21.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"4fbe6778354e643faffe6d5aacbf19f31f6930aa","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-11 09:29:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"90b349f8986d12d2175bca53bb260b9c6f5328a4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-11 13:01:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cfe6bf38b700ca9551c9eb54e9ea319e53eb9bd8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-11 13:09:58.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f8297dac37f1225f0a498006639192a73767be54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 13:10:36.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"312549ecba48db4b9e3176e6f63b711ea73067a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 14:27:14.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f231e9a1846845c19e2f3f5304cfb3fd : SUCCESS in 39m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/aac619ba14724f1ba17d9852333aaa87 : SUCCESS in 1h 12m 17s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8e4a0c145e854595a0f7b01fabe39d60 : SUCCESS in 1h 15m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1feaf31638140ecb7d1ee346faa9ddf : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f460cd1c5854431b9e8b4ad4987b81fd : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80539f7ae09142a7aba2080f88f4345d : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c7c4921043a491eb7fd90e9abdd608f : SUCCESS in 6m 07s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2bc72e112fe846b6b6ef0387b16370e5 : SUCCESS in 21m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e0756baf40c6450493f0cd7f0341b3b6 : SUCCESS in 53m 24s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ca9d69a846744ec7b9470c783eae2b7a : SUCCESS in 1h 08m 34s","accounts_in_message":[],"_revision_number":5},{"id":"2eccab74536716c17159881ec98f0c38f2f864cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 14:27:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"29a2eed4bdc2b1d756f4275aa1c0eec503a1806a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 14:29:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f63501be794643638d8e2c30e2a2066a : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"556e8f9eeef71d240bef961cee0e3e5d357e3711","revisions":{"1b8970810957a2055b74170058c00b06760d5f00":{"kind":"REWORK","_number":1,"created":"2021-10-08 20:36:39.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/813252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/813252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/813252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/813252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/813252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/813252/1"}}},"commit":{"parents":[{"commit":"8cdf952c6be7e15c1e6a7589371bc827386d4cd2","subject":"release: Use Xena sources by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cdf952c6be7e15c1e6a7589371bc827386d4cd2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-08 20:36:23.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-08 20:36:23.000000000","tz":120},"subject":"Fix GitHub tarball URL for monasca-thresh","message":"Fix GitHub tarball URL for monasca-thresh\n\nChange-Id: Ie388938fc207c982c5650221baee354a5bb9be30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b8970810957a2055b74170058c00b06760d5f00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b8970810957a2055b74170058c00b06760d5f00"}]},"branch":"refs/heads/master"},"1364ca8e503325c43e061e99173439f76aa5aa49":{"kind":"REWORK","_number":2,"created":"2021-10-08 20:38:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/813252/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/813252/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/813252/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/813252/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/813252/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/813252/2"}}},"commit":{"parents":[{"commit":"8cdf952c6be7e15c1e6a7589371bc827386d4cd2","subject":"release: Use Xena sources by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cdf952c6be7e15c1e6a7589371bc827386d4cd2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-08 20:37:56.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-08 20:37:56.000000000","tz":120},"subject":"Fix GitHub tarball URL for monasca-thresh","message":"Fix GitHub tarball URL for monasca-thresh\n\nChange-Id: Ie388938fc207c982c5650221baee354a5bb9be30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1364ca8e503325c43e061e99173439f76aa5aa49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1364ca8e503325c43e061e99173439f76aa5aa49"}]},"branch":"refs/heads/master"},"c3421782f99d02b058327564a138d731915c6753":{"kind":"REWORK","_number":3,"created":"2021-10-09 12:27:57.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/52/813252/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/813252/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/813252/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/813252/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/813252/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/813252/3"}}},"commit":{"parents":[{"commit":"8cdf952c6be7e15c1e6a7589371bc827386d4cd2","subject":"release: Use Xena sources by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cdf952c6be7e15c1e6a7589371bc827386d4cd2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-08 20:37:56.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-09 12:27:32.000000000","tz":0},"subject":"Fix GitHub tarball URL for monasca-thresh","message":"Fix GitHub tarball URL for monasca-thresh\n\nChange-Id: Ie388938fc207c982c5650221baee354a5bb9be30\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3421782f99d02b058327564a138d731915c6753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3421782f99d02b058327564a138d731915c6753"}]},"branch":"refs/heads/master"},"461cb7efc66f6c83b9bd924486011572baea5cb1":{"kind":"REWORK","_number":4,"created":"2021-10-09 14:51:49.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/52/813252/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/813252/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/813252/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/813252/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/813252/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/813252/4"}}},"commit":{"parents":[{"commit":"8cdf952c6be7e15c1e6a7589371bc827386d4cd2","subject":"release: Use Xena sources by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cdf952c6be7e15c1e6a7589371bc827386d4cd2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-08 20:37:56.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-09 14:51:05.000000000","tz":0},"subject":"Fix GitHub tarball URL for monasca-thresh","message":"Fix GitHub tarball URL for monasca-thresh\n\nAnd the options docs, they were misleading.\n\nChange-Id: Ie388938fc207c982c5650221baee354a5bb9be30\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/461cb7efc66f6c83b9bd924486011572baea5cb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/461cb7efc66f6c83b9bd924486011572baea5cb1"}]},"branch":"refs/heads/master"},"556e8f9eeef71d240bef961cee0e3e5d357e3711":{"kind":"REWORK","_number":5,"created":"2021-10-09 16:04:22.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/52/813252/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/813252/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/813252/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/813252/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/813252/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/813252/5"}}},"commit":{"parents":[{"commit":"8cdf952c6be7e15c1e6a7589371bc827386d4cd2","subject":"release: Use Xena sources by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cdf952c6be7e15c1e6a7589371bc827386d4cd2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-08 20:37:56.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-09 16:04:20.000000000","tz":0},"subject":"Fix GitHub tarball URL for monasca-thresh","message":"Fix GitHub tarball URL for monasca-thresh\n\nAnd the options docs, they were misleading.\n\nChange-Id: Ie388938fc207c982c5650221baee354a5bb9be30\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/556e8f9eeef71d240bef961cee0e3e5d357e3711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/556e8f9eeef71d240bef961cee0e3e5d357e3711"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
