)]}'
{"id":"openstack%2Fkolla~813304","triplet_id":"openstack%2Fkolla~master~I5ec4e8cfc6048954aaf731808afb70ceb67b6ee5","project":"openstack/kolla","branch":"master","topic":"fix-update-alternatives-centos","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2021-10-12 08:57:02.000000000","reason":"Michal Nasiadka replied on the change"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2021-10-12 09:12:33.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-10-12 09:12:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5ec4e8cfc6048954aaf731808afb70ceb67b6ee5","subject":"Add missing reno","status":"MERGED","created":"2021-10-10 20:37:52.000000000","updated":"2021-10-12 15:51:06.000000000","submitted":"2021-10-12 09:12:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"813304-fix-update-alternatives-centos","meta_rev_id":"7f6cb722d2bfa81822d0a123f61a54a5d0b27748","_number":813304,"virtual_id_number":813304,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-12 09:12:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-10-12 08:57:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2021-10-12 07:50:59.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-10-12 08:57:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-10 21:55:01.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-10-10 22:19:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-11 10:02:03.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2021-10-11 10:28:11.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-10-12 07:50:59.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"2fad7cb30bbb59063520f148925cd71f92d71f95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-10 20:37:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20d2dcbbbe198f21d25801de623a95020c9c0083","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-10 22:19:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4f376b3f37774ada8cbe2c70891030f6 : SUCCESS in 29m 24s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/72505b3c6ef54501b30ed8c85106b0a1 : SUCCESS in 34m 33s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/98362d1937974f139fb00037344f3cb5 : SUCCESS in 1h 08m 11s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7491f9caa3814d1fac97b4a7887c9b3b : SUCCESS in 1h 06m 08s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/05a2185f2c3f4cf7a8668395d1fe72ac : SUCCESS in 1h 28m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/716bf71e2bb84013872a29ddd38969b4 : SUCCESS in 30m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4e29edbef5b54b54977166c9a560fe5e : FAILURE in 15m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c60090af0c6f430d99fc69590c2daa1e : SUCCESS in 1h 04m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e791ff0661904cb7bc3526b9f620474a : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/216319e8b46849639d0f8ad8a12a3276 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d9229d438634ced9592657baa95dcdd : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c68e75c4a834bccb544bcf336dcf5d7 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/765bf39cf9224098bc1159e142877b45 : SUCCESS in 9m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/79dfaa9f2ad7453ea38f6c92f6623ea3 : SUCCESS in 21m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1852201918724afc81b9840040477a1d : SUCCESS in 35m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cd09c722ba2346e68d95825eae17937e : SUCCESS in 50m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0af214768c47414b95bb1cdf7548fc0c : SUCCESS in 52m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/de442d0cefd54d99a71e45ca86b132f7 : SUCCESS in 1h 03m 34s","accounts_in_message":[],"_revision_number":1},{"id":"d45a13da2d885db5fd113ca33c36fc14540b6f7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-10 22:55:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1aa219f8bfd146259d4b759d17f98f32 : SUCCESS in 2h 03m 53s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/45356f6fc75c4c8ab4e5da78c7aa9422 : SUCCESS in 1h 32m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/26e4a26be08746e99475df7c9ebd2bee : SUCCESS in 2h 15m 16s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a5d343f244c43c597dada8ae27f5172 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31053765f55d43859e8e086d47d33cb3 : SUCCESS in 7m 04s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e9a3fc73f659438782f83d7c3e07bf4a : SUCCESS in 1h 38m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a08a2c9ec1506228ae41696be41608b60b98c473","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2021-10-11 10:02:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eaa38e0edebbb9941f7b1b33337e6c40024132c2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-11 10:28:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5db25aced49b5937d89a3dd3a4f69646d31108dc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-11 10:32:35.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"cecc3b0f3b25cff980f9bcb7fb22fa9455555e3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-11 11:01:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0975cdf497ee4a01a82bc649ce5c496b243a954e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-11 11:03:32.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f85cb4eb8f0eee11e4ec9a0b5dcb1588b4d13f3d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-11 11:19:11.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"1f03641b165811413ca8017b56a051221bf87f88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-11 11:23:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fd9236a35242a576a217a1959ead3057d768752f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-11 11:25:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4d89f8822bb9e7c2828b82fe75192b6baff958c0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-11 12:52:26.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"372a3cc43bc0c8d36cb56b42cdd994248c483618","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 12:58:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/877b5bbbe7444e81870f324391095e7c : SUCCESS in 26m 18s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/91786c7afa2d4673a9a3a2f38b670244 : SUCCESS in 33m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6241386d095a4770bca22bbdfa7b33a2 : SUCCESS in 1h 15m 09s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4eff5724c89b4e028c4ce6bcaac27a03 : SUCCESS in 1h 13m 56s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ec7d962dbd684b09b94809f598aa321f : SUCCESS in 1h 30m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/55a85d6c38fe4c56a379fd15617a9f1b : SUCCESS in 27m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e539286dceba4cb3bbddce999f68268c : FAILURE in 24m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/80e6619025424817a5f64ef953e034b9 : SUCCESS in 58m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eea2974cdf84176954315e174f765bc : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9299623f5364b6aa604f7a3b9753306 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcbf5da354ea479a8079bda0d81f1dc4 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e499b9b37c6e4ad0bd3753c7ec2600e5 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c33e60f2d4524a0c9df10115820a406e : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1ebd8ba328942b5a0cdf966f0a27109 : SUCCESS in 5m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2625dce1fbde49f5b26ec011186285f5 : SUCCESS in 21m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e9997c5443714bf683592fd2ca60cbb9 : SUCCESS in 29m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1ff80ef377334e1f94c6895478197ec3 : SUCCESS in 52m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e6052b2996b94583a01e38458935e08e : SUCCESS in 51m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4c321791742845bf9050024a71aa9f30 : SUCCESS in 1h 06m 16s","accounts_in_message":[],"_revision_number":3},{"id":"626ee60d8395042d9f41c9e3617b94310456dae1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-11 13:05:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"08150eac52eeb3d98c5b7f6a69389ce778025a62","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-11 13:06:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"12792ab297974c6bc17722a4d483c5089f3a1918","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-11 13:14:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f465aed148e97fb3b71859ca644c08a8c3e00d88","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-11 13:14:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f60e7eebf59a96fc45388591b837ef34971d3342","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 13:23:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2722933baa9d4d59bd9986226bbf1dd8 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/495f439578fb4f6a80d499403ea8a0be : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54ce4342fda5426ab3f5bc8451f4668e : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":4},{"id":"aff74168fd0e69d97723f1148a599693f464a5b2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-11 13:24:17.000000000","message":"Topic neutron_update_alternatives removed","accounts_in_message":[],"_revision_number":4},{"id":"94bdce55ac648f6ff9ce3c90410442c9a6ac988c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-11 13:24:21.000000000","message":"Topic set to fix-update-alternatives-centos","accounts_in_message":[],"_revision_number":4},{"id":"a43557abe0f2702fe3bc20007fe573ccf7a8b375","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-12 07:50:59.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3cae103bb63940e9c7818f22e580d03a9bb40209","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-12 08:57:02.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d7be264e4f6625886838e917da451a2f37fc6114","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 08:58:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"692052f6e6762e83ab6ac3e6fd14c6edf67b3f74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 09:12:33.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a9f594fc5034681a7edc3af0e075c07 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/372480cd48ef42e697e740c7307f4b11 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/639a1552207e45899a8eeb3e3f421337 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":4},{"id":"6016dd848fd15e1f9a360324801e89cd9d4771f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 09:12:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"a6d1edd65ab26b138ecaa16df9174599388f394a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 09:14:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11774b5ed0594e12bfb385a254fdc5a0 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d317292825654590a2adc169e6d43a53 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":4},{"id":"c09a9bcaec5497a0050c918ad5fad586c1b2e9fa","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-12 13:51:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ffd3552cab2a378750a67f6b7adfb740b9a1fae2","tag":"autogenerated:gerrit:revert","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-12 13:55:44.000000000","message":"Created a revert of this change as I487f0f5e795b27faf39530d459df13d816082fa2","accounts_in_message":[],"_revision_number":4},{"id":"900e07413c5d0ee6088e196f2a45224bbc7545e1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-12 15:13:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fe59521e6e8c76ada88c8908d06675491302ea78","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-12 15:15:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"59848d6a0061a3c7b4ff5b136e7d0e0f9f603624","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-12 15:18:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dd589cfe88c28a92eeab4a5d4124eea375d7908b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-12 15:22:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7f6cb722d2bfa81822d0a123f61a54a5d0b27748","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-12 15:51:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1b6d1bf16215f7a166b6b6578ddd9ca242751d3e","revisions":{"5aa44e30a9eb5d220e21e788d9fb7d8353f855b6":{"kind":"REWORK","_number":1,"created":"2021-10-10 20:37:52.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/04/813304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/813304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/813304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/813304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/813304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/813304/1"}}},"commit":{"parents":[{"commit":"ab92558f41f59e5cb265730ea1093ad96926def9","subject":"Merge \"Use RDO\u0027s Xena Delorean packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab92558f41f59e5cb265730ea1093ad96926def9"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-10 20:32:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-10 20:32:38.000000000","tz":180},"subject":"Fix KOLLA_LEGACY_IPTABLES workaround","message":"Fix KOLLA_LEGACY_IPTABLES workaround\n\nAdd the forgotten colon\n\nTrivialFix\n\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nChange-Id: I5ec4e8cfc6048954aaf731808afb70ceb67b6ee5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5aa44e30a9eb5d220e21e788d9fb7d8353f855b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5aa44e30a9eb5d220e21e788d9fb7d8353f855b6"}]},"branch":"refs/heads/master"},"28cc1f61123522de9e72b3379e5180fe5c87121a":{"kind":"REWORK","_number":2,"created":"2021-10-11 11:01:24.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/04/813304/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/813304/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/813304/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/813304/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/813304/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/813304/2"}}},"commit":{"parents":[{"commit":"ab92558f41f59e5cb265730ea1093ad96926def9","subject":"Merge \"Use RDO\u0027s Xena Delorean packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab92558f41f59e5cb265730ea1093ad96926def9"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-10 20:32:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-11 10:58:42.000000000","tz":180},"subject":"Fix KOLLA_LEGACY_IPTABLES workaround","message":"Fix KOLLA_LEGACY_IPTABLES workaround\n\nAdd the forgotten colon\n\nTrivialFix\n\nChange-Id: I5ec4e8cfc6048954aaf731808afb70ceb67b6ee5\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28cc1f61123522de9e72b3379e5180fe5c87121a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28cc1f61123522de9e72b3379e5180fe5c87121a"}]},"branch":"refs/heads/master"},"4a540e865fc4d0599ae697aed8f4b6fa1bf7ad77":{"kind":"REWORK","_number":3,"created":"2021-10-11 11:23:50.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/04/813304/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/813304/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/813304/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/813304/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/813304/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/813304/3"}}},"commit":{"parents":[{"commit":"ab92558f41f59e5cb265730ea1093ad96926def9","subject":"Merge \"Use RDO\u0027s Xena Delorean packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab92558f41f59e5cb265730ea1093ad96926def9"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-10 20:32:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-11 11:23:23.000000000","tz":180},"subject":"Fix KOLLA_LEGACY_IPTABLES workaround","message":"Fix KOLLA_LEGACY_IPTABLES workaround\n\nAdd the forgotten colon\n\nTrivialFix\n\nChange-Id: I5ec4e8cfc6048954aaf731808afb70ceb67b6ee5\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a540e865fc4d0599ae697aed8f4b6fa1bf7ad77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a540e865fc4d0599ae697aed8f4b6fa1bf7ad77"}]},"branch":"refs/heads/master"},"1b6d1bf16215f7a166b6b6578ddd9ca242751d3e":{"kind":"REWORK","_number":4,"created":"2021-10-11 13:14:16.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/04/813304/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/813304/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/813304/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/813304/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/813304/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/813304/4"}}},"commit":{"parents":[{"commit":"ab92558f41f59e5cb265730ea1093ad96926def9","subject":"Merge \"Use RDO\u0027s Xena Delorean packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab92558f41f59e5cb265730ea1093ad96926def9"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-11 13:13:08.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-11 13:13:08.000000000","tz":180},"subject":"Add missing reno","message":"Add missing reno\n\nAdd missing reno for I151420d3036410dd5ce6d90907d72dd3c20643b5\n\nChange-Id: I5ec4e8cfc6048954aaf731808afb70ceb67b6ee5\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b6d1bf16215f7a166b6b6578ddd9ca242751d3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b6d1bf16215f7a166b6b6578ddd9ca242751d3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
