)]}'
{"id":"openstack%2Fkolla~813305","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I151420d3036410dd5ce6d90907d72dd3c20643b5","project":"openstack/kolla","branch":"stable/ussuri","topic":"neutron_update_alternatives","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2021-10-13 15:26:22.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2021-10-13 12:43:02.000000000","reason":"Maksim Malchuk replied on the change"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-10-13 08:21:09.000000000","reason":"Mark Goddard replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-10-13 15:26:22.000000000","reason":"Change was submitted"},"30523":{"account":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"last_update":"2021-10-13 15:26:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I151420d3036410dd5ce6d90907d72dd3c20643b5","subject":"neutron: Fix update-alternatives on CentOS","status":"MERGED","created":"2021-10-10 20:42:44.000000000","updated":"2021-10-13 16:22:36.000000000","submitted":"2021-10-13 15:26:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"813305-neutron_update_alternatives","meta_rev_id":"0ecc6a82d5be5ca6fb0017759d146a0cfc2f7e68","_number":813305,"virtual_id_number":813305,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-13 15:26:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2021-10-13 08:21:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":2,"date":"2021-10-12 18:58:22.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2021-10-13 08:21:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-10 20:42:44.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-10-10 21:54:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-10 21:56:13.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-10-11 09:58:50.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2021-10-12 09:04:53.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"6a8d24dba8f7f27ea7f5d22b30f9ac23bc0f2f16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-10 20:42:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51fc0c88305ec3185c5dc231779ca8070639f66e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-10 21:54:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4b5935f3e0584275a30613b0b235464a : SUCCESS in 23m 44s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f7aa997bae06409b9bb26bd00f22afeb : SUCCESS in 29m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2d82202a83274a6bb39e7d9cd465808f : SUCCESS in 49m 41s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6841eeb98db44ee4acaad0c6ff6784bc : SUCCESS in 56m 09s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ba6ff0b9ee4f4475bce41cf497d900bb : SUCCESS in 1h 04m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/63db374b05774c5397580b5979ffe8d5 : SUCCESS in 22m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c5c106388e5d4e7688f25d66580c05c6 : SUCCESS in 20m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/002405e37dc04cebb332737dbcaea4eb : SUCCESS in 56m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc981ea282c942d080b0a0dc37dcba78 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f70d618b3bf48d3835efa71b4c48e3e : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cab91d10799547d5a1362351947385a0 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3188f5c2b2374d31b128aa7bfcc3ed58 : SUCCESS in 6m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1f69281792cb44ed8489ea6984d38294 : SUCCESS in 20m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b6187033e8ac4cdf960861db82dbecc1 : SUCCESS in 29m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b16457b05d649aa968ff9e645386658 : SUCCESS in 55m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2b8df7470f98400f8294c13322f6c1bc : SUCCESS in 47m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6358404787f2421e8d9c2b7d043c9132 : SUCCESS in 1h 00m 58s","accounts_in_message":[],"_revision_number":1},{"id":"6f6d61b099db3b3784c3918b55b1d352a360edd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-10 23:17:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/0d84ab53ab834de793180760cb054c18 : SUCCESS in 2h 27m 33s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2ad04bb3a9674f1ca92a6be5ec978326 : SUCCESS in 1h 47m 52s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cd2ce87788e047b796597f1c4e0fd954 : SUCCESS in 2h 31m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/961f01da39654648877829ef43b6e1ce : SUCCESS in 1h 47m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd3cad88c09a64d42c3b257cc4fdcb595be6d3b2","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2021-10-11 09:58:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a21ee0bd599a7c1d67c7694c755396795f2204a4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-11 10:08:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cb06f7acd8c394d7f4d328898a854f074730c0fd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-11 10:21:56.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"942d2a7e1a6e46fdc9179f3862febccf6b973920","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-11 10:25:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"460b54767c1d051717b26b108f7e10ea94985c62","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-11 10:27:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"158b4417326383edb11dd6f355bd81d205ea18b9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-11 10:27:58.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"5ff6084880542d8ce282b2f633a1f0f8c8d3f34b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-11 10:28:38.000000000","message":"Patch Set 1: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc5af5a0fc83b952f971f3e24d506b255a98253e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-11 10:31:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ee1715057e495f15885b36bdfb0ff83257fdf99","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-11 12:41:01.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"f086fa7c351c6572139782cf95bdb95e1f2b9267","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-11 12:54:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ae293831a58d822952fbcdd0df2bafac8b679af","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-11 12:55:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a495fc0bd79d3c09083e21f6271cfe3bb6ec263d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-11 12:56:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9ac04faa41a2f1be546f0c97e09c64d06235712","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-11 13:06:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"87afeed2353d55e453da225c1fe532ede44813fb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-11 13:07:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a315a6ad6d8395aecd326d2d99941f397496ac60","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-11 13:08:01.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"237def39a89a4d3f9efdb2b03c6e30be52aad1c1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-11 13:17:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"832ea504416c5e70696d65d8e8bf7f4a67b2d410","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 14:21:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8093736e1ccc4c1587efbca95afdbf4c : SUCCESS in 26m 57s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4dfd20c2399743f8adb4d9b52881e4fd : SUCCESS in 31m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4dc5404f0d8c424caadba281dfb0d63e : SUCCESS in 1h 04m 01s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/43cc15235b4340f0a7eb6cc8fe107f85 : SUCCESS in 1h 02m 23s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0ff632698a8543939e700999d9b84cc4 : SUCCESS in 1h 11m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/15fb7180a8704daf94dafe4ca4947753 : SUCCESS in 22m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b7af3f2ab83c4c71b7758c08fd48dc9d : SUCCESS in 17m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/466121f3c32846d88901cd459169e6b3 : SUCCESS in 57m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbac1d060b24425d84204d887bd9c433 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9be175295e1c41c0868a4273d24270af : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c0f7afa6b72e4f6fab8f8af1cd954efe : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a61a6d7efea94c81a16b4c95cb69dad0 : SUCCESS in 5m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/035bb976a2b240f48daa48a3f421b576 : SUCCESS in 22m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/09d056ffb1af4a31afe956d0dc70a9cf : SUCCESS in 28m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b0f438afd95145489c186c6f580b8706 : SUCCESS in 57m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/860c196094664752854ea6799957b599 : SUCCESS in 52m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/22d27223d58d4023a6399c86d2ae66f3 : SUCCESS in 1h 07m 28s","accounts_in_message":[],"_revision_number":2},{"id":"67fbbf02e656942e616586c64676b1120c07e508","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 16:10:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/39edec132e0a443eb7519f81d20f3e7a : SUCCESS in 2h 34m 53s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d9c195201c454adba3035b4b44067731 : SUCCESS in 2h 34m 59s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ecb3cc76acb649d1a8057d4e1386340a : SUCCESS in 2h 52m 38s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f5188088bcc64e148aba839180a32a5c : SUCCESS in 1h 38m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4c6421068444100cf2af70a5c5ac59eced51ea6e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-12 09:06:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6d1823cd6478c35e99036aa62c2c4ab75d7c787b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-12 18:57:14.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f6a124098e5736057a37f1ff804aafa1c6171826","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-12 18:58:22.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e823c98d8394ce02a60026e9de52748df2e5f9b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 20:20:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d0bf00e37a2c4722871169626eae1c75 : SUCCESS in 38m 03s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0447e01d5c8b48c1abd8b8ee86edfa5d : SUCCESS in 37m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9860e5f7d287442b9fc542a3b54587c3 : SUCCESS in 1h 00m 42s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/955e99e4f02048f1ad0fcd44d28b6109 : SUCCESS in 51m 41s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9abe5300a95747e7aa55ed05f1fa0934 : SUCCESS in 1h 10m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6df5cea5837349b49fe7208482252dc6 : SUCCESS in 24m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a42418439fa14aeeb175d8026f16eda9 : SUCCESS in 21m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/07d29fe6a8df42e8a4dce29f32a0cd9e : SUCCESS in 54m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47e378b7db144faf974c47ede9fb3f31 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83a7d53645ca42abb29921ec122d2806 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5baa6202d39d4c60a7236cc741df9891 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0545d73a512241f8a68c0333a8b44424 : SUCCESS in 6m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b45573f9b1b840d98ab0eef7cd8e3c69 : SUCCESS in 22m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/64f5a29913c9455f8bdc1b308e49fc74 : SUCCESS in 32m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f772ba2a9a8148c6893c2eeedc5f3dac : SUCCESS in 54m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bc8b08d27d0540039e04ddc7e28e6b14 : SUCCESS in 50m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5182995a2419465dbb24987fc650e712 : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":3},{"id":"dff584f675378ef60edac16b4bf8ae9158b5b3bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 21:35:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/a00a7bb453ee4429a77d4d0d67ecdb29 : SUCCESS in 2h 30m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fe3ee6d1c4a3406eb0a9de8173944203 : SUCCESS in 1h 44m 48s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ce58062cc013464293ab69caabd8c263 : SUCCESS in 2h 35m 57s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f4c326ed2ec643fd889bb9bbd3e48210 : SUCCESS in 1h 45m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c25ba7fd4088ef2472ef972b2ae0c8938ab4d2bc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-13 08:21:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cb47a8d732151561aef92be476a3530bea8a6cfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 08:21:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"da3a3b75a5c366f48b4ae736d7910507e696f350","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 09:36:15.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/569674ed87ec4b12843d4308daf5f04c : POST_FAILURE in 25m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2aaa003e19ab44f4aadf1a0e11da832d : SUCCESS in 54m 32s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/803de3f17d0044e08a8e31b5018be5b8 : SUCCESS in 1h 02m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a707a39418c4e77b1f470259aa6f32a : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f3f26a20d6f4345b2041c13cc8fd87f : SUCCESS in 8m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5824945ec1ff44d4a2bef5cb423b0d7f : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e60aff19ca554700b997a7ba3437ae38 : SUCCESS in 8m 56s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/771dc2c512c24b1bae1e554cad02a79b : FAILURE in 23m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2410e09f03604f6a8ad3cad722091505 : FAILURE in 17m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4de4e234d5a845a7bb19f7c97b13ebaf : FAILURE in 37m 33s","accounts_in_message":[],"_revision_number":3},{"id":"d56be01a22f839bd617374ac3d0efaa1f3a33a28","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-13 12:43:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a5d1840c42e0ffbb2fac27c0d189bb8099ebc0f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 14:05:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1875d88eb9e340faa61a3efb0211b312 : SUCCESS in 33m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1e707e73c0a44a7f98297bfbe246e236 : SUCCESS in 38m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/15a7b2e5835542718af260c9d94751f0 : SUCCESS in 1h 11m 02s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/fc6a5d2cd3cf49969e8409af96b3c6a2 : SUCCESS in 1h 10m 42s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3ff32b350ba14a05b4bcf9a5deb3380d : SUCCESS in 1h 18m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6631ba4e74e942a0b90b0f6bf77cee29 : SUCCESS in 29m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e9bd6471742349949e593d0735f2ecb7 : SUCCESS in 26m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fd072d22ae744f8abf26aa25a47f3e32 : SUCCESS in 1h 05m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33450813aa8947f39b09406da2a2000a : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51eb083dc0994fdebe7db6c56096f9ad : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6303c99869a34a3986159a9dd6a27dbf : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeff4f5f6352483ab261b6ca83c8d112 : SUCCESS in 9m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3e858b58dc9e440e8221cdc3c327c430 : SUCCESS in 23m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4e283fba99dd4a5ea9f965307f6bf4ef : SUCCESS in 30m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dcfcffdb3fb84b8597384d2ce0fd2aa3 : SUCCESS in 54m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/18aaf950991a4c149faadef0b9568645 : SUCCESS in 57m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/829ba69ca0054597a1dd610889333b95 : SUCCESS in 1h 02m 57s","accounts_in_message":[],"_revision_number":3},{"id":"624d1c564bd368517d3504c0f2acec3dc025cb3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 14:06:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"969032a1af2addd9f7c16409dd184c82b5efd168","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 15:26:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/af030f2b06e14f72aaf1f4fdf5d5902a : SUCCESS in 31m 39s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9fc87967bf57481ca6a222c117d13308 : SUCCESS in 1h 04m 33s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f5c4034dc4aa439fb7358a3f2ae8a0a8 : SUCCESS in 1h 14m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac61100382c347969c4e21a4134bcb4b : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06b37cba48334653a72111b413fdfd84 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/275af460980042ae9d06c5d8dae024c2 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c43dde4f6a74e158f1459e6a2d986b2 : SUCCESS in 7m 07s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/21558f21e6c2474e9ccd10517c546f85 : SUCCESS in 29m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6e1f5fa9d36a4d04b9be876a33ffebd4 : SUCCESS in 57m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f64381ac530840efba83186ac2659751 : SUCCESS in 1h 09m 24s","accounts_in_message":[],"_revision_number":3},{"id":"8eb5ed1db05dd897a76e087cc0fcc3054981a0c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 15:26:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"384363d3bba0121e3155eab6e3b76feb95b73c90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 15:29:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77de5d5f3a1141ffb3f2e950777376e7 : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":3},{"id":"0ecc6a82d5be5ca6fb0017759d146a0cfc2f7e68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 16:22:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/30d662260f88465bab1c1a4d26a2c29c : SUCCESS in 2h 02m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c7e993d47f0949a7bb9874aac16670cf : SUCCESS in 1h 30m 40s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/71b88086f701426ebe22cdbf83e37900 : SUCCESS in 2h 18m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3b4c80082cad4c86ab3715ec20636266 : SUCCESS in 1h 30m 19s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f179f8383bec039bd82670217faa7e16ef17fcfa","revisions":{"84c27a50e6ea97121ad8c222f635efab4b2a02f0":{"kind":"REWORK","_number":1,"created":"2021-10-10 20:42:44.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/05/813305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/813305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/813305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/813305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/813305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/813305/1"}}},"commit":{"parents":[{"commit":"23af39b046a1c1ce9802d4f358410af99d2f10b9","subject":"[debian] Use upstream MariaDB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23af39b046a1c1ce9802d4f358410af99d2f10b9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-03 15:26:13.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-10 20:41:06.000000000","tz":180},"subject":"neutron: Fix update-alternatives on CentOS","message":"neutron: Fix update-alternatives on CentOS\n\nThis change also includes I5ec4e8cfc6048954aaf731808afb70ceb67b6ee5\n\nChange-Id: I151420d3036410dd5ce6d90907d72dd3c20643b5\nCloses-Bug: #1884778\n(cherry picked from commit 7a4ba91fa667a295a7e1023abfaa15e0874a371a)\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84c27a50e6ea97121ad8c222f635efab4b2a02f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84c27a50e6ea97121ad8c222f635efab4b2a02f0"}]},"branch":"refs/heads/stable/ussuri"},"a80f314011f5035b29feb473e7ea7b4f36e65c1a":{"kind":"REWORK","_number":2,"created":"2021-10-11 13:06:55.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/05/813305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/813305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/813305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/813305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/813305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/813305/2"}}},"commit":{"parents":[{"commit":"23af39b046a1c1ce9802d4f358410af99d2f10b9","subject":"[debian] Use upstream MariaDB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23af39b046a1c1ce9802d4f358410af99d2f10b9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-03 15:26:13.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-11 12:39:01.000000000","tz":180},"subject":"neutron: Fix update-alternatives on CentOS","message":"neutron: Fix update-alternatives on CentOS\n\nChange-Id: I151420d3036410dd5ce6d90907d72dd3c20643b5\nCloses-Bug: #1884778\n(cherry picked from commit 7a4ba91fa667a295a7e1023abfaa15e0874a371a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a80f314011f5035b29feb473e7ea7b4f36e65c1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a80f314011f5035b29feb473e7ea7b4f36e65c1a"}]},"branch":"refs/heads/stable/ussuri"},"f179f8383bec039bd82670217faa7e16ef17fcfa":{"kind":"REWORK","_number":3,"created":"2021-10-12 18:57:14.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/05/813305/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/813305/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/813305/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/813305/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/813305/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/813305/3"}}},"commit":{"parents":[{"commit":"23af39b046a1c1ce9802d4f358410af99d2f10b9","subject":"[debian] Use upstream MariaDB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23af39b046a1c1ce9802d4f358410af99d2f10b9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-03 15:26:13.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-12 18:57:14.000000000","tz":0},"subject":"neutron: Fix update-alternatives on CentOS","message":"neutron: Fix update-alternatives on CentOS\n\nChange-Id: I151420d3036410dd5ce6d90907d72dd3c20643b5\nCloses-Bug: #1884778\n(cherry picked from commit 7a4ba91fa667a295a7e1023abfaa15e0874a371a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f179f8383bec039bd82670217faa7e16ef17fcfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f179f8383bec039bd82670217faa7e16ef17fcfa"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
