)]}'
{"id":"openstack%2Fkolla~814783","triplet_id":"openstack%2Fkolla~master~I8e0c55e7995e5269f285c6d9c75e8c4b7b85a38b","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2021-12-08 21:07:07.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-12-08 21:07:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8e0c55e7995e5269f285c6d9c75e8c4b7b85a38b","subject":"Debian: enable building Bifrost images on x86-64","status":"MERGED","created":"2021-10-20 13:51:26.000000000","updated":"2021-12-08 21:08:55.000000000","submitted":"2021-12-08 21:07:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"814783","meta_rev_id":"9888f3d25ea6274d71650852f87afc804dc81aa1","_number":814783,"virtual_id_number":814783,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-08 21:07:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-12-08 16:27:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-12-08 16:27:20.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-12-08 21:08:55.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-12-08 16:27:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-12-08 16:27:20.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-20 15:16:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-20 19:13:23.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-12-08 16:27:20.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-12-08 16:27:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-12-08 18:31:45.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"71e329d468d00de370376d54fe66b54ca213feff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-20 13:51:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6aa5f08d0ed3b22f8a576cfa23f7f02fba81fe8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-20 13:52:05.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"34679004a08085282870c3e9d62101edf754b2c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 15:16:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/0e3e4dc5a77e497789506a0f4db6c977 : SUCCESS in 28m 47s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/eeff58c1bdd647a4aa88e28c1eaeb6bd : SUCCESS in 39m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ab00a9b5e4b8450683618d07b4be9f13 : SUCCESS in 1h 10m 29s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e419cfeea80a47d59bf6f267670bf8d0 : SUCCESS in 1h 05m 44s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4fe9f49126ea45a1b2ed4fdede7f9b84 : SUCCESS in 1h 17m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dc912c7d72fc44068d56039ccbed1ab9 : FAILURE in 30m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c241b0beeac0446eb085d1953cc77d0a : FAILURE in 21m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/662fb44c4de74ce69d0259b1b328ad01 : SUCCESS in 1h 06m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c93faae8896b4176b2f1610c822149e0 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b211a243db2f431188d21ffdd7abcdb6 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b0b93dfb09f4caba661f943dec062c2 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24429d160736451bae6944fdfd34bc7e : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f420947b8af48c0897bce03911fac9e : SUCCESS in 6m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7ee2a08ab0214f59812f503a08ec9723 : SUCCESS in 26m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/27ef3f5d74f0469b89f339c7e69bb868 : SUCCESS in 24m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5893499af1434da5a43c3ae6ecea368f : SUCCESS in 51m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aead92f1c20b4bc581ddbf81faee89f0 : SUCCESS in 48m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7ceb5a4651b64dfc9146c181ed7fdddb : SUCCESS in 1h 12m 05s","accounts_in_message":[],"_revision_number":1},{"id":"5e7bcd6df6e457b63c948df581bb4aef5e305bfb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 15:37:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/55a957354b3a4296b4694a7175d45437 : SUCCESS in 1h 26m 11s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c098847fb78c4aab8d49cf7870ef7bb6 : SUCCESS in 51m 26s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e4e315ac3d434e0e8a4ab607c1970a9f : SUCCESS in 1h 43m 18s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/095f5ddba4ca4ca497ccd073439422a9 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2241138df7a4437992492e7502d49b05 : SUCCESS in 7m 45s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/70de174e80864558960fc32a36f31a6a : SUCCESS in 41m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"80770add402833d16921b95906157d8e13d5609f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-20 19:13:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"12577954c1f0397a87b3d12498dc73806ae6d099","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-27 10:04:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"73d71e66a64801a9d3fa7bc4ae55b7d8e66a52af","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-27 10:04:47.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"e519e68f3e7f7dc43c7a12450e8b8fc9ac69f02d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 11:40:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4d41bafcdcba4fb7a71d688e517e6271 : SUCCESS in 54m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e8c4d1335b5a4155b3c771d4a5c63e29 : SUCCESS in 31m 53s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ed4a43e358444d08943ee4d7e349b170 : SUCCESS in 1h 17m 16s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/cc999af97053435f8a77f4d85034747e : SUCCESS in 1h 34m 39s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/67bd62086c1a492ea64f5aa196efe8ba : SUCCESS in 1h 24m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b3d69770ebd34335b0189d0264fcec54 : FAILURE in 42m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9c7e788bd14141008697d4b11f58492c : SUCCESS in 16m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/722447b522f64c43948ca6be835d1d83 : SUCCESS in 1h 02m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aabc5b234e00452488260bcc8b032aa8 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d9dc4f7c1b74a82aab852a5e7566c94 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80d453486c8e46c7a91fdc2793079e9b : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63f5258fd8ea4c7196e8bcab73efec44 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/065eff4421d74f42bc73bb49f94c5a8d : SUCCESS in 7m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a345800c3380466aac512d2c527cf8d7 : SUCCESS in 33m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/303f9ea5ceac42e3829cf6c373772e0e : SUCCESS in 58m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d10c2d46eac4e09ac2c50a3c426efb9 : SUCCESS in 52m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/077e4a038c734ed0ad8ea6b148eaa228 : SUCCESS in 40m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/92688119025942cabc6f6b5e4060d8b7 : SUCCESS in 1h 01m 28s","accounts_in_message":[],"_revision_number":2},{"id":"271527f6c4d6fc2feee6be6c071099d4e092b0f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 12:06:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e3009298ce6748aebcbeb444f083ceb8 : SUCCESS in 1h 34m 02s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4e9cc49f08184dd88fa90be6d4003525 : SUCCESS in 1h 59m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/03490de2389240f0955793378e6a38d6 : SUCCESS in 1h 42m 38s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/386c689cfec84a4da61e70fbd75bf766 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5054f6a4330e434db9a442115f7697f9 : SUCCESS in 6m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/be09e370f07c47998ededa2d2f8715ac : SUCCESS in 1h 02m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0ad6fe705270112cc0580ca434e3fb71416c00f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-27 21:12:36.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d8374c82ca9715936aad61c820e6049d94c36596","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 22:39:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/28facd11e4de4dd7a13266a0d14c43b8 : SUCCESS in 1h 16m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d4f486c7cd184ad7aa0f8b9d0a0090f2 : SUCCESS in 36m 30s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ebe53e41b9104231981376a37a91aff3 : SUCCESS in 1h 20m 33s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d7ea3f14c5145f58224b2f5de2efa14 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3bfc9519b3f54d22a55b262ad1da6634 : SUCCESS in 6m 33s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0068c0441fe342909b091833579b0ba6 : SUCCESS in 1h 01m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9672ea4ca18329cac26ee5f1f8f7d3dbaf48bebd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 22:39:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/47d4689204c2475a951817debdcae0c8 : SUCCESS in 26m 49s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0f4f854c0e9a4e2d91db72afe25eca5b : SUCCESS in 51m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5aef6d731ba14fb5997f179290cde08a : SUCCESS in 1h 11m 29s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a7d3b8b0b5ed4abbb1a7f6251e38e21e : SUCCESS in 1h 14m 23s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2a22d4ab67fc4aecb9fefe0a440ef126 : SUCCESS in 1h 25m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b0e82ea2238c4f1d96891b81f879f1eb : FAILURE in 35m 14s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f9be06e5badb479b99c30f5579476764 : SUCCESS in 18m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e175d4bb231a4e29b1942ea7b97bc424 : SUCCESS in 59m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07758bb23ee3425abd6cc6d7376f9c6c : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e45df927c5e40bea0c42e300d9489dd : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e41001990b314d7d8f88ff1646010991 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb5eb22c52c24964a8bcbecbdb356fd3 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/713d7dcf316a46a39f2a7320c25b2008 : SUCCESS in 7m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8a06fa85f38f4542b2131638e775a6b4 : SUCCESS in 21m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b96250dc3d5d416e906c3a119fafd8c3 : SUCCESS in 32m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf33bef39d6741b18b8f253cd868397f : SUCCESS in 52m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1ea9aab6d3d04be0b6a5635e8d05b263 : SUCCESS in 53m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/647bc0c15d99420f883625e60171367a : SUCCESS in 1h 05m 30s","accounts_in_message":[],"_revision_number":3},{"id":"e327174541ea45dbdb09238361eb02e843306f80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-02 09:24:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7d891f4453d469d338808ee0f8ac628ca5492b88","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-02 10:02:32.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"1b5612fa966faba4bb8e5c8d7cf924d479e1a781","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-02 10:02:55.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3e410821059d47e099dc400896c240aed32f90f8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-02 10:32:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"428713f2ff2332fa2b4ab9d6162e170684523620","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-02 10:51:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"49892738c0f32c20c33b558ba2d7b7e279c0920d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-02 11:04:26.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a7fc8c307475ed6cfe2b1ae6be20bd9173018f5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 12:15:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4917587cdf214c728ca44963ef893525 : SUCCESS in 25m 25s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/dee5c9a607a1497eaeff50e30d7a04e8 : SUCCESS in 50m 26s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2aa84bbb71e34254b277c5241bc526ba : SUCCESS in 1h 09m 53s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/db53cba121b047f08ad2a75402937552 : SUCCESS in 1h 13m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/26acb31abfdd41d1b5afe2ad588fb84e : SUCCESS in 1h 20m 17s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0686cebbcdd3402eba6393d009e33286 : FAILURE in 38m 57s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e3386788a250465b8d03b986900f0451 : FAILURE in 19m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fd1c8e31e2d74ca78aefcff549dfca7a : SUCCESS in 59m 14s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2a8af417d571452286d9b5618c6d7446 : SUCCESS in 1h 19m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f929305d8b04226a84785be7470f304 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8dcbd9af32a141edb9650809c871daa6 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26524075c96b4b38a7911eb9cb6125da : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e056258add7f4033ab7e8699a84c0041 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed705d0e76914be9a8b85b1e8fba78e7 : SUCCESS in 7m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cc455227e33f4efea82bed1b658fafef : SUCCESS in 17m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/317dfffe24f84bd39785d0ad4bf0997e : SUCCESS in 41m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/51480e43e55e43209b054d5884fa03c2 : SUCCESS in 57m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f18a8f1c02274629b162740f7acb67bd : SUCCESS in 55m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/12e814344de14bd3a115676957f2eeec : SUCCESS in 1h 13m 46s","accounts_in_message":[],"_revision_number":5},{"id":"1704c07f1cc63652170212206eeace0219c6c456","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 12:51:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/bf3d2f40f8134e358a397e7534dfc5e5 : SUCCESS in 1h 47m 47s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f1684fb39dae4bb0889fdfeb34c7ba3a : SUCCESS in 1h 57m 40s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c8343088f8414d3cb29af5c1498cfeaa : SUCCESS in 1h 23m 25s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8edd41e6e0554d1a8cbf6057b88011d6 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c5b61222f487417cb442c9a694a6e940 : SUCCESS in 9m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ae9b9770f184a7f976c71fa1d563040 : SUCCESS in 1h 19m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"acaa7113b5f0ffeda348cc68a63994287e5cc19d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-01 16:01:52.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"7f258605c03ab93c192cda9b3f9aef1de963e8b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 17:37:53.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ac750f508f5f4b7bb70c1b5a306e9490 : SUCCESS in 23m 34s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/496d352d224d42fca7b212f1773067d8 : FAILURE in 44m 39s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/13e2bf8274bc40c08af50610766b2179 : SUCCESS in 1h 04m 57s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/cf7351108a6c46ebb25de5f4fd3a6850 : SUCCESS in 1h 16m 27s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6cc4df84d2ea4e529a52e5ee3c67852e : FAILURE in 1h 33m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ba505e7f5fbf43259d370ed8081bb191 : FAILURE in 30m 17s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2f1a607144a34c36be28cd9b5e58b3ab : SUCCESS in 18m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/46c2faee4c224484b2bc6be9e06717ba : SUCCESS in 1h 06m 45s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/718c1e37bfb54e70aee9a685bbbfe40e : FAILURE in 1h 06m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ce4eb0d67174192bdc8501172d9034c : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8659e2ca7e44689a102e452f83e1668 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39534e3839844b099fd1dd8606bb563c : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73f0a4d4ace04bb0b95269eb754220e1 : SUCCESS in 8m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e66f1cc1975344549b810659fc784ccb : SUCCESS in 19m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b6fb6eac58ec4eec89484e8b90345c90 : FAILURE in 27m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0b8716455bdf4cf5bdce6edf57fbd449 : SUCCESS in 54m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f7aa253934d64921a7d0b6b80db7c4d8 : SUCCESS in 52m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2d473abc502d47d6a6f1defd7fb2dc80 : FAILURE in 51m 48s","accounts_in_message":[],"_revision_number":6},{"id":"b3cc5c4dd10527a8be4b79e7cda45e62774b49d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 18:39:33.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3cfb81b9fedc41dcb3a1a4f5f3337700 : FAILURE in 1h 54m 02s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b97cfe674b5d44e0babbd2f137697084 : FAILURE in 47m 37s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6f5708bfd7b146fba97af366f1cc28ae : SUCCESS in 2h 03m 25s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c567416dd0544f73a0c4d2d7990f3e11 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/488c6be13872482fa6cb41c18457110e : SUCCESS in 7m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ee2262bcb464643878c1612272fea90 : FAILURE in 1h 25m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"877d70099e2de6577a396f5a5653452aef05277d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-02 09:07:46.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"e1eb7666b306417cb50cbf77ad6bfb4897db66ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 10:34:08.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ff49ed50a2104c5983a623d9ebc2a952 : SUCCESS in 36m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/65e0ae1cdc6949c699eda19aa5a811af : SUCCESS in 35m 51s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3350740d93444f08a9ae93d16d6cab41 : SUCCESS in 1h 14m 19s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4ad3e101927842ffba5c82fa6d0504aa : SUCCESS in 1h 18m 38s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/acea6329bfe24a8ca8cf0ae8a92a1857 : FAILURE in 1h 23m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7fc3b4cdd9af4dc6af7e7a4bdd923dd1 : SUCCESS in 23m 19s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cd23b124fdda45c0b7d04c41bdd47c38 : SUCCESS in 22m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8627d3090cdd4deda27011784784527c : SUCCESS in 1h 15m 00s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/45437a3e226849f58d5904ee6c192e56 : FAILURE in 1h 00m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb5e97873091402fb718a2b789694470 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15f513e715cf4351beb018e2736713cc : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9d5cd4946e844129d3c5dd673a33b66 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9edd3e69becd41019843d3582b1cfc56 : SUCCESS in 7m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/446d3490e62d421c8fa6b90467358c48 : SUCCESS in 18m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9def04a3ee094abdbdd9c9b6fb31a24b : SUCCESS in 33m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a77d40bbf8d247e0913b3d0e02ad95d6 : SUCCESS in 52m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5271cdc2f1af4c60b9a2a8a9f19a7f09 : SUCCESS in 55m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/212cdceaf4f7495e8b79baad03db4b79 : FAILURE in 1h 11m 45s","accounts_in_message":[],"_revision_number":7},{"id":"cef4c5c3f36c098619358a2c98bf9406404ea7b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 10:37:07.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b969977fe6044fe28339d537011e42fd : SUCCESS in 1h 23m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/07d95502b4ef4f36860b98538a2443ac : SUCCESS in 1h 07m 28s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6a16fd80e9fc41af97ed49304a7a3e7a : SUCCESS in 1h 24m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d0a2658c4874314abaf8320d142da3a : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/173dfd7c7a2f47bfad6c76cb0579f867 : SUCCESS in 6m 38s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/990b1f4225ef4fcab66cee5cb5b52191 : SUCCESS in 37m 38s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e29932347c6ca334f5fd566c52eca29ef54e5ee6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-05 19:03:18.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"d4e7103de795794f8858790408424f8e8bc5d50f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-05 19:07:23.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2cab32fba6750c459c96f26de70313c045386ef6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 20:24:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/0c061a398d1d44a9a65d077445090bb2 : SUCCESS in 37m 50s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ae282fc97a9f4e09a7ca81237becb94b : SUCCESS in 42m 32s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/35f37a97f8e441d9ba95f8718d39b2ad : SUCCESS in 1h 12m 07s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f562d21ba0d94bc8ad0b56e9de11bbf9 : SUCCESS in 1h 11m 16s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/cafa1482b2584203aaf4796af11a27bd : SUCCESS in 1h 19m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b39c5af737ba4c3588ce500a17a4ef27 : SUCCESS in 32m 58s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2d3ced7030e04bd4a9e6a0fdc16fbd27 : SUCCESS in 17m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fc7bffcbe08d4558bb3fbd71ac8ba37e : SUCCESS in 1h 00m 58s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d2509328621b45b5ac0c782fc44dc083 : SUCCESS in 1h 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03ad7d28e002462bab0e6f891c2965b5 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f1be72a11dc49e69dd8e34faa52bd25 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/767c5f318c1b46f28ca0e489fbb524e2 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f77cc45c6846daac07ce88e26ad81e : SUCCESS in 5m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1ed6b68551884938884a1a69b270fb47 : SUCCESS in 19m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/526133a9e3864bcdae34cad523687377 : SUCCESS in 30m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9231b25038244f069a942ffeada4aa37 : SUCCESS in 56m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/db11795d71fe42a685fb1ebd7d36f34c : SUCCESS in 54m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/64c7f29d53d34c8698624d1fa72687dc : SUCCESS in 1h 07m 00s","accounts_in_message":[],"_revision_number":8},{"id":"71ab2a2af4d670d7b96ce5c129925c1c56e227ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 22:42:50.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ba101fc6b73247bcb90a0735fe1afc96 : SUCCESS in 1h 43m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2c04d454898f4282939b7bc6c256a8be : SUCCESS in 41m 48s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/00e86b680c35451aa5b8577e44f528a4 : SUCCESS in 1h 26m 35s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b170fd35d9148fbb5cac87c1bcdbce2 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1781079f1d3d480184d6e27d27503be4 : SUCCESS in 4m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/97ca46b7fe36447d9d734c6fcbc9555e : SUCCESS in 32m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7bd48e015dce7a258021584b853a68cc314febf3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-08 16:27:20.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"e3a5c31850b8b1ae301802788c005e1f54d489a1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-08 16:27:44.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"068781ed055c28208013d0dcd3d41d35bbd854b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 16:28:06.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"6a6c9b5534a254bf19c154d6d296b7e4731e051d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 18:00:09.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3f91f68dda694f01b2c19817b3d4637f : SUCCESS in 46m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2e4af0142a644e4c8bc4826b457cf2ed : FAILURE in 34m 48s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/251c771738af40f390f632be5abb23b0 : SUCCESS in 1h 30m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a3619cb041784d5a93eeaef8083bfd41 : SUCCESS in 32m 23s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4bcaaf1b55634d62b40aef3e17f8d58b : SUCCESS in 55m 51s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c22d0ea613de454a9d8e12350addc214 : SUCCESS in 1h 14m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f7ceec913194450b1e1abef38baea44 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/704d2214cdd9490c9f5b15b5d62bb342 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da363fdf99ec4c7b90fed16afb070b10 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16e3d26821544ad7963cd13548785b59 : SUCCESS in 7m 37s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cbc64b07427c441a85e316993f09b0b5 : SUCCESS in 30m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b433b95c0c524c328ad3ffc5f59b1037 : SUCCESS in 56m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dabaa6d5d1384003a53d60f17509fcb0 : SUCCESS in 1h 06m 46s","accounts_in_message":[],"_revision_number":8},{"id":"ee16cbdd6edb32cb10de980312096a0652a86177","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-08 18:31:45.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fa60b1fd57cc16fae16c5ae99433bb23defa2574","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 19:42:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9bff008c04684519939a85b95b116500 : SUCCESS in 26m 51s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/71f53e32e50b4ccf9d556a1cb52211bb : SUCCESS in 34m 05s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/32d9e51fb991448dac6cfb239cd809dd : SUCCESS in 1h 04m 20s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b08b72c36cc94555bdb522b33d61461d : SUCCESS in 1h 07m 02s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a31ebce2444b402fa0eacb3e0b7830de : SUCCESS in 1h 07m 54s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a52c3f57898c42d78f3f8d3ddfe10d11 : SUCCESS in 30m 55s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e097f150d0324a80a245b00d052124d0 : SUCCESS in 14m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dfa91ffcdc3f46868d0a953f27191e96 : SUCCESS in 59m 24s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c4bd1eb6077847918ec50a8b0d46bce8 : SUCCESS in 1h 01m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7716b358db0d4772b133f8a483106325 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e68ba606843e4fa382dbb0ddccf3ba99 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22c01bbc837949b39caf3ac850b892a3 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89db19eedd5449049732562f2c46c0a0 : SUCCESS in 6m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a273b7b0b7864245bf5581e7dfab44c5 : SUCCESS in 21m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5f5b0a0e9231450297bfd76a7bddb43a : SUCCESS in 34m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/16d1eacae06949829f1a5c3f5a8efa97 : SUCCESS in 50m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0c3f3aad525843899b0220ce05f6720f : SUCCESS in 39m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7e5f77673040440ea4212ee93d8f77d7 : SUCCESS in 1h 04m 45s","accounts_in_message":[],"_revision_number":8},{"id":"447614f3fa3844501543e273b2d134b63fd704a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 19:43:07.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f51947edfae068c19b1386f850a6d0aaef210cc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 20:53:47.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9d5a4b43d48842afa09bffc9d7e7ec88 : SUCCESS in 1h 59m 38s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/84fc9fb7c2af4be1a7b516c6dfde1440 : SUCCESS in 1h 28m 45s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9e6047e4570c4de999f88f4056d4b3eb : SUCCESS in 2h 08m 58s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67f7b773a51d4e0095e0e348c18374ea : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1953e572025c4778a63c8191f9bf5adb : SUCCESS in 6m 27s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b2380b61dee48c38729217ec03d061c : SUCCESS in 1h 23m 34s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"85c8392f2797ca79033a25bc2f5a8679115235a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 21:07:06.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/223daa3519114c8c9fceca456c1036b9 : SUCCESS in 40m 33s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0c9b7426c7bd44d2999e92618745e185 : SUCCESS in 1h 05m 22s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ca6eb68bb541418ebc1f58953897023c : SUCCESS in 1h 20m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/63f0e57a31334b069a431518039b71fd : SUCCESS in 26m 15s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ae52e15c40794de5ae1abe3ceab93548 : SUCCESS in 54m 24s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/81d948fe801f40cfaf76fd963b11948f : SUCCESS in 1h 03m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc7cf80c84764902a01369c0f6f4b5b5 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b86e2bc81864aab95884bec8bb43f6c : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e64c05d812b748fc89011cc89590617f : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec1cbf2d87ef43f3bf445b74b1c15e44 : SUCCESS in 7m 10s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a495625f0a7145e2bf1ffe08d953a40e : SUCCESS in 28m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3301fb0e8e84482daccd2bd31cf9f3da : SUCCESS in 49m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2f8a4df5221d41739e8d4b3224270472 : SUCCESS in 58m 28s","accounts_in_message":[],"_revision_number":8},{"id":"7a18e658cf5e51ac593eda78ee78253a8134161d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 21:07:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"f4294b5afe8cd41cf168c41a7cdea5d1edf4796c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 21:08:55.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0169097114bf46049e99025159ec072c : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"023b9f325285e4a4c9ba8fe6dff54743a4e0cfec","revisions":{"222f03dde40777146453fc573dbc27af5d27ed61":{"kind":"REWORK","_number":1,"created":"2021-10-20 13:51:26.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/83/814783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/814783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/814783/1"}}},"commit":{"parents":[{"commit":"3c04addd59a5855985cfa4412c306b213dcf93b4","subject":"Add Python3 yoga unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c04addd59a5855985cfa4412c306b213dcf93b4"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 13:50:34.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 13:50:34.000000000","tz":120},"subject":"Debian: enable building Bifrost images","message":"Debian: enable building Bifrost images\n\nhttps://review.opendev.org/c/openstack/bifrost/+/806328 fixed building\nof Bifrost Debian images.\n\nChange-Id: I8e0c55e7995e5269f285c6d9c75e8c4b7b85a38b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/222f03dde40777146453fc573dbc27af5d27ed61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/222f03dde40777146453fc573dbc27af5d27ed61"}]},"branch":"refs/heads/master"},"3c62d8b8d5b06dadca153a311f704b3366b5db17":{"kind":"REWORK","_number":2,"created":"2021-10-27 10:04:37.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/83/814783/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/814783/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/814783/2"}}},"commit":{"parents":[{"commit":"1766ce8314bb263c7671eb61eeb34bc5f554d643","subject":"Merge \"cinder-volume/ubuntu: add lsscsi and nvme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1766ce8314bb263c7671eb61eeb34bc5f554d643"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 13:50:34.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-27 10:04:14.000000000","tz":120},"subject":"Debian: enable building Bifrost images","message":"Debian: enable building Bifrost images\n\nhttps://review.opendev.org/c/openstack/bifrost/+/806328 fixed building\nof Bifrost Debian images.\n\nChange-Id: I8e0c55e7995e5269f285c6d9c75e8c4b7b85a38b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c62d8b8d5b06dadca153a311f704b3366b5db17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c62d8b8d5b06dadca153a311f704b3366b5db17"}]},"branch":"refs/heads/master"},"481b2e1ff815ec36ef9d6811d7b5aa376245d082":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-10-27 21:12:36.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/83/814783/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/814783/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/814783/3"}}},"commit":{"parents":[{"commit":"4b3ad066d1a7cf4e93fe774a140b09dd5eedc6a8","subject":"CI: make Debian source jobs voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b3ad066d1a7cf4e93fe774a140b09dd5eedc6a8"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 13:50:34.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-27 21:12:36.000000000","tz":0},"subject":"Debian: enable building Bifrost images","message":"Debian: enable building Bifrost images\n\nhttps://review.opendev.org/c/openstack/bifrost/+/806328 fixed building\nof Bifrost Debian images.\n\nChange-Id: I8e0c55e7995e5269f285c6d9c75e8c4b7b85a38b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/481b2e1ff815ec36ef9d6811d7b5aa376245d082"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/481b2e1ff815ec36ef9d6811d7b5aa376245d082"}]},"branch":"refs/heads/master","description":"Rebase"},"6bebb93cd12be39400484fbf6ffe5278c78207cb":{"kind":"REWORK","_number":4,"created":"2021-11-02 09:24:30.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/83/814783/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/814783/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/814783/4"}}},"commit":{"parents":[{"commit":"5e40e0894f1a6e8d7a06ddff274d6420b51f0266","subject":"Merge \"neutron: Use update-alternatives --display instead of --query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e40e0894f1a6e8d7a06ddff274d6420b51f0266"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 13:50:34.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-11-02 09:24:22.000000000","tz":60},"subject":"Debian: enable building Bifrost images","message":"Debian: enable building Bifrost images\n\nhttps://review.opendev.org/c/openstack/bifrost/+/806328 fixed building\nof Bifrost Debian images.\n\nChange-Id: I8e0c55e7995e5269f285c6d9c75e8c4b7b85a38b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6bebb93cd12be39400484fbf6ffe5278c78207cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6bebb93cd12be39400484fbf6ffe5278c78207cb"}]},"branch":"refs/heads/master"},"0b2bc5f0e4c76177bc29f39451462bec1196f6c0":{"kind":"REWORK","_number":5,"created":"2021-11-02 10:51:07.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/83/814783/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/814783/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/814783/5"}}},"commit":{"parents":[{"commit":"5e40e0894f1a6e8d7a06ddff274d6420b51f0266","subject":"Merge \"neutron: Use update-alternatives --display instead of --query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e40e0894f1a6e8d7a06ddff274d6420b51f0266"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 13:50:34.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-11-02 10:50:25.000000000","tz":60},"subject":"Debian: enable building Bifrost images on x86-64","message":"Debian: enable building Bifrost images on x86-64\n\nhttps://review.opendev.org/c/openstack/bifrost/+/806328 fixed building\nof Bifrost Debian images.\n\nOn AArch64 Bifrost tries to install x86-64 grub.\n\nChange-Id: I8e0c55e7995e5269f285c6d9c75e8c4b7b85a38b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b2bc5f0e4c76177bc29f39451462bec1196f6c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b2bc5f0e4c76177bc29f39451462bec1196f6c0"}]},"branch":"refs/heads/master"},"b90881bc12ff5fd3369bdeb7e1b4ee40b6eaa305":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-12-01 16:01:52.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/83/814783/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/814783/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/814783/6"}}},"commit":{"parents":[{"commit":"13de905c8da05f00bf857d68b389470384957165","subject":"switch source images to follow master branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13de905c8da05f00bf857d68b389470384957165"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 13:50:34.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-12-01 16:01:52.000000000","tz":0},"subject":"Debian: enable building Bifrost images on x86-64","message":"Debian: enable building Bifrost images on x86-64\n\nhttps://review.opendev.org/c/openstack/bifrost/+/806328 fixed building\nof Bifrost Debian images.\n\nOn AArch64 Bifrost tries to install x86-64 grub.\n\nChange-Id: I8e0c55e7995e5269f285c6d9c75e8c4b7b85a38b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b90881bc12ff5fd3369bdeb7e1b4ee40b6eaa305"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b90881bc12ff5fd3369bdeb7e1b4ee40b6eaa305"}]},"branch":"refs/heads/master","description":"Rebase"},"6ebb6c70f82218441721f486d8d97f02ae37b884":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-12-02 09:07:46.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/83/814783/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/814783/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/814783/7"}}},"commit":{"parents":[{"commit":"7cc7b30dcd23170a35b0ff00aa6f8e7e1efd56c4","subject":"switch source images to follow master branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cc7b30dcd23170a35b0ff00aa6f8e7e1efd56c4"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 13:50:34.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-12-02 09:07:46.000000000","tz":0},"subject":"Debian: enable building Bifrost images on x86-64","message":"Debian: enable building Bifrost images on x86-64\n\nhttps://review.opendev.org/c/openstack/bifrost/+/806328 fixed building\nof Bifrost Debian images.\n\nOn AArch64 Bifrost tries to install x86-64 grub.\n\nChange-Id: I8e0c55e7995e5269f285c6d9c75e8c4b7b85a38b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ebb6c70f82218441721f486d8d97f02ae37b884"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ebb6c70f82218441721f486d8d97f02ae37b884"}]},"branch":"refs/heads/master","description":"Rebase"},"023b9f325285e4a4c9ba8fe6dff54743a4e0cfec":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-12-05 19:03:18.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/83/814783/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/814783/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/814783/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/814783/8"}}},"commit":{"parents":[{"commit":"8cf23a6c312c5318c05c6e6b97f1a3f7ca34f0e4","subject":"switch source images to follow master branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cf23a6c312c5318c05c6e6b97f1a3f7ca34f0e4"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 13:50:34.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-12-05 19:03:18.000000000","tz":0},"subject":"Debian: enable building Bifrost images on x86-64","message":"Debian: enable building Bifrost images on x86-64\n\nhttps://review.opendev.org/c/openstack/bifrost/+/806328 fixed building\nof Bifrost Debian images.\n\nOn AArch64 Bifrost tries to install x86-64 grub.\n\nChange-Id: I8e0c55e7995e5269f285c6d9c75e8c4b7b85a38b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/023b9f325285e4a4c9ba8fe6dff54743a4e0cfec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/023b9f325285e4a4c9ba8fe6dff54743a4e0cfec"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
