)]}'
{"id":"openstack%2Fkolla~814788","triplet_id":"openstack%2Fkolla~master~I8d7632e84dc26ed2022541996a529c2263133c6a","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2021-10-20 20:31:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8d7632e84dc26ed2022541996a529c2263133c6a","subject":"update information about unbuildable images","status":"MERGED","created":"2021-10-20 14:19:00.000000000","updated":"2021-10-21 08:52:02.000000000","submitted":"2021-10-20 20:31:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814788","meta_rev_id":"5be7d61715eb6840ed2c6d77087ad3fbdf893017","_number":814788,"virtual_id_number":814788,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-20 20:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2021-10-20 19:11:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2021-10-20 19:11:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-20 15:56:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-20 19:11:34.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-10-21 07:40:29.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"}],"messages":[{"id":"51ae22c5edc13956913fcbed392ca0b8238701d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-20 14:19:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb87e9824f425312161c337da663928f686ec9c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 15:56:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/dd58e5fdbcb441698885e2dd7909d035 : SUCCESS in 22m 29s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/da03941cb93246578f03b8b827e63438 : SUCCESS in 29m 48s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/126ac9e9f273458f95aa528f37b9bfab : SUCCESS in 49m 45s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/48f882d1c0174926935420b8cfb0ff65 : SUCCESS in 58m 57s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/dde649bbe08e4556a0c0db4fe3220810 : SUCCESS in 1h 31m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d57fa2c111dc4f5786bc0a7634d63cb0 : SUCCESS in 28m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/49ee768538cc4f51a19a58019642f75a : FAILURE in 17m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ed0e420fa6c147ca9e07af234d6249d5 : SUCCESS in 1h 00m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/196fe2a51f784d8ba4306990cc43e2ea : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64d18b664b7e4c0bb8ceee52b4a3bce4 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fe2c997be234eaba079b77b7b85d35d : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea7873b93fbf4a3b99f34846a1df24fb : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00f27e0717fa47bcbda7054d6977908e : SUCCESS in 6m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c024f7b59b83448fb3de84214b049c0d : SUCCESS in 20m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cc461e0d1166402799b534ef1c16ed1a : SUCCESS in 27m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c74a16ea94334e0fad4bc99e83e4f14b : SUCCESS in 53m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cfcfdfa46d30442e914813a9624a0b1e : SUCCESS in 47m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/37d13de934654e2a877ca153820ee8df : SUCCESS in 59m 19s","accounts_in_message":[],"_revision_number":1},{"id":"504a67d4048430fd9698658bd26eb02600340f87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 16:05:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7e1b95fc093f44c79ce6a82f15c1ed95 : SUCCESS in 1h 43m 52s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d5f63f352bc74d238a5f12e327d0e4e5 : SUCCESS in 45m 35s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f08ba96f0d764f68a4055df1d9982322 : SUCCESS in 1h 40m 14s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c81b2b95fe5e482285a3f063a5edc3dd : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/89ba8443a3cb46d0a885d0d044c3c64c : SUCCESS in 5m 14s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ce8a9d88b1a84496bfed4541d09f0816 : SUCCESS in 38m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a2cee3cecf3fd69d13bfc7d860b0f67716b460a4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-20 19:11:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"47064a7cc16bc719431bc6d46172f01e8c4dadbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 19:11:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ebcef1f2b6c3910df5c831a311f8380b314c16df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 20:31:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2f0c48188c0e49d8823544611ca93e1a : SUCCESS in 35m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/666fd56b5e174537b6f4f44f188ec2d6 : SUCCESS in 1h 07m 46s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7692656b95204b4facbdeb5058a1350a : SUCCESS in 1h 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fca22196569847ee913acc3dcee73270 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21e6985223964a7e9d6c477dc8fe04dc : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/364409c2202f485aad0568e4e1938327 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa841de47b504d298dda51c980c6b2ae : SUCCESS in 7m 26s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c0d25b7b14f6455e8925a7e954430254 : SUCCESS in 34m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/16ba916be322430591a84930783861e1 : SUCCESS in 47m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3f1d6f6fbb7a481896ab1116638bc22f : SUCCESS in 1h 06m 44s","accounts_in_message":[],"_revision_number":1},{"id":"f48d3e5b8f47809285f1905ae79e50ffa0df1e51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 20:31:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3a2a6c8f42d8439c40eb773eeca5f0d3b907e442","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 20:33:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a1df949b0bb43bfb12a6c5665479d11 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1},{"id":"b10a79ad58e753456db590ce00c85abfe7f1c54c","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-21 07:40:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d34e344af35360baabb296913fbe58885e869db","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-21 07:57:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5be7d61715eb6840ed2c6d77087ad3fbdf893017","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-21 08:52:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"65cf4146ed17aa6d00d917c34343abc18fe8fb15","revisions":{"65cf4146ed17aa6d00d917c34343abc18fe8fb15":{"kind":"REWORK","_number":1,"created":"2021-10-20 14:19:00.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/814788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/814788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/814788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/814788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/814788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/814788/1"}}},"commit":{"parents":[{"commit":"3c04addd59a5855985cfa4412c306b213dcf93b4","subject":"Add Python3 yoga unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c04addd59a5855985cfa4412c306b213dcf93b4"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 14:18:26.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 14:18:26.000000000","tz":120},"subject":"update information about unbuildable images","message":"update information about unbuildable images\n\nChange-Id: I8d7632e84dc26ed2022541996a529c2263133c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65cf4146ed17aa6d00d917c34343abc18fe8fb15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65cf4146ed17aa6d00d917c34343abc18fe8fb15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
