)]}'
{"id":"openstack%2Fkolla~814796","triplet_id":"openstack%2Fkolla~master~I8d8258877e8b0990ae6e1d41a2d56caf64807d72","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2021-10-21 14:00:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8d8258877e8b0990ae6e1d41a2d56caf64807d72","subject":"monasca: enable for Debian/aarch64","status":"MERGED","created":"2021-10-20 14:42:35.000000000","updated":"2021-10-21 14:06:06.000000000","submitted":"2021-10-21 14:00:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814796","meta_rev_id":"6f69f69f7b22150cf938147d34718b06d9d8f01f","_number":814796,"virtual_id_number":814796,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-21 14:00:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-10-21 12:32:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-10-21 12:04:28.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-10-21 13:26:37.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-10-21 12:32:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-20 16:12:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-20 16:13:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-20 19:12:16.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-10-21 06:19:25.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2021-10-21 12:04:28.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-21 12:32:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-10-21 13:26:37.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"fca0dc73eb9e4a12e1c5bd5b1f73618d2a83fa51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-20 14:42:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4bfb0d52057ef3dc006e0965e882c1f14fde013","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 16:12:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d386c12f0ddb466cbef696bdaa2f56a2 : SUCCESS in 50m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e8d7ae83cc794ccb9154d67128c5602d : SUCCESS in 48m 06s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1316733c14ae436cbef165eadb82bcfd : SUCCESS in 1h 22m 02s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c4c37e261f94004b3b220e9f4517127 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/699629c80ad346409b61df88f07a53bb : SUCCESS in 7m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0b641e8a812d404f80362ab70ad22370 : SUCCESS in 41m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"91a3a69a66c43adedb2b7fb65c51bff3bfb582ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 16:13:38.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/c16a9eccef15416d95fa9a958f766b6f : SUCCESS in 30m 27s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4f5e452a75d24cbfae19a310af951b8d : SUCCESS in 35m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a059bf7488e34317b92edeaa2f02a05e : SUCCESS in 1h 13m 43s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0234e73743f843a39db77b740702dfb0 : SUCCESS in 1h 11m 39s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1f315cbda1064c18b9586538d7b8fad8 : SUCCESS in 1h 16m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/694b044ddcf14ee5ad87f5701f0aa659 : SUCCESS in 33m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4d56ffcbae7f4092a53f6792a7ea032b : FAILURE in 19m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ba9bfbdf0aac4b479a828889124c9b35 : SUCCESS in 56m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cac8efd41d804ca6ba5cad89e2fc835a : FAILURE in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83a2c20d6b214c339f9d132a33a151d5 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89baf69cc5a54875871b2af185a5d6f1 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f618a25784fc4d5f8fd97e4e7186ea83 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a7cc71ec742487fa2dcaeb2063bfa80 : SUCCESS in 6m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/643ea7da85944e548ba8e92ab23c79df : SUCCESS in 21m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c8bba89e65b64052864158f760d4c267 : SUCCESS in 31m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cd80a1cf12fb4e6bb265943fd16264b2 : SUCCESS in 52m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/305a0e3220b84227a6f2d9645552f6cf : SUCCESS in 54m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e60353b32139402e8ca17f78969b922d : SUCCESS in 58m 35s","accounts_in_message":[],"_revision_number":1},{"id":"b2349c3345ec6d05c184fa0bfedf119edbb75c30","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-20 19:12:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74009df2cda9a924b44ec94e17f4b14fae884b9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-20 21:02:36.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"536bbb2b19333d2799731c6fab1aaf6c43f67317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 22:20:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/be6274b107c542d1bc0954328b846e3e : SUCCESS in 25m 46s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/96465e1d5b6e48488e650e475ad064b3 : SUCCESS in 37m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/37f76c4f6fca41ae888f1f106bbb6387 : SUCCESS in 1h 16m 23s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/41183ab8824a4a5e9cf2f2107bc0bf9a : SUCCESS in 1h 09m 11s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d6073fa18e6049a78802ebcf0fbcfe9c : SUCCESS in 1h 01m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/499e4cd247d5470bb0f6219b3ad03cf2 : SUCCESS in 20m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bb4e0e7a56104dd49779d03d1c9fdf53 : SUCCESS in 21m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/892938969f6d434285f26259c61dc979 : SUCCESS in 1h 02m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1f820dbecb2445982f653a08d1c802e : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8ac0c0239954c399d39bc1de9ab955e : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e641481dabe1496292f30061d118058e : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc4ec98616344f62a7895abe56a68343 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9388b135e8446109ff6fe1561f5c8e9 : SUCCESS in 7m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/087947f3251242a78c311da3907361ba : SUCCESS in 18m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7bdbbeb6df994574baae0b25e86a4fed : SUCCESS in 29m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/325274884e7044179919e39ec664cf52 : SUCCESS in 58m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/06324cfbcf9441b19efd3f51a9486f5d : SUCCESS in 54m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/36eff42411a7496da4fa92e5d09505b3 : SUCCESS in 1h 05m 37s","accounts_in_message":[],"_revision_number":2},{"id":"e456ca70e568819ee29d05e7ecb4112090b925ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 22:36:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2aa61be2a1174f20bf957d74f5b22808 : SUCCESS in 57m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ca5afbfea0454ba6b1daa421e5e4e2df : SUCCESS in 49m 05s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5d7f72ec49c5416fb94c26d6ddc48f1b : SUCCESS in 1h 26m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/944db919f3a440e189d0fcbf100f2c73 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1dbde71168c84469b3ea6917147bc03e : SUCCESS in 4m 46s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5bb1af95220145bca24181b82332aff9 : SUCCESS in 41m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5c29726999a97d561c3d05572d0b0e53adc96eda","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-21 06:19:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fc20391fdb7aba48d22092f35091b29238042f93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-21 07:55:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8c0e15be33ee996e5a868d421c9a8d93404fb0e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 09:16:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4af0b9a33b154d17ac9950b98c44dfb9 : SUCCESS in 27m 17s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e3075ee0aa514d2b86c8c1adaaab8d48 : SUCCESS in 37m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0e908c1136a94f388e8a22e0da234d04 : SUCCESS in 58m 56s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/91d70fe46ee44d65bc492bae686ffb36 : SUCCESS in 57m 58s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/485bfbd36fb243ae8ee8ed05516c6ba7 : SUCCESS in 1h 19m 05s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/052b964ac26a415fbd6aa645f8f26c9d : SUCCESS in 27m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9200089c3fcc4bc79228eb4553c0e613 : SUCCESS in 17m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f782ab8ca34c415bb3b522423956b60f : SUCCESS in 48m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2d15431cb194daaa11f7bf1d9674955 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f5887cf074b4b33a6d1f6bbf1a39ef7 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6578e1a6752b4718a887d6a09bdf23e7 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de58e54d4db74bbb9e3655a2b7a00cb3 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88cfb1cf51c84062a41807e32763c915 : SUCCESS in 6m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/abd9d19dd14d42cba45d57b45d07a66d : SUCCESS in 22m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ab4bfe2b23624c6197a9cef2f67df55c : SUCCESS in 35m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/067e19edbf0e45cfb174e83960abe9b0 : SUCCESS in 55m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/746523fd644c4a5388242d6667a5aee7 : SUCCESS in 49m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d9b86ccc1e0d4ca3996b709cd96cae07 : SUCCESS in 1h 08m 10s","accounts_in_message":[],"_revision_number":3},{"id":"aed38203e8d2d89258ce768bf3aee834a62f0842","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 09:29:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1cab684ea08f4654b8a813dfb2ffbc55 : SUCCESS in 1h 24m 04s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7f47c5f0eeee4246b216f5440d7a35cc : SUCCESS in 39m 51s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6bd9ae9da0b5451caaa2e9ac53f3f7c9 : SUCCESS in 1h 25m 08s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1fa370f9e4194f84946bd10b3e48ed21 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d4ae81cbfd8e4a7993d202f7ff634ff8 : SUCCESS in 5m 42s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/170fd2b2d5f544ff9e4b8e19518c0ec5 : SUCCESS in 46m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"548917c90d46ed746f3bbb43ba53b7a49ff3f006","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-21 12:04:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4090621d95791745c9376add5e5aa098569553e8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-21 12:32:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"84302363908f13f3eaa01460dd1a9cf65670e70d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 12:32:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a81d6b7677cc5e4e24789b31260de15e07d41892","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-21 13:26:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"af94fa53f21356b61f5bc99fb3e1054f99b4a4c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 14:00:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e2b8bab30895485ca6372bc3aca48ecf : SUCCESS in 40m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/83af265a4f8a48a19fa021ca7e6d54ed : SUCCESS in 1h 14m 22s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f495856d93c7420a96c667c6ff99b0f0 : SUCCESS in 1h 27m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fb6bea1d820402480e95c6b971cb884 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fee65297e444faebc99c7d8a4a7e075 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cff8551282445c9a2e6189366705be4 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84360e28f5794b5fbfbf032208e7a19f : SUCCESS in 7m 38s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/be70d8e114234c4b8fb7cb772766b5dd : SUCCESS in 33m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7cb6cd07de1649f4a990bbbb94a6bfec : SUCCESS in 1h 00m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9aded7ab3c494937bebfdc3b7f6972ce : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":3},{"id":"95ab5f5129b8fe656fbd87ae5ec81c737dbcb3dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 14:00:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6f69f69f7b22150cf938147d34718b06d9d8f01f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 14:06:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7ef27e4cca5461086f1bf71a3caddbd : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5dfc361b6567db81afe0ad87b1e510750ea6b51d","revisions":{"955eaf4023e9bcb18c05cdd570ec1852826508a4":{"kind":"REWORK","_number":1,"created":"2021-10-20 14:42:35.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/96/814796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/814796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/814796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/814796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/814796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/814796/1"}}},"commit":{"parents":[{"commit":"65cf4146ed17aa6d00d917c34343abc18fe8fb15","subject":"update information about unbuildable images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65cf4146ed17aa6d00d917c34343abc18fe8fb15"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 14:42:05.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 14:42:08.000000000","tz":120},"subject":"monasca: enable for Debian/aarch64","message":"monasca: enable for Debian/aarch64\n\nCentOS and Ubuntu has too old librdkafka-dev\n\nChange-Id: I8d8258877e8b0990ae6e1d41a2d56caf64807d72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/955eaf4023e9bcb18c05cdd570ec1852826508a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/955eaf4023e9bcb18c05cdd570ec1852826508a4"}]},"branch":"refs/heads/master"},"9954ca63ee687c71e2b709c75a8b20cc912bce8c":{"kind":"REWORK","_number":2,"created":"2021-10-20 21:02:36.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/96/814796/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/814796/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/814796/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/814796/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/814796/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/814796/2"}}},"commit":{"parents":[{"commit":"65cf4146ed17aa6d00d917c34343abc18fe8fb15","subject":"update information about unbuildable images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65cf4146ed17aa6d00d917c34343abc18fe8fb15"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 14:42:05.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 21:02:36.000000000","tz":0},"subject":"monasca: enable for Debian/aarch64","message":"monasca: enable for Debian/aarch64\n\nCentOS and Ubuntu has too old librdkafka-dev\n\nChange-Id: I8d8258877e8b0990ae6e1d41a2d56caf64807d72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9954ca63ee687c71e2b709c75a8b20cc912bce8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9954ca63ee687c71e2b709c75a8b20cc912bce8c"}]},"branch":"refs/heads/master"},"5dfc361b6567db81afe0ad87b1e510750ea6b51d":{"kind":"REWORK","_number":3,"created":"2021-10-21 07:55:53.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/96/814796/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/814796/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/814796/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/814796/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/814796/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/814796/3"}}},"commit":{"parents":[{"commit":"6014b503ddab254e2a8a9db397c46b91fe189ed1","subject":"Merge \"openstack-base: drop anyjson\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6014b503ddab254e2a8a9db397c46b91fe189ed1"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 14:42:05.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-21 07:55:46.000000000","tz":120},"subject":"monasca: enable for Debian/aarch64","message":"monasca: enable for Debian/aarch64\n\nCentOS and Ubuntu has too old librdkafka-dev\n\nChange-Id: I8d8258877e8b0990ae6e1d41a2d56caf64807d72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5dfc361b6567db81afe0ad87b1e510750ea6b51d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5dfc361b6567db81afe0ad87b1e510750ea6b51d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
