)]}'
{"id":"openstack%2Fkolla~814857","triplet_id":"openstack%2Fkolla~stable%2Fxena~I547aa9dac5f003135dab8c497aea8dbac89c38e6","project":"openstack/kolla","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2021-10-25 11:06:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I547aa9dac5f003135dab8c497aea8dbac89c38e6","subject":"centos: switch to RDO final release of Xena","status":"MERGED","created":"2021-10-21 05:58:09.000000000","updated":"2021-10-25 11:08:55.000000000","submitted":"2021-10-25 11:06:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"814857","meta_rev_id":"d0e91b255d063a92f101a6df413d50dd48cfb9ae","_number":814857,"virtual_id_number":814857,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-25 11:06:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-10-25 08:27:23.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-10-22 11:54:07.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-10-25 08:27:23.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-21 08:02:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-21 18:52:41.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-10-22 11:54:07.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-10-25 08:27:23.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"539a14194d3123fd3678a601eb357933368bb8ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-21 05:58:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bce972f7d94afe6512b8b3c8bc1410277db59cb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-21 06:37:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4075d0ab0bd12c2931816315f6d406249e61dd27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-21 06:43:16.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"38e8c0cd16700bd4aaf15bf59603d585bf30afef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 08:02:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7cca3de5881a41809287b32f46dbf00d : FAILURE in 9m 45s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c56657a6b67b41abb450a25e785cc0de : FAILURE in 9m 55s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/919c35e0b7c34c6b902222da56cd64ad : FAILURE in 16m 20s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a865ed3d237e46f59e66a7a6d9166893 : FAILURE in 16m 58s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/02b42ca9456c42579b4bf724ec02f29b : FAILURE in 17m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8ec1bbd8c087425bab0d952d67ce0d16 : SUCCESS in 25m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bade1822f33543b693578581be092723 : SUCCESS in 18m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/56993506b3074207a99de1d76dcf9042 : SUCCESS in 1h 00m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da697fb708df430da107b82def329284 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acf8391a079b4643b37f20126d864494 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb41dafad05d45dd9df518f7f45907ee : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77da6dba542b45d992c63a98c311e884 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06621917a90346c9a76a181460d63a25 : SUCCESS in 8m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/68b1fa1df3bd4d5697661b0374a3a656 : SUCCESS in 31m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf0cf24889c84a7c922b6a633eae3a8a : SUCCESS in 39m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/391f279e60864a7ab374eb889e7110df : SUCCESS in 56m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fcfb3da29f0e4c2ab8e31b8602600583 : SUCCESS in 55m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d1bcb0a6b2ee4a4885e82509041b8ccc : SUCCESS in 1h 11m 06s","accounts_in_message":[],"_revision_number":2},{"id":"ba435729bb0d7caac2267735e50e25a9cc41d4bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 08:38:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/fe479b1a70c842a588257c0d9c545934 : FAILURE in 16m 04s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e1e0c2f71e664fd9a2a0f5624e54e758 : SUCCESS in 37m 05s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9d47762f08894f329504345a01e9c8e5 : SUCCESS in 1h 52m 15s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/17a5615cb3704ddcb232f6d8a06d7888 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b6185ed50047493582f25be06c19a608 : SUCCESS in 6m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/bd36ed18ddf44a22ad58b356826dade6 : SUCCESS in 1h 06m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"17f296887d512bc0f9dd918497df44d47e751cb3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-21 18:52:41.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"170b69a887df97d503e6d6596f29cc409d456ac1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 20:11:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/0e5a6a82a47b4bd3b3e4aff43baf1595 : FAILURE in 29m 05s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f6bb30435f284ac3b108ed8d75c593bd : FAILURE in 32m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6524ae104b4f44828fb037b7b80ac282 : FAILURE in 22m 49s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/948ce4f8e5f543c49f23d1de4281034a : FAILURE in 26m 28s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/54411ed9ebe94f35bffc219576cf452c : FAILURE in 23m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2b34d172bd0d4601b3b0ad997f17b286 : SUCCESS in 31m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3d4f7abdc66144c4b3d61045efc7e1d4 : SUCCESS in 18m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/29e124fd809a4d6eb898cba9c29b246a : SUCCESS in 48m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e22e43adbc8841a59f7043d27d240457 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19b6716523374f6c84b8a0bc8ed0344c : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/823013e2f148487692812afa6cf56c74 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe695318e8fb4dc69b428d48cd959ac7 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/319076b4f55e46f1ad174d2c91f991e8 : SUCCESS in 6m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1448a0626d44412ea76b3c15994e674f : SUCCESS in 23m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9951d9f80973446fbb778430f3ff5f44 : SUCCESS in 33m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9c1ededa127f4872bdf2526953baf1e3 : SUCCESS in 47m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a4a3c0ea4f4742a38270d7a528d8d3a6 : SUCCESS in 53m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4bab7e6e979d4e52bf5fd48880a11a53 : SUCCESS in 1h 08m 57s","accounts_in_message":[],"_revision_number":2},{"id":"6c295cf20a20c0fea9b7c1a77dd4ed034554905e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 20:30:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/10be58714f3d4c37b557e465e267bc4d : FAILURE in 53m 30s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/715b177b2c4b4310bf49adb1210b599f : SUCCESS in 49m 36s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dcf576aa6c7d4b20be19b55210a7d891 : SUCCESS in 1h 37m 28s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad51266287e4446f970f9b40d4de51f9 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/082450d8926d42a4a2004e8b61853f05 : SUCCESS in 4m 33s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2eb979f369424a019e0ea6d0cc57b132 : SUCCESS in 43m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7ac8a4af0b545b8eb82ae5a257dce456536e9d91","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-22 07:49:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9040294c4dc83fe6dca97742ec9e7a9001350a68","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-22 09:33:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0a9b19d233884f127081344e0315d1b591244fad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-22 10:08:26.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"756ae6ad950518233ef48cb105d63415fab33128","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 11:37:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/034393deb71c4f72ae4761964c1e3754 : SUCCESS in 30m 10s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8083a2e2837f4754bc8d51d7630bf801 : SUCCESS in 39m 54s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/efa9d92326f446ae808f6eb6291c8494 : SUCCESS in 1h 15m 23s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c1d983d84cc44fa39dd5d11c4b572138 : SUCCESS in 1h 14m 05s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c1063c8037524e1baffa80d11f2e0b01 : SUCCESS in 1h 27m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/38e190b7bc424792986bda2d83fc2c19 : SUCCESS in 29m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/da81cce4088144bfbee3ee5cca903d91 : SUCCESS in 21m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ebcd1c51b7154defb5f8eb779683cdee : SUCCESS in 59m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e59d354fbd4743a18c6d1fc8badc1dba : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b343760e4ad74b0c9a55c0917ce25f09 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0052edd687f14349bf029ad86e905a5e : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/920706e330db430292acd18c76b52bc3 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fd0acbfa436450b913b70db21fb8f56 : SUCCESS in 6m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b4eebfc657534fa398a5bc3e97f83ee5 : SUCCESS in 20m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4b4a0c5bbcb44cd1b3e03f51f186193f : SUCCESS in 26m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3a69f3910cef427fbc83c1f28a102cd4 : SUCCESS in 56m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/232f98f05e3c46abae66251c68d2a0a3 : SUCCESS in 53m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d0b66a3c91e74246a07a06067863248c : SUCCESS in 1h 12m 08s","accounts_in_message":[],"_revision_number":3},{"id":"096a4721b32af539ee0637834b8ade66f9d679bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 11:44:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/712c584ea3824bd78d67162a6e590e34 : SUCCESS in 55m 38s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0695c5fc8eab49b7a47a9dd1e05d3d13 : SUCCESS in 56m 41s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6a9aac9623dd48bb8ac8de41db77ded0 : SUCCESS in 1h 34m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6fe7033b36da491d94f4b5cc1dfcea54 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/97b79f55c13240d0aa3bb7aae51fefa9 : SUCCESS in 6m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f8ca68a904654f3f98363c22d983a10c : SUCCESS in 52m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f40ffd09eb250733ea3a7a7447cdd172f70e4e14","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-22 11:54:07.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"90ffaf0814a43ce7f3068d0df4db81cd574e5e87","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-25 08:27:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dc0a2df7b2cee9fddaa84b456e8b9b493a8aa4aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 08:27:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1e836dda029526dfe4a68522ae1db9ff25fc4e28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 11:06:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4037a62c06f74b779a4310e8cdca7c60 : SUCCESS in 35m 02s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6409ddc7c329439098920e6bf32f3af6 : SUCCESS in 1h 07m 36s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/95426963e51b4d7cb5f0778a2c7597c2 : SUCCESS in 1h 30m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d6b26c7db034c64a92eca1565902c80 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c335ae41dbf427896586d5f3d65aa3e : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea80a534e22a4602863aed101874f035 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67f5091a556a43b5b9079ef08f99ab81 : SUCCESS in 6m 44s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6c5f34deddc41d791f94c72cd9bd8cd : SUCCESS in 27m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c913be461eb64dd28fbd8e83f0b0b957 : SUCCESS in 50m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bbebe65df92a47cbb77f1986cf071e89 : SUCCESS in 56m 06s","accounts_in_message":[],"_revision_number":3},{"id":"7a35a63ea82b9ac37fb065c1e56d02f6d88e67b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 11:06:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d0e91b255d063a92f101a6df413d50dd48cfb9ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 11:08:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90347df25af04102ae61f8eb85e40459 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"46f8f3f63d0fb7e6a8153b515a360587eeb8c6e9","revisions":{"e36af4692d8dcc9763161ca5fa956b25ff7a1f09":{"kind":"REWORK","_number":1,"created":"2021-10-21 05:58:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/57/814857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/814857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/814857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/814857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/814857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/814857/1"}}},"commit":{"parents":[{"commit":"9eed029a83acd18d8403312ee3b9dbba0bd8999e","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9eed029a83acd18d8403312ee3b9dbba0bd8999e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-21 05:57:54.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-21 05:58:05.000000000","tz":120},"subject":"centos: switch to RDO final release of Xena","message":"centos: switch to RDO final release of Xena\n\nChange-Id: I547aa9dac5f003135dab8c497aea8dbac89c38e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e36af4692d8dcc9763161ca5fa956b25ff7a1f09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e36af4692d8dcc9763161ca5fa956b25ff7a1f09"}]},"branch":"refs/heads/stable/xena"},"73975ce7f6bc5c8fc6102842cdfea96cfb973578":{"kind":"REWORK","_number":2,"created":"2021-10-21 06:43:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/57/814857/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/814857/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/814857/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/814857/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/814857/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/814857/2"}}},"commit":{"parents":[{"commit":"9eed029a83acd18d8403312ee3b9dbba0bd8999e","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9eed029a83acd18d8403312ee3b9dbba0bd8999e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-21 05:57:54.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-21 06:43:16.000000000","tz":0},"subject":"centos: switch to RDO final release of Xena","message":"centos: switch to RDO final release of Xena\n\nChange-Id: I547aa9dac5f003135dab8c497aea8dbac89c38e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73975ce7f6bc5c8fc6102842cdfea96cfb973578"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73975ce7f6bc5c8fc6102842cdfea96cfb973578"}]},"branch":"refs/heads/stable/xena"},"46f8f3f63d0fb7e6a8153b515a360587eeb8c6e9":{"kind":"REWORK","_number":3,"created":"2021-10-22 10:08:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/57/814857/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/814857/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/814857/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/814857/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/814857/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/814857/3"}}},"commit":{"parents":[{"commit":"9eed029a83acd18d8403312ee3b9dbba0bd8999e","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9eed029a83acd18d8403312ee3b9dbba0bd8999e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-21 05:57:54.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-22 10:08:26.000000000","tz":0},"subject":"centos: switch to RDO final release of Xena","message":"centos: switch to RDO final release of Xena\n\nChange-Id: I547aa9dac5f003135dab8c497aea8dbac89c38e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46f8f3f63d0fb7e6a8153b515a360587eeb8c6e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46f8f3f63d0fb7e6a8153b515a360587eeb8c6e9"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
