)]}'
{"id":"openstack%2Fkolla~814860","triplet_id":"openstack%2Fkolla~master~I86e6f7d377e3068a830d178f9b3a077c9a710dd1","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2021-11-23 11:37:29.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I86e6f7d377e3068a830d178f9b3a077c9a710dd1","subject":"[DNM] Remove debian ovn-base from unbuildable images","status":"ABANDONED","created":"2021-10-21 07:45:57.000000000","updated":"2021-11-23 11:37:29.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a0c5c682736ce624ae72ef16ca0aba6cc650fdbc","_number":814860,"virtual_id_number":814860,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-21 09:18:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-21 09:18:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-21 18:58:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"}],"messages":[{"id":"dc11155560d5d136e38c634c47f8cad544322dc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-21 07:45:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0323c2ad92f71b5e53f60ca8ed7b24ab753087f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-21 07:48:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18c7795d867a790aa375b534a1a77771db205730","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 09:18:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a078f25a42d94f29af2fe0a02324fa36 : SUCCESS in 29m 09s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/95a0f717a6764679b8a188a64cc1477b : SUCCESS in 44m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1b14d945400e4d65b9ea2acce2bf18c6 : SUCCESS in 1h 13m 41s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/bb25a9474d4d49b29e97f0dad5718c15 : SUCCESS in 1h 11m 38s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1d41a200d4f14e4da927120804038d7d : SUCCESS in 1h 29m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6ca7d9a0044e45439dc4e1db270dcbd7 : SUCCESS in 29m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/17cc31a23a01437cab90b456d56a1eda : SUCCESS in 18m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5a210f6526134bd0907c19974bec04ef : SUCCESS in 52m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a06ecc82cf428a8088a8a8ea9cbae1 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d6723cfde2446f7943bfbc2ef54df0e : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec18bf714458454383dfca412a273bb9 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc1758b831dd4cf6a5dd7d7d0c51fe40 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6cb5d4a732e4b94b4b5ea47961fe835 : SUCCESS in 6m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c203eaf4d5a24516b1624664a36c123d : SUCCESS in 22m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3d8f60055dde486dbed37f081f005475 : SUCCESS in 35m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dd19925737a74464bfc96cc044f9be6e : SUCCESS in 54m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/37501ed25e614d61b8223bd7bc47e239 : SUCCESS in 55m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ffc6d2cc3ac644368e4d239f680c0e28 : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5e262045e9bcda8ecca6b504b8bb640af81301cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 09:22:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c6b201890de146bfb200bfb0740ac8bb : SUCCESS in 1h 28m 49s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a30696bd6ae843b8bf140c39981581e7 : FAILURE in 52m 52s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ef270b218eef4d74a500cca53e1564b6 : SUCCESS in 1h 28m 34s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/906b85670f784b3db8ee793f9f0e6d73 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1efcc081997c4111aef1bd821dbddc95 : SUCCESS in 5m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2d720ccc84b14936806d06597cefb694 : SUCCESS in 47m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae8661aed905abbcb77d57f5e1b748913869ed8e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-21 18:58:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f4a488be34c9e64d77da45ab077f5e748fe93705","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-21 18:58:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a0c5c682736ce624ae72ef16ca0aba6cc650fdbc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-23 11:37:29.000000000","message":"Abandoned\n\nOVN images are supported for both x86-64 and aarch64 architectures now.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92bcf82ccfdcd14a2a801ba2db0f7512f0a03059","revisions":{"92bcf82ccfdcd14a2a801ba2db0f7512f0a03059":{"kind":"REWORK","_number":1,"created":"2021-10-21 07:45:57.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/60/814860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/814860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/814860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/814860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/814860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/814860/1"}}},"commit":{"parents":[{"commit":"65cf4146ed17aa6d00d917c34343abc18fe8fb15","subject":"update information about unbuildable images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65cf4146ed17aa6d00d917c34343abc18fe8fb15"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-21 07:43:17.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-21 07:45:55.000000000","tz":120},"subject":"[DNM] Remove debian ovn-base from unbuildable images","message":"[DNM] Remove debian ovn-base from unbuildable images\n\nChange-Id: I86e6f7d377e3068a830d178f9b3a077c9a710dd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92bcf82ccfdcd14a2a801ba2db0f7512f0a03059"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92bcf82ccfdcd14a2a801ba2db0f7512f0a03059"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
