)]}'
{"id":"openstack%2Fkolla~814900","triplet_id":"openstack%2Fkolla~master~I645f5d04ff2be3a5857765e901eccfe0b56ff199","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2021-10-22 17:42:27.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-10-22 17:42:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I645f5d04ff2be3a5857765e901eccfe0b56ff199","subject":"ovn: Add Debian x86 images","status":"MERGED","created":"2021-10-21 11:34:28.000000000","updated":"2021-10-22 17:44:19.000000000","submitted":"2021-10-22 17:42:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"814900","meta_rev_id":"fcecd247ae46902afa68abdde93d89a2dddcf670","_number":814900,"virtual_id_number":814900,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-22 17:42:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-10-22 16:19:15.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-10-22 13:17:14.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-10-22 16:19:15.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-10-22 16:19:24.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-21 12:51:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-21 13:58:23.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-10-22 16:19:15.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"0d26023aa9cf5a6e2d0b95918546e09698c08f80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-21 11:34:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f482e641d4c8587613e1b36487873b0765b57c75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 12:51:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e000f85f427740afaabd8b2b32e99a2f : SUCCESS in 35m 31s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4dd5cfb97b004182af18bd7b793e515f : SUCCESS in 37m 16s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d667d779c7cd4dacb84a75fc3113743d : SUCCESS in 52m 46s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2d5c40a398a445669dd855461ee359df : SUCCESS in 1h 10m 24s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c6bb0a32a56d4de5851a4dd809a9f9ba : SUCCESS in 1h 10m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/404a33f863ea4b64b57f0a64eb7a0013 : SUCCESS in 35m 26s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/eab0fb0d1bb44c0494521f6157763789 : SUCCESS in 16m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/923dc810398c42e1afbc3c11ac32cf64 : SUCCESS in 55m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/196d81f753fa4eaf8176e9b833239d01 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a28f574319ee4727bb320a7943c3dd4d : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7230c2e2b749464eabeaf6cbbfbcf1ef : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fe15aa511dd4d7c80eae08ace4bb2af : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccf93c3329524930b78b4ccf4a7fff99 : SUCCESS in 5m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/02e6e7a6c9a84581a9d9d276a13e40b4 : SUCCESS in 22m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/76c7f47bee804e098f8ef40184fccd96 : SUCCESS in 30m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0b37ce2ecb234bd49434f9e79a9dda61 : SUCCESS in 48m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fb8707f72c704be1af72b2af3124a3cb : SUCCESS in 52m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/21c14ae8898249bfb3476b11d83b87bd : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":1},{"id":"cffc2701d7f15d589646476aa9db9de666c995b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 13:14:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/8e6d4efe2b8a4987bd3ab72eb0c4954f : SUCCESS in 47m 47s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/58472901be5c421197dba5763f5f1976 : FAILURE in 46m 55s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8b9d495314704eb2a983ecb05bf8b7f3 : SUCCESS in 1h 38m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b99086379bd427b88058e02aa22d4bd : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/adb0b27bdee24ba0b81cb717e0ef1cf9 : SUCCESS in 6m 01s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/68e354b378cd4ac3a79c7b390309a0c8 : SUCCESS in 42m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2babd040acb29e42aa16c707740dd5d8ae98c41d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-21 13:58:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"918e9952c5e815acc02b0dadee256af4cdd0dc44","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-21 15:31:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32d3f5d64e13ff12a8f9d5cd71137be2c4c37b91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-22 07:54:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5fa26fd575a40474d4d72f35933008a496d7c8ad","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-22 09:35:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1614d1902986915536ee0dfbb627645d6de4c333","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 11:04:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e9e6c62493574bb38e6f8aa7fc5d1f8d : SUCCESS in 26m 45s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/bc16e7f2103144d29993750bd3628953 : SUCCESS in 46m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/993db8e2516f494582bf51c979c62953 : SUCCESS in 1h 15m 17s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/32073cd971874a00ae4e7bde7e140a27 : SUCCESS in 1h 04m 26s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/76ded99e29af4c1ea8b2257aba2ea487 : SUCCESS in 1h 28m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cd332de9bdad4d33ba6f373352e32aef : SUCCESS in 26m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3c3bc434ad1d4bbea6e1d7d24084220a : SUCCESS in 17m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/82f4ef3c3fd5469e9b85ac490e1a8e24 : SUCCESS in 1h 00m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9f17f1fb71b4532af6945ae304c26f8 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da630457c2384b70b453c4ac82b485de : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfbc87ba8a96463bab15cdffc08ab23c : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/526447c988c04ac5b90d55e0595a7b96 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adefe5e7e12541e6a8510dec89ef43b4 : SUCCESS in 6m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/98a02459a56443f28589ce523289d248 : SUCCESS in 19m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef8e79257195444ba5b1a0193314f676 : SUCCESS in 34m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6b77d89aee0f4bbab2255380762c9064 : SUCCESS in 55m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ae4c6ce10f6f4d449d4bc3a02bdbec69 : SUCCESS in 53m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fab028edf1ea4b0396bb12988770ff2a : SUCCESS in 1h 07m 53s","accounts_in_message":[],"_revision_number":2},{"id":"5b7119df840f4f7896f012de2a94926eb8b1f7fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 11:32:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d7e9761099a2478ba50d1efb6c6cc362 : SUCCESS in 1h 26m 37s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0a3945ed26ad44bba3ed5dd70fe36bb8 : SUCCESS in 1h 16m 50s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f2917d3c051c4246aa4deda4e7613aa1 : SUCCESS in 1h 50m 04s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca1efb7445094497bc662cb482678567 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/086a4ab177274d9ab0b792359b2cd736 : SUCCESS in 6m 42s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d4b97d0023d249f8998a2636065e64a9 : SUCCESS in 1h 14m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5c7d180dd8c623807e5c1c82940447e9a9c7e1a1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-22 11:36:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4e97983b5ba13cfade8761379164d278ce90d5a0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-22 12:37:39.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6f0fca3928284fc22ada20c78a166b08a03458a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-22 13:07:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2055007ac5ff80c484e6a9e730ad66af119f47b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-22 13:16:54.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"2970aca2a1068600a51b9ca441ee0d640f19ac2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-22 13:17:07.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"6b6b6917511be136f5173d53b5988052a12d2cb3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-22 13:17:14.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"7c8c40d120da1c22f0654acdbf53b161578ec302","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 14:48:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/932ac4d89f6843ebb61197a8a18d10ec : SUCCESS in 31m 27s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c265934cbbdc4aab8fef09e3c7c9748a : SUCCESS in 38m 09s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/467344f82b664c308067dbf498bf8141 : SUCCESS in 1h 12m 19s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8391d1179ba0405db28816be06694318 : SUCCESS in 1h 15m 14s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/64a451ef5a334edea270b100a4e90499 : SUCCESS in 1h 28m 05s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/37275f18358f42d6a7fb13a7d3287e46 : SUCCESS in 30m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3fe666aced23438aa42156635d0a0f8e : SUCCESS in 18m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c25e2675703e4cd5a87213a498656eae : SUCCESS in 1h 01m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfc5d05e1c754cae81e66bb465c44a9e : SUCCESS in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0941fb1d1364240a9620feee751f159 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8887c5342ba8410b9ef8c319c8775fb8 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18e77f8b49064bd68c3160abd58bf54b : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d18dacf77ff49428664ed623d8fa6bf : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07c452f377474d5592cbd7deb06a35b9 : SUCCESS in 4m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2639d40d8b184870a1b26d285af516f4 : SUCCESS in 21m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fc8bad1373e4465dba500699ca502061 : SUCCESS in 38m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5aa8c808edb440428a351fc7a3e5089a : SUCCESS in 51m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bcb3ef9d9f2042caa17fd1ea393d573f : SUCCESS in 50m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d09697eedeff4df2a7f4bcdf966f9844 : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":5},{"id":"275554c765c67abad7da79e29593b540904adebc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 15:06:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/fe596ed2a18c4a859e755247868f1982 : SUCCESS in 1h 45m 32s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e3b56773463a4a9390b5d8851ee71732 : SUCCESS in 1h 26m 28s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3139a7ea25c4424cb3e87ee9dc7ee3ba : SUCCESS in 1h 47m 29s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f6e9515de11498d8db41f454121d084 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e943342fcb04cadbad0c82db2e5240a : SUCCESS in 8m 41s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b4fbc7814e494a6a926d8852fa8cd289 : SUCCESS in 1h 41m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f60be44f3bdb349316313054285a6e32f32e5d2c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-22 16:19:15.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d9d416203cb5674de95e363b447e39eefdcaa010","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-22 16:19:24.000000000","message":"Patch Set 5: Backport-Candidate+1","accounts_in_message":[],"_revision_number":5},{"id":"2103fe7cc8a8c4a8b9991cc88f2b01f399c9491c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 16:19:41.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8039c01bea220defaa4f4abb82ae28feb57c143e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 17:42:26.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/85b4caa1a37e4530a94f9ce34840f729 : SUCCESS in 48m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a4dc89307ec84ce5a9027c4689e9b9ba : SUCCESS in 1h 14m 31s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6b286f30c4c34eb79dcb291707d8d6a4 : SUCCESS in 1h 19m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37512464e0e840d29e312a8bbc14db11 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0702d724ebf945af8140e95b10e0298a : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccb74bf4e252464abf879c607440b24b : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c7a7a512320444c9ede57d932530d7c : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f955d061a63488485a30b6d7e2772c1 : SUCCESS in 4m 29s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/db9b856cd11040b283ada28e4e2bfd8c : SUCCESS in 33m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/17d0f33c97b24a94aca2291b57e9656a : SUCCESS in 53m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1400cec6ef0d4dfea10d3e6c000729c7 : SUCCESS in 1h 04m 26s","accounts_in_message":[],"_revision_number":5},{"id":"49ba4ae0ac521d2ba40da7cf54c419f8ff61869a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 17:42:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"fcecd247ae46902afa68abdde93d89a2dddcf670","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 17:44:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8daa48dee1d241fd82b0211c2aff5912 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ff6d406d2654ee8a6d92b808e54f1fe : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3df1a1dde9dda0b4dfd6c91da65b53e58bb0d219","revisions":{"46e1870949ece14c51b825813d2827ece87e865f":{"kind":"REWORK","_number":1,"created":"2021-10-21 11:34:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/814900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/814900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/814900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/814900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/814900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/814900/1"}}},"commit":{"parents":[{"commit":"6014b503ddab254e2a8a9db397c46b91fe189ed1","subject":"Merge \"openstack-base: drop anyjson\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6014b503ddab254e2a8a9db397c46b91fe189ed1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-21 11:33:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-21 11:34:24.000000000","tz":120},"subject":"ovn: Add Debian images","message":"ovn: Add Debian images\n\nChange-Id: I645f5d04ff2be3a5857765e901eccfe0b56ff199\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46e1870949ece14c51b825813d2827ece87e865f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46e1870949ece14c51b825813d2827ece87e865f"}]},"branch":"refs/heads/master"},"ec29d910d8cc3e4442c0ddc27cb8ac0542573609":{"kind":"REWORK","_number":2,"created":"2021-10-22 07:54:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/814900/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/814900/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/814900/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/814900/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/814900/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/814900/2"}}},"commit":{"parents":[{"commit":"6014b503ddab254e2a8a9db397c46b91fe189ed1","subject":"Merge \"openstack-base: drop anyjson\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6014b503ddab254e2a8a9db397c46b91fe189ed1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-21 11:33:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-22 07:53:49.000000000","tz":120},"subject":"ovn: Add Debian x86 images","message":"ovn: Add Debian x86 images\n\naarch64 still has no installation candidate\n\nChange-Id: I645f5d04ff2be3a5857765e901eccfe0b56ff199\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec29d910d8cc3e4442c0ddc27cb8ac0542573609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec29d910d8cc3e4442c0ddc27cb8ac0542573609"}]},"branch":"refs/heads/master"},"08fbd2427a86b87dfb8822790c6d66eae279f04f":{"kind":"REWORK","_number":3,"created":"2021-10-22 13:07:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/814900/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/814900/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/814900/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/814900/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/814900/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/814900/3"}}},"commit":{"parents":[{"commit":"6014b503ddab254e2a8a9db397c46b91fe189ed1","subject":"Merge \"openstack-base: drop anyjson\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6014b503ddab254e2a8a9db397c46b91fe189ed1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-21 11:33:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-22 13:07:07.000000000","tz":120},"subject":"ovn: Add Debian x86 images","message":"ovn: Add Debian x86 images\n\naarch64 still has no installation candidate\n\nChange-Id: I645f5d04ff2be3a5857765e901eccfe0b56ff199\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08fbd2427a86b87dfb8822790c6d66eae279f04f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08fbd2427a86b87dfb8822790c6d66eae279f04f"}]},"branch":"refs/heads/master"},"7cf012eec9d6351568c36aee48b438b8c8e266b1":{"kind":"REWORK","_number":4,"created":"2021-10-22 13:16:54.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/00/814900/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/814900/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/814900/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/814900/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/814900/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/814900/4"}}},"commit":{"parents":[{"commit":"6014b503ddab254e2a8a9db397c46b91fe189ed1","subject":"Merge \"openstack-base: drop anyjson\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6014b503ddab254e2a8a9db397c46b91fe189ed1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-21 11:33:59.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-22 13:16:53.000000000","tz":0},"subject":"ovn: Add Debian x86 images","message":"ovn: Add Debian x86 images\n\naarch64 still has no installation candidate\n\nChange-Id: I645f5d04ff2be3a5857765e901eccfe0b56ff199\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cf012eec9d6351568c36aee48b438b8c8e266b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cf012eec9d6351568c36aee48b438b8c8e266b1"}]},"branch":"refs/heads/master"},"3df1a1dde9dda0b4dfd6c91da65b53e58bb0d219":{"kind":"REWORK","_number":5,"created":"2021-10-22 13:17:07.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/00/814900/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/814900/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/814900/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/814900/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/814900/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/814900/5"}}},"commit":{"parents":[{"commit":"6014b503ddab254e2a8a9db397c46b91fe189ed1","subject":"Merge \"openstack-base: drop anyjson\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6014b503ddab254e2a8a9db397c46b91fe189ed1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-21 11:33:59.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-22 13:17:07.000000000","tz":0},"subject":"ovn: Add Debian x86 images","message":"ovn: Add Debian x86 images\n\naarch64 still has no installation candidate\n\nChange-Id: I645f5d04ff2be3a5857765e901eccfe0b56ff199\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3df1a1dde9dda0b4dfd6c91da65b53e58bb0d219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3df1a1dde9dda0b4dfd6c91da65b53e58bb0d219"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
