)]}'
{"id":"openstack%2Fkolla~814915","triplet_id":"openstack%2Fkolla~master~I50ec5dd7e97028a15758c46eee8c45d8052d0363","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2021-10-21 16:07:04.000000000","reason":"Change was abandoned"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-10-21 14:43:28.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I50ec5dd7e97028a15758c46eee8c45d8052d0363","subject":"Drop \u0027--infra-rename\u0027 argument","status":"ABANDONED","created":"2021-10-21 13:16:16.000000000","updated":"2021-10-21 16:07:04.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"eadceea578431380cadeeef24775b46947e04ffd","_number":814915,"virtual_id_number":814915,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-10-21 13:43:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-21 13:36:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-21 13:43:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-21 13:57:15.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"}],"messages":[{"id":"7b1dbdfc0f40cfc6ad373f8498416a147b8b7da9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-21 13:16:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9069324193e584a4c2e4cf1a4aacd58abdd29514","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 13:36:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b28d64f1560444fc97677efec2658b58 : FAILURE in 9m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/efad29f80d3c449c95d77cd1a14e6fdb : FAILURE in 8m 31s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8ec7a6c5c6e8407395d1ea483619d133 : FAILURE in 17m 31s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ee6b9e62e544b3580ed185de73d5012 : FAILURE in 6m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b01c07e954844cf6bab02378f6cc3ae8 : FAILURE in 6m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6253c07c10ad4747ae3df5e83f31a174 : FAILURE in 7m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"98915671c6b6917e563c76c5bc067d26d0bb0602","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 13:43:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/923e42f4b44244348226391bc02d20c1 : FAILURE in 6m 38s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ee2931acb97b4b8d8edd53c3f95820c4 : FAILURE in 5m 44s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7c4c482e6b6f4405adb0318a93934439 : FAILURE in 12m 55s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c1477bc957cf4456811c74c2e1b67425 : FAILURE in 14m 53s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ef248e59116b4ce0ac00894393fad4c3 : FAILURE in 13m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/913d9a259a8144d08201d4b001037a65 : FAILURE in 8m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b85d71fa847a4aa3a77a1add416e6f02 : FAILURE in 4m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4511f312cbc5496894633e4ef73b41a8 : FAILURE in 16m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16814a54e63c460fa0ce2f1f498666cd : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27be5cf0e782477daa46807d70de38a2 : FAILURE in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/303514269f004834ac4de912b8b6d6d6 : FAILURE in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/099f96cd4ecc42c195aff1ed7b59d480 : FAILURE in 4m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7451e454396444a8a5350eb6eb611bc7 : SUCCESS in 6m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/de4a18630cef4dbf979a4e97d889e31b : FAILURE in 6m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6d17c3c154134750a88fc80ffecad8ec : FAILURE in 5m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d62e61d3797e46b59ed07f446ca712f6 : FAILURE in 10m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7ffa1a39f36541eaa13cd14082e60d4f : FAILURE in 10m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b9b0820cf417434ab5a3020c5def08fb : FAILURE in 11m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e3490a6a808d63dac063589fd4b6607cc92e6acd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-21 13:57:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"640757325d1dba7348d0b4d2ffb44c1b2070eaf4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-21 14:10:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce06f1e50195c2ebcc8a800661738cb5be6cfbce","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-21 14:43:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eadceea578431380cadeeef24775b46947e04ffd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-21 16:07:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f3e87ae89745fc94f18010275e204297076ebe6","revisions":{"7f3e87ae89745fc94f18010275e204297076ebe6":{"kind":"REWORK","_number":1,"created":"2021-10-21 13:16:16.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/15/814915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/814915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/814915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/814915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/814915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/814915/1"}}},"commit":{"parents":[{"commit":"6014b503ddab254e2a8a9db397c46b91fe189ed1","subject":"Merge \"openstack-base: drop anyjson\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6014b503ddab254e2a8a9db397c46b91fe189ed1"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-21 08:29:26.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-21 08:29:26.000000000","tz":120},"subject":"Drop \u0027--infra-rename\u0027 argument","message":"Drop \u0027--infra-rename\u0027 argument\n\nThere is a plan to move to source only images. So infra images stuff\nwould not be needed \u003d\u003d this switch is worthless.\n\nChange-Id: I50ec5dd7e97028a15758c46eee8c45d8052d0363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f3e87ae89745fc94f18010275e204297076ebe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f3e87ae89745fc94f18010275e204297076ebe6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
