)]}'
{"id":"openstack%2Fkolla~814944","triplet_id":"openstack%2Fkolla~stable%2Fxena~I8d7632e84dc26ed2022541996a529c2263133c6a","project":"openstack/kolla","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2021-10-22 17:55:04.000000000","reason":"Marcin Juszkiewicz replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-10-25 11:06:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8d7632e84dc26ed2022541996a529c2263133c6a","subject":"update information about unbuildable images","status":"MERGED","created":"2021-10-22 17:49:11.000000000","updated":"2021-10-25 11:12:23.000000000","submitted":"2021-10-25 11:06:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814944","cherry_pick_of_change":814788,"cherry_pick_of_patch_set":1,"meta_rev_id":"355c8934dab60f34da867a0044ec12eb0d4cbc72","_number":814944,"virtual_id_number":814944,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-25 11:06:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-10-22 17:55:04.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-10-25 08:32:43.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-24 08:31:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-10-25 08:32:43.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-22 17:49:11.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-10-22 19:32:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-24 08:31:51.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2021-10-25 08:32:43.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"069798087a13c98b5c64477dedf223959c6a3ebe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-22 17:49:11.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8abfbe2df248226e48ea21ce5d14f6ab11598254","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-22 17:55:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"edb25fc916e465359c13c5bb5cdb7599e5e040ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 19:32:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/eaf344680d6b4da389ff2f00867c7b57 : SUCCESS in 35m 27s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/49f411a1886f4001a9f7204b434b6ef8 : SUCCESS in 40m 39s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/acb64086df494451ae0142c8f37a00fb : SUCCESS in 1h 10m 11s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e093d94e8b5947fdba9a604c228cc756 : SUCCESS in 1h 08m 01s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d397963d76564c4cb91b645e5e201fd7 : SUCCESS in 1h 35m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/56330a1e7ae8420b866c380acfc98d46 : SUCCESS in 22m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6da3beea5ac24a32a44e5459b3a53473 : SUCCESS in 17m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5b9b7dcdbb8643498f4b4e19d0c6d1aa : SUCCESS in 1h 02m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e6558cb727a4ef68b700211a001d17f : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76cfa0a6b12c4fd2b38f90a2f82a162e : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5b8c217be43437aa254df27d04e4847 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1903922625fe4fbc86b4b8b8a318d895 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b096dc713eab49d2bbd93a9367fdd509 : SUCCESS in 6m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ced9ac7ffee44d5cacdcf889e02072fa : SUCCESS in 22m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3996e1610fcc40559e68b09943d6741b : SUCCESS in 32m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/767772de58ff48498253ca7b60324bc2 : SUCCESS in 49m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/719afdec939a4e4f9e2bcc59a2f67ee4 : SUCCESS in 45m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2833b8c3f2f54049be695ef0d2834442 : SUCCESS in 1h 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"aa1bf0887458bf6978da2e64517f93947bef1f44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 19:46:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ce36169d8274a63a1fbc454a1d2aa84 : SUCCESS in 1h 19m 12s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d2927bbf6642486791e9a6208980c49f : SUCCESS in 1h 10m 06s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/da4ca82cd6694d6d8f79c31ef8f95e5a : SUCCESS in 1h 55m 22s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/93070e6a172e46c5914aeca13ac8b4c8 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0e8c0f02f13402a95467bc29a35b8d0 : SUCCESS in 7m 04s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5c0217a41b804301aee7d33080f39997 : SUCCESS in 1h 12m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2ac2597418e36b74457bb499a50ebe578b5ddca","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-10-24 08:31:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4fbfdc24fd74ee999d506b5be872a50ac4f3865a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-25 08:32:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5211b47ca6da7e3399b1c3204d3c5e11f1fbfbc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 08:33:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2eb196efe25d84fb86cb6faeef1007c1980a831","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 11:06:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2d55f383b5da4046b73113e8c3e2c95c : SUCCESS in 34m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/73f3ab5ee423437e83606763b99b826b : SUCCESS in 1h 13m 40s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/89f87e9cb1cb4e0fa58d9b2a80636dcc : SUCCESS in 1h 26m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1b95a8b979f430da2114887514ad8e9 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86fd11f6afab45e4bae71b36caf5b8c0 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a7f30c54caf4520b951ac7d3e615f00 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c095f6fa44074004a67ec0ff31af633d : SUCCESS in 5m 55s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/701ace872df241bfae6e6d3100548dd2 : SUCCESS in 32m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f1f49370ee23437cacfd2ff5a077e216 : SUCCESS in 56m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f388fdc677424bd29053b53d4513dadb : SUCCESS in 1h 15m 35s","accounts_in_message":[],"_revision_number":1},{"id":"dc08f18fb42bffea96bd4c9184f9e65dcebf98fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 11:06:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"355c8934dab60f34da867a0044ec12eb0d4cbc72","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 11:12:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22bfdf409a2f4d42918883299376ce82 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"698866e432423102a4237197cb75c52991e09819","revisions":{"698866e432423102a4237197cb75c52991e09819":{"kind":"REWORK","_number":1,"created":"2021-10-22 17:49:11.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/44/814944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/814944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/814944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/814944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/814944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/814944/1"}}},"commit":{"parents":[{"commit":"9eed029a83acd18d8403312ee3b9dbba0bd8999e","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9eed029a83acd18d8403312ee3b9dbba0bd8999e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 14:18:26.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-22 17:49:11.000000000","tz":0},"subject":"update information about unbuildable images","message":"update information about unbuildable images\n\nChange-Id: I8d7632e84dc26ed2022541996a529c2263133c6a\n(cherry picked from commit 65cf4146ed17aa6d00d917c34343abc18fe8fb15)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/698866e432423102a4237197cb75c52991e09819"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/698866e432423102a4237197cb75c52991e09819"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
