)]}'
{"id":"openstack%2Fkolla~815441","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~Id488635110c6158dd23466250c9d57c4f0bc03ff","project":"openstack/kolla","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2021-10-28 09:29:23.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-10-28 06:41:43.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"Id488635110c6158dd23466250c9d57c4f0bc03ff","subject":"cinder-volume/ubuntu: add lsscsi and nvme","status":"MERGED","created":"2021-10-26 14:25:00.000000000","updated":"2021-10-28 09:31:07.000000000","submitted":"2021-10-28 09:29:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815441","cherry_pick_of_change":815440,"cherry_pick_of_patch_set":1,"meta_rev_id":"a9e0edaabcb5265a4810397ee500e0d88e113bb2","_number":815441,"virtual_id_number":815441,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-28 09:29:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-10-27 09:09:38.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-10-27 18:51:59.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-10-27 18:51:59.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-26 14:25:00.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-10-26 15:58:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-27 09:09:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-27 18:51:59.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"}],"messages":[{"id":"0b056daf39d140c0a457645de54a809461f2f136","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-26 14:25:00.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"b5c0fb3e68011041c97c093fb1504a547c2ac003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 15:58:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/0f862e585c804eaeb481650f71278353 : SUCCESS in 19m 24s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/22ef59ea501d49488cc927c9c9f8cf11 : SUCCESS in 1h 02m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/aefedefb3e9241829bfc66f31c04bbec : SUCCESS in 1h 10m 03s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f7eb54d5158d4073bc641bdf74473170 : SUCCESS in 1h 15m 09s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/660b644dae704c2291f6f41eb05886cb : SUCCESS in 1h 24m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a9e3c2aa7f2a424c8b1f489de9408b35 : SUCCESS in 33m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/33b1f98647544cdd9ae131124a8c45ba : SUCCESS in 23m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8cb66ca702f84dc1a877509af103dffe : SUCCESS in 1h 00m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac0a2b65118f482881bf401d4671b95d : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35d26b7dabdd42799f45d854e65c0ae0 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7a36055fd7d4f45883f2078335b2a04 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e25809ab2e947bf9685d16c8113566c : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/486aab9ac8e240ba953a6b9b5ebd05a5 : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a00df55d275944cdb9210303ebfcc5fb : SUCCESS in 6m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1d13b59575a44f119ca5a47bc340578d : SUCCESS in 18m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1093e63522c9454f99e17ec74886189b : SUCCESS in 44m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/98de48feb3a147c69bd830c2aaed2f8a : SUCCESS in 1h 01m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b3416faf78ab4aa58c228585187f8abe : SUCCESS in 51m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9ba3092ab84544139d1e1bb49c978cc9 : SUCCESS in 59m 33s","accounts_in_message":[],"_revision_number":1},{"id":"eebc1ef0714c64db4e1bb4ccc02ecfedf05d4784","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 16:31:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/459fede3af714b6a9a5b78722e982fc5 : SUCCESS in 1h 18m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/039a074e4c9144eca45e7a604cb6dfb3 : SUCCESS in 45m 59s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/44ed5280dcbe4d6ab3eccac06bcdb78f : SUCCESS in 2h 04m 41s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3ca1499df88f4360921109402bb455f4 : SUCCESS in 1h 26m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8da4a283896b30842bd215f8ccc61865362cb70f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-27 09:09:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"824dca68b98c635baef196703b50ed086237e15a","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2021-10-27 18:51:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b92de0d3445eccf5b9c0c14a07268bd84ca4c38a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 18:52:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2e3a20c111dc12fd378c639822882c13f466c48c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 20:08:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b93378ab48244fa1bf108a4de875b13c : SUCCESS in 45m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e75c4cf86b29464480fd717987f72ffa : FAILURE in 26m 44s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2d1f2083bf4247db8d5bec57be3186a5 : SUCCESS in 1h 14m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2905182eb1054dfc818ebc297e7aaec9 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8486f2d9d8884ae1ab94facfcb2d4d96 : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c58574790af4150810a0ba2fdedb3ec : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aadc3fa301cc456c91401d92ee53ff33 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee12161f46ad46ee945b38979bc61dc1 : SUCCESS in 5m 51s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/837a4ac627da4caa8efa0be1b358bcbc : SUCCESS in 41m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2651a83e9a82465385eb33d9fd5483fc : SUCCESS in 59m 43s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3adbb9a63a0f4038a7098814611049e5 : FAILURE in 56m 52s","accounts_in_message":[],"_revision_number":1},{"id":"67b157fd124974c73cf93ad4b8003e3cb5ec8df8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-28 06:41:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"627867a3b9f16e6ac50cb13300d4a5b934f58251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 08:08:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fb8814b3f6f54289a1eaaf529e9e507a : SUCCESS in 24m 40s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/64e4f92060aa4454a0f6881adb511732 : SUCCESS in 38m 35s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/357b4e5f9a294326a5f4c8197678d321 : SUCCESS in 1h 16m 47s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/1aa563577dd345ae86d73952e181a644 : SUCCESS in 1h 02m 32s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/715d56473ab74c7aac4e85e7b3fcc1d7 : SUCCESS in 1h 22m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/96937edb5a894207b05cdb9b06eeb4b9 : SUCCESS in 31m 45s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/42cac826c6c54a4f99e30027a7a5d6db : SUCCESS in 19m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ad0e2f8acca04589b0aeb12d78b5ef4d : SUCCESS in 55m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0513820b4e7f4ac2b4f4bf719fe4c620 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a68a84b1b0b34a3eb037b2e5877d7a0e : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18ec1e54f4e64638ae7959b8accfe657 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6aeed1762340460c908bfe2b9bb292f1 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e18b445fa49a45dcb001d5327ae6c54a : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d87a481ca4b4adc9bcbf9b04ab097cb : SUCCESS in 4m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9939aa6cbac646fda68ee0097ec8e2d6 : SUCCESS in 20m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ec9c1ebb85d04dc38a09efdbe6fe888e : SUCCESS in 26m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d9e662a6c0d14417ac22a53bff54bf02 : SUCCESS in 50m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/930da6f76acc4a5ba5d9324c2a6ff665 : SUCCESS in 51m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d70a779fbaf04acc8cc2901535b94e92 : SUCCESS in 56m 42s","accounts_in_message":[],"_revision_number":1},{"id":"9e6acb67ed8ae23d10bcb60e9f769716146f9b9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 08:08:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b453f44768cf3a488acfdebc8109d0bf57686b5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 08:48:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/35a0bf371ef444468db6ee29af86bb36 : SUCCESS in 51m 59s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ea88896ac79f49d5a8a0d48ab1d93a29 : SUCCESS in 1h 21m 05s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/61a0bb79296143d2ba6dc83a7976f891 : SUCCESS in 1h 59m 50s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0f6020a0876541f2b1f65a341d4cbfc9 : SUCCESS in 42m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"507df651d67b0eca471df71e8b48dff56d0865ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 09:29:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e9d493162a144648ab2415833aa060d0 : SUCCESS in 51m 26s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/24a67b631e914343bf8ecdc537ce7d47 : SUCCESS in 1h 13m 16s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9f2e2ba9bb6549648f7c4cd39955b62e : SUCCESS in 1h 15m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1da25797751a4c88ba24586b1dcbc5e7 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe041a3b772a4424bc8da9625551f413 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c01bf3dc6f447dca0540ad13d4d4dbc : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea24081f8a00437bafcd807e48eecf2c : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ae7ab994d5840eb9da098cea663393d : SUCCESS in 4m 43s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/af6cff029c69488cb60ca425d6b677fa : SUCCESS in 37m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c9c880cbcee34fa29d0f8940e4c34ea6 : SUCCESS in 50m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1727a8323e824db58c1b42c692cd216e : SUCCESS in 54m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f6cb6532e584ec3abfd69e842f67dd10365887f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 09:29:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a9e0edaabcb5265a4810397ee500e0d88e113bb2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 09:31:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8d48de9c4c141b3b890c0066983679b : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03717acb1f274563bc8f3a64b356c400 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6448d3d21e13f1c70ceb20acb89583c58d594e52","revisions":{"6448d3d21e13f1c70ceb20acb89583c58d594e52":{"kind":"REWORK","_number":1,"created":"2021-10-26 14:25:00.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/41/815441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/815441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/815441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/815441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/815441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/815441/1"}}},"commit":{"parents":[{"commit":"c9a71e977eb731755b27270613af5b72b63ef9f1","subject":"gnocchi: update to 4.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9a71e977eb731755b27270613af5b72b63ef9f1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-12 11:24:47.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-26 14:25:00.000000000","tz":0},"subject":"cinder-volume/ubuntu: add lsscsi and nvme","message":"cinder-volume/ubuntu: add lsscsi and nvme\n\nCloses-Bug: #1942038\nChange-Id: Id488635110c6158dd23466250c9d57c4f0bc03ff\n(cherry picked from commit dfb65af8845b184c533fab451fd609b2f7de8859)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6448d3d21e13f1c70ceb20acb89583c58d594e52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6448d3d21e13f1c70ceb20acb89583c58d594e52"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
