)]}'
{"id":"openstack%2Fkolla~815442","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~Id488635110c6158dd23466250c9d57c4f0bc03ff","project":"openstack/kolla","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2021-10-27 20:30:15.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-10-27 20:30:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id488635110c6158dd23466250c9d57c4f0bc03ff","subject":"cinder-volume/ubuntu: add lsscsi and nvme","status":"MERGED","created":"2021-10-26 14:25:08.000000000","updated":"2021-10-27 20:32:00.000000000","submitted":"2021-10-27 20:30:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815442","cherry_pick_of_change":815441,"cherry_pick_of_patch_set":1,"meta_rev_id":"fac8749fa5e4123d80e25ce9c9ada4732a147988","_number":815442,"virtual_id_number":815442,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-27 20:30:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-10-27 09:09:57.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-10-27 18:52:18.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-10-27 18:52:18.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-26 14:25:08.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-10-26 15:44:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-27 09:09:57.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-27 18:52:18.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"}],"messages":[{"id":"a2fd4870f849d461e45f799bc6bd75d73d84ab4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-26 14:25:08.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"842938809252e30e315dcd6f7a606e9aa593c6cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 15:44:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/13786f64731643f983db1db89df6ca05 : SUCCESS in 24m 23s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/de91149ab1d54de2bb5dfa80e6a064a6 : SUCCESS in 24m 50s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f2220804ad1e4a4397ef6ab032993084 : SUCCESS in 33m 22s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/03ef9940844c4b1c940f6ad014f0288c : SUCCESS in 26m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c41a6e837c684fb4be9df03b479bd5e5 : SUCCESS in 59m 49s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/705228110e244372b228cddd6700efa5 : SUCCESS in 1h 01m 48s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0bcd6e754d534f07a2911a62761847f3 : SUCCESS in 1h 04m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/530094fe7278428ba8cf4a52f3762c21 : SUCCESS in 1h 00m 46s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c9ee9988505f49fb9f4f8f0bd83274f5 : SUCCESS in 1h 05m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9b1474ab51b94809a84016b86117aa60 : SUCCESS in 1h 13m 46s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e7d27e4f785540f683a5d19426203dc8 : SUCCESS in 20m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d09a1cc179564f33844dab7a1e9a7bef : SUCCESS in 23m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e7e1df6a8432471aadb8aa3f10bbaec6 : SUCCESS in 53m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56cf91de5193490da7ec8924a104bfb1 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8e9860e14bc40d599ba1d7500a3cdf2 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c6d2cad08c14d87862322f30dc93173 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/416be66cd11b4b66a8ab9cf82233a089 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/191cb27eb23e42f08dea6c5915c945d1 : SUCCESS in 5m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/281b1de5771a48f4902c111ca2404dc3 : SUCCESS in 25m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/31059cde3f3b4098bcaae9d082730bc0 : SUCCESS in 24m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/af6766755827426dbfec68d39ce4cd2b : SUCCESS in 58m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4850be778a1341e09e69c473436867d2 : SUCCESS in 58m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/45ffadfdc9a1417f94a902a11b834b59 : SUCCESS in 1h 04m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b540d18b52a9eac7d1ae34f048e79ad0b409a0e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 16:56:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/ec3a2ee0041749d1985aac33672a702b : SUCCESS in 1h 44m 08s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d6912095e9fe494083aef094099ee740 : SUCCESS in 1h 12m 33s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/743f543d37ce4005bb12f15f97e1e14a : SUCCESS in 1h 03m 00s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/77771ed5bc2a4934a8b084d3e0965bb7 : SUCCESS in 2h 28m 21s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ee6163573dc74f6f92cb7604cd060042 : SUCCESS in 1h 32m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0fc91c810eb4683fb82918091c4553ecc1942945","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-27 09:09:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9a257a11b3201df222344619cff43d07f74d2918","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2021-10-27 18:52:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"51e51002649d672714a641b6e4495771e9110a84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 18:53:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"55aac4392fcd05b9eb40d5a9cc37fad08afb8c0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 20:30:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/24016c9688fa48588f547f62efbcb092 : SUCCESS in 28m 53s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/dd5f95984dc64aec83c3f257ab9d9beb : SUCCESS in 31m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5ae68f56be8b4d76b0a08a670900b11d : SUCCESS in 1h 02m 47s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/648be5ebe0384acc8983338d5968d246 : SUCCESS in 1h 06m 44s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/dbe92315b1e347d985aff6ba3a36e1a0 : SUCCESS in 1h 02m 29s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b7de2f0229ee4e7ea3b2a8e742ff95dc : SUCCESS in 1h 08m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a4aa4ac0bb64101ae92c8e035e08bdd : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c401be8280f447dbe4e2779f45269b7 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36e63f0ec86847c8a7956488dd3de79a : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/567d5aa5b4d9470c83bcf5779bf57f5e : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b3d2a48d339432999b285a70646b42a : SUCCESS in 6m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d1ac1da2481a4cdd9c584b8f4140db8d : SUCCESS in 23m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/207f5fdf55874bfab48a333feb14fbe0 : SUCCESS in 53m 13s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e7adf02c1a9b43c2bdd52f708e707bbd : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":1},{"id":"562f3e16a7bb72d29c79a6472e604d50514cb2e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 20:30:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fac8749fa5e4123d80e25ce9c9ada4732a147988","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 20:32:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/819d8d033c984c7a9154a54b3c3d6c92 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4fa0f8b006f459ea062936f9383c437 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"49a5b7e55381b4b1ef7aa0bb23178f13a924ee62","revisions":{"49a5b7e55381b4b1ef7aa0bb23178f13a924ee62":{"kind":"REWORK","_number":1,"created":"2021-10-26 14:25:08.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/42/815442/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/815442/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/815442/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/815442/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/815442/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/815442/1"}}},"commit":{"parents":[{"commit":"8c9bcb0e7dcb652227995db1af3e9f6205943371","subject":"[debian] Use upstream MariaDB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c9bcb0e7dcb652227995db1af3e9f6205943371"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-12 11:24:47.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-26 14:25:08.000000000","tz":0},"subject":"cinder-volume/ubuntu: add lsscsi and nvme","message":"cinder-volume/ubuntu: add lsscsi and nvme\n\nCloses-Bug: #1942038\nChange-Id: Id488635110c6158dd23466250c9d57c4f0bc03ff\n(cherry picked from commit dfb65af8845b184c533fab451fd609b2f7de8859)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49a5b7e55381b4b1ef7aa0bb23178f13a924ee62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49a5b7e55381b4b1ef7aa0bb23178f13a924ee62"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
