)]}'
{"id":"openstack%2Fkolla~815443","triplet_id":"openstack%2Fkolla~stable%2Fussuri~Id488635110c6158dd23466250c9d57c4f0bc03ff","project":"openstack/kolla","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2021-10-27 20:39:44.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-10-27 20:39:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id488635110c6158dd23466250c9d57c4f0bc03ff","subject":"cinder-volume/ubuntu: add lsscsi and nvme","status":"MERGED","created":"2021-10-26 14:28:14.000000000","updated":"2021-10-27 20:45:48.000000000","submitted":"2021-10-27 20:39:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815443","cherry_pick_of_change":815442,"cherry_pick_of_patch_set":1,"meta_rev_id":"57fc06f54429eb4be41f8a1da77d05d8c9243149","_number":815443,"virtual_id_number":815443,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-27 20:39:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-10-27 09:10:16.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-10-27 18:52:30.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-10-27 18:52:30.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-26 14:28:14.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-10-26 15:52:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-27 09:10:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-27 18:52:30.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"}],"messages":[{"id":"5302335b70710568c2d2a6c6bca1669db901fa44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-26 14:28:14.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"99037b2bf2474b5e1d192ad7c894d9bdf1d66381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 15:52:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/59d853c859fd46af9ccfd85524b92cb3 : SUCCESS in 24m 58s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/20acfa048e784209a61a118e9e15d4cb : SUCCESS in 31m 14s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/af3ebfd510a641eb83eb344b4f1a7dee : SUCCESS in 1h 09m 02s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1799af65ad2f4432a3135ea85537411a : SUCCESS in 1h 03m 07s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/bcd0618969f94e8f88b01d55092ebe8b : SUCCESS in 1h 07m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/73875ea625494aacb1812c81f3eef38f : SUCCESS in 23m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a0798241bb2d4e0aadf6b95ce559f94c : SUCCESS in 20m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6516c46e71f748e1a86e8da145dfd3da : SUCCESS in 54m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e51dd5e1d874bb7afefe64c0af4d02c : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea980ae197e246be920eae7ec87e4b45 : SUCCESS in 6m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f3d7b42c71be434b9be88dfefffd7116 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4017427461641b59121d849e7dafd5e : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e2241b979904386aad8750a94182a98 : SUCCESS in 7m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2239b1acfeca4498964008fcc87775b4 : SUCCESS in 30m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/84220056bde24b338656c7e0a0383666 : SUCCESS in 28m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/313828070cf04fada3a3948c6c41d087 : SUCCESS in 55m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0d030ce9cc6d45d786b3c87375a533e2 : SUCCESS in 48m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a3f1620da5a244528ef655808a2e5522 : SUCCESS in 53m 57s","accounts_in_message":[],"_revision_number":1},{"id":"cf2e453097f562a1f24769584861122931f3985e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 16:47:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/b3f0735e85da482faabe640a3f269d28 : SUCCESS in 1h 09m 55s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/636486fb6def4f00921dc40d6f8a3250 : SUCCESS in 1h 07m 06s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0e5ab624066f4af7a5e0f15036456a1c : SUCCESS in 2h 17m 41s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0c5c4f4ac2d14786a0da828de0318bb6 : SUCCESS in 1h 23m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed44d0067daa57d21ca1362c14a168b87ad13f13","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-27 09:10:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5df066ace3dd378289be47d3ad3d755fd0de1860","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2021-10-27 18:52:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"01d28e525d8136aaaba265209e3635147ee77e2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 18:53:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"135f0742c09157ab14115f48b2fbd972cd9f50c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 20:39:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/deb178c7c9a14de9a2b56392db572a1d : SUCCESS in 31m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c006b85fd36f4c9fb0f296412d836172 : SUCCESS in 1h 05m 03s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/04f7dcd4dd3a4e99ab84b9f70a21a43b : SUCCESS in 1h 16m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51d6b0178b344b8caf16ab9e4d3a3b66 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/638d202e10b14b40a3670d31b57e6354 : SUCCESS in 7m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a41212c8ce24f95b61055c682d30297 : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c1afebb169c453fa329f3aaf82a3204 : SUCCESS in 8m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8aa6942afc8549709e2cc46d13e6667f : SUCCESS in 6m 22s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/280a749453d14356b9e74ef8f7886066 : SUCCESS in 29m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/85beffe25c87487d841275eb998381c1 : SUCCESS in 55m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fd16f0d2c23b4f349d2ec2b0cf88249e : SUCCESS in 1h 04m 15s","accounts_in_message":[],"_revision_number":1},{"id":"e1b1ec49298423e27392d0fdec1e824df3ff000b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 20:39:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"57fc06f54429eb4be41f8a1da77d05d8c9243149","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 20:45:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4339380a0934b3facb7ba77d2c11bcc : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6d116e78099405492305ff1275c177f : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5445c1e54a1e96d568bc9f45b066e479640e3895","revisions":{"5445c1e54a1e96d568bc9f45b066e479640e3895":{"kind":"REWORK","_number":1,"created":"2021-10-26 14:28:14.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/43/815443/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/815443/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/815443/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/815443/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/815443/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/815443/1"}}},"commit":{"parents":[{"commit":"f179f8383bec039bd82670217faa7e16ef17fcfa","subject":"neutron: Fix update-alternatives on CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f179f8383bec039bd82670217faa7e16ef17fcfa"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-12 11:24:47.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-26 14:28:14.000000000","tz":0},"subject":"cinder-volume/ubuntu: add lsscsi and nvme","message":"cinder-volume/ubuntu: add lsscsi and nvme\n\nCloses-Bug: #1942038\nChange-Id: Id488635110c6158dd23466250c9d57c4f0bc03ff\n(cherry picked from commit dfb65af8845b184c533fab451fd609b2f7de8859)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5445c1e54a1e96d568bc9f45b066e479640e3895"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5445c1e54a1e96d568bc9f45b066e479640e3895"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
