)]}'
{"id":"openstack%2Fkolla~815781","triplet_id":"openstack%2Fkolla~master~I78da7eb2b6b5eae2b04fd95b24615bd4a572a1d6","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2021-11-01 12:49:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I78da7eb2b6b5eae2b04fd95b24615bd4a572a1d6","subject":"neutron: Use update-alternatives --display instead of --query","status":"MERGED","created":"2021-10-28 07:20:20.000000000","updated":"2021-11-01 12:50:50.000000000","submitted":"2021-11-01 12:49:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"815781","meta_rev_id":"8422485bc033ad081012e89a130f1f95fe07ad96","_number":815781,"virtual_id_number":815781,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-01 12:49:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2021-11-01 09:24:00.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-11-01 11:21:09.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-11-01 11:21:09.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2021-11-01 12:50:50.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-28 08:31:15.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-10-28 08:54:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-28 16:18:02.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-11-01 09:24:00.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-11-01 11:21:09.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"cc93aeccc63a507abdee7ee1b92efe848886381a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-28 07:20:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfdcd6f538e211a2d826356d3131979d3e7bb408","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-28 08:31:15.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3aa13f4e9fd037d35f8638184ea75fecf2ba5a0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 08:54:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a76e38c5246b400f965d211033d98754 : SUCCESS in 46m 29s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c6c2c11c9a5b4bb09a73c57dd049b39c : SUCCESS in 43m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/256aaee02d3746259971e05e3596fe15 : SUCCESS in 1h 08m 04s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/56ca851abe22456290978560097a5a7a : SUCCESS in 1h 17m 15s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8e35f67bfe4f4f1d866779b04b114387 : SUCCESS in 1h 18m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/19d157d73ea049b6b508c7b1279c909d : SUCCESS in 24m 11s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/909a01c72ee440a5a30f59d89fa665e7 : SUCCESS in 20m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3fcea4f2fa344813ac4d20af3f52e8c2 : SUCCESS in 55m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/079bd182c4874ad39d8d09c159d67fa4 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/038f6284be6e464bb4f15c0681505f2e : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b212c1c1a014c76b7b2303576e1f73a : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c15f04c325e4170bd76a89d27ca7624 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6abfcb15372a4e2cb60d5daecf1c4030 : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2405e6c0248d488b9d9e0ce75c7ff7a0 : SUCCESS in 4m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a18ec9256731400f9ef926736d36315e : SUCCESS in 30m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d246143a349340fd94fce6196539269a : SUCCESS in 40m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/764d43e1ce024a4a89f8b5088da4e9c6 : SUCCESS in 56m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4f9af33c23a247c6b0e780466eca8d39 : SUCCESS in 1h 00m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b3c2042c10ca42b89462d45106454273 : SUCCESS in 1h 32m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b51a0a60866aab7ebab455e4fbcc61ecf0fba719","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 09:22:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d97ae77575614530a7dc1adc5a31e0e9 : SUCCESS in 1h 04m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fb4824b59e464764ab9de9aeb33752e9 : SUCCESS in 1h 05m 00s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/85e7a8a7849849b8a5236884c6ed6d64 : SUCCESS in 1h 53m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/511112be78714de185d3caf2163e61ee : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a5335fc69c04949b96f0d12c0cc9554 : SUCCESS in 5m 22s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6c2618af2039494a8110d74d0b4d5a7a : SUCCESS in 56m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6bf5c34e70811f5013f2a0e3f87798cec90f09fd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-28 16:18:02.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"60bdf445d583b8b5dad6e34c88535d34af0e9a5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-29 09:59:51.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2ed9b3f3850e852e1a7211e39092ad5c2070b771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 11:29:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/25ccbc63591e4872ab00612c4cbe3f9c : SUCCESS in 44m 48s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/20a1e53e395544dc9db0c31ed8ecc940 : SUCCESS in 47m 15s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/585087a63dab4d8ea444f2ea464f88bb : SUCCESS in 1h 21m 51s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/afaa5b5ea99d4e4fa7f793906c782eaf : SUCCESS in 1h 16m 29s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d5741cf32de64770942660806e8eb17c : SUCCESS in 1h 27m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a3a8a8f5a4484052996204dee8f6ebfd : SUCCESS in 29m 03s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/32a0fddca89f4e7aafd66bbdac64dae2 : FAILURE in 17m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8e402b03916d4bf9b1d2496147c09009 : SUCCESS in 48m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50d4c669e8974baead935106ab0bc7b5 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/431b2d363cf749a58b9e2847c7dd26e0 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88cebad0a4e44a96b1da8b6781252148 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ab46d4018e34710bb162e13c27b2a63 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07139a31b8fc405dafa2506839888fc1 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a925739a330742b5aefb5c74fca6afd2 : SUCCESS in 4m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2a21ddf96dec49ad9cc9ab3d0bd45253 : SUCCESS in 19m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8b45bcad24994cfb8b681fbc3b967488 : SUCCESS in 32m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d8be9c866945435e8590c01f5eff72be : SUCCESS in 52m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/80ada755404147ba9fffccbf0fe0470d : SUCCESS in 42m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2852240831604585a22556bd4245726d : SUCCESS in 1h 03m 36s","accounts_in_message":[],"_revision_number":2},{"id":"9a7ca3a24e46c13fba5ceac8e5636b509cef9de5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 11:51:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/fa0826874e2a4d228fc4d5506754fb45 : SUCCESS in 45m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/55e8fd47ac0b47c0ac703d18be9f2ad2 : SUCCESS in 1h 25m 42s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4753d032ea6b41f8957e94cda03d19e8 : SUCCESS in 1h 29m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8a01ef899f34dd3aa88f19ad3d06259 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b14eb3495ef54add97c5b652e44d9aa4 : SUCCESS in 4m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9cd14707389f4a8ea381d5c3db5e9531 : SUCCESS in 1h 49m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"43585abef13bd213b1cb84ef18ccc5b6bc894d47","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-01 09:24:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b52b35afc0460a9087972667b4f4480e5b50e879","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-01 11:21:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5e9f45c34685c93c094029131eaefb84f9626901","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 11:21:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"513b4cbbf8706d934b25533fd80ee507ddf119f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 12:49:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/df0133f67d3d4a21a7d10b7276ad1e0f : SUCCESS in 37m 28s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/386489ec77bd46a984b75a377313acc9 : SUCCESS in 1h 26m 44s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/22d4b55f1d604b49b1af3cf01f06a1dc : SUCCESS in 1h 24m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0aeaf00eaa094d50a6a36577536a2697 : SUCCESS in 29m 52s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/21a73a41629449a78dc902c2987c987d : SUCCESS in 1h 04m 09s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/77f314b31e79434aad99a45ecbbcb7ef : SUCCESS in 1h 13m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4134ccecc6b5498b9cc3eb05bc57ce8b : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d47e1d4f64f8478db2f6730683cc4f08 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36c5a45331be4d3da304e89fc890b034 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e25e19b7de24277aa01ec2f7990bf3f : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ae64c0aa6314a22955728c11e146f72 : SUCCESS in 4m 32s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/78e1f3b51e3b4ceb860b10539da34b42 : SUCCESS in 28m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/257973c100ad4759b8cba78c63981934 : SUCCESS in 51m 44s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a769c1a7115e4e3ba71f51fd5dc03683 : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":2},{"id":"30a6ae55deb36b38a86d6f0396435d2728fee4ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 12:49:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"84a57a5684364114913c53b108d628a315d0ad37","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 12:50:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46af264b99834778a46d15a577fca9ea : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fe8ca58e9c34156b463f12873d0ef88 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2d1bbe72579f9209e151133352d4d6761f764f33","revisions":{"4ae57106e694c4482dd0df41dcee41747a714242":{"kind":"REWORK","_number":1,"created":"2021-10-28 07:20:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/815781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/815781/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/815781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/815781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/815781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/815781/1"}}},"commit":{"parents":[{"commit":"4b3ad066d1a7cf4e93fe774a140b09dd5eedc6a8","subject":"CI: make Debian source jobs voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b3ad066d1a7cf4e93fe774a140b09dd5eedc6a8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-28 07:17:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-28 07:20:18.000000000","tz":120},"subject":"neutron: Use update-alternatives --display instead of --query","message":"neutron: Use update-alternatives --display instead of --query\n\n--query is not supported before update-alternatives 1.15\nand CentOS uses 1.13\n\nCloses-Bug: #1936947\nChange-Id: I78da7eb2b6b5eae2b04fd95b24615bd4a572a1d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ae57106e694c4482dd0df41dcee41747a714242"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ae57106e694c4482dd0df41dcee41747a714242"}]},"branch":"refs/heads/master"},"2d1bbe72579f9209e151133352d4d6761f764f33":{"kind":"REWORK","_number":2,"created":"2021-10-29 09:59:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/815781/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/815781/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/815781/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/815781/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/815781/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/815781/2"}}},"commit":{"parents":[{"commit":"4b3ad066d1a7cf4e93fe774a140b09dd5eedc6a8","subject":"CI: make Debian source jobs voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b3ad066d1a7cf4e93fe774a140b09dd5eedc6a8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-28 07:17:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-29 09:59:51.000000000","tz":0},"subject":"neutron: Use update-alternatives --display instead of --query","message":"neutron: Use update-alternatives --display instead of --query\n\n--query is not supported before update-alternatives 1.15\nand CentOS uses 1.13\n\nCloses-Bug: #1936947\nChange-Id: I78da7eb2b6b5eae2b04fd95b24615bd4a572a1d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d1bbe72579f9209e151133352d4d6761f764f33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d1bbe72579f9209e151133352d4d6761f764f33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
