)]}'
{"id":"openstack%2Fkolla~815824","triplet_id":"openstack%2Fkolla~master~I2bb89a30a88c8b889c5af2fc609e236ed966f7ba","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2021-10-29 17:27:49.000000000","reason":"Change was submitted"}},"hashtags":["cleanup","debuntu","trivial"],"change_id":"I2bb89a30a88c8b889c5af2fc609e236ed966f7ba","subject":"base: drop commented repositories","status":"MERGED","created":"2021-10-28 10:58:07.000000000","updated":"2021-10-29 17:30:01.000000000","submitted":"2021-10-29 17:27:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815824","meta_rev_id":"ad57b2930fac842880883fe1317e14f923958c53","_number":815824,"virtual_id_number":815824,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-29 17:27:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2021-10-29 15:56:43.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2021-10-29 15:56:43.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-28 12:35:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-29 15:56:43.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"674daf2f4e21f706e3072af320bfd459214578a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-28 10:58:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53ddc23755a47c9baca16fefa8becf6e2c43a42b","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-28 10:58:22.000000000","message":"Hashtag added: cleanup","accounts_in_message":[],"_revision_number":1},{"id":"18e3348c33c445144e363125183a23b8094e9391","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-28 10:58:26.000000000","message":"Hashtag added: debuntu","accounts_in_message":[],"_revision_number":1},{"id":"2781200846c3005b0242d19ddf0b1a4868e3c128","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-28 10:58:30.000000000","message":"Hashtag added: trivial","accounts_in_message":[],"_revision_number":1},{"id":"c400f04f992f437992e0c99dda4f381e59a78281","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 12:35:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ee7bd62e96cb4fb787c503f7c119d68b : SUCCESS in 24m 17s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b1e48bb2800d47609c0ea58205b25f1a : SUCCESS in 51m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d4fe5ad813c84895a16260bfb730368c : SUCCESS in 1h 29m 51s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/567c8fc9680849f6aa75bb07078c1ead : SUCCESS in 1h 03m 45s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/68e21a0f7c814fb1a59e87a67757ae66 : SUCCESS in 1h 33m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b92e3c6d022d4c3bbbe8616d951fd5b5 : SUCCESS in 30m 54s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/070cd007f93f44da8adc1a20bce64a2a : SUCCESS in 19m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/69540dff819a4e778f6ef89e48b0626a : SUCCESS in 1h 04m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e6072f05b0d4b57a44daa33ef4c97ba : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7dfcd1f91a54aad88130bf1b977a12a : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71eea45d59fc4faaa9df50981a0fdfca : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d284570eb6b340dc9267d6d3131d241d : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a7d40eea66a40f9bfdcc3c3c1908360 : SUCCESS in 6m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/45077038083b4b18ba12527699fdf5aa : SUCCESS in 21m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6737e6a6d5bd49c49b51a622a2f4df2b : SUCCESS in 37m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3ac5fd13bfbb42f0812e8a76a9b6e9a7 : SUCCESS in 54m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5d08f46b61744ab9a00096c5e903fcf4 : SUCCESS in 1h 08m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/28c02aaeb5a648d7ad52dde646ebdc74 : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":1},{"id":"bf39a748f5254a6f830fa44b9449cf3f92c3c6fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 13:06:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7db9f969b00246d4b249710d2247b165 : SUCCESS in 55m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/687f53da23ac427eb25b2502448d3d8d : SUCCESS in 46m 45s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c4b6ab5895b146479bbf3b70809a0ea6 : SUCCESS in 2h 05m 38s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3ac8a037b734bd6ad8eb3d2902dccb6 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/967e506b937a4f9c8293e7f0d7e4c61c : SUCCESS in 6m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2fdfae2063744394a991a1a4d6a7eafc : SUCCESS in 1h 03m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e2cc285691c70dab14162562a469929ceb5662c3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-29 15:56:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b1d02788ac97c60dda450d9df3af38ca8c5d4c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 15:57:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5fd10e76c67088ece6179803e03de432084b3ea6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 17:27:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2e448076c2fa4997b3def78c1292aaa3 : SUCCESS in 37m 35s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c0a1085b22894792baf599bd9e514790 : SUCCESS in 1h 10m 53s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/60a07acc5686430893e0a7abc925b13e : SUCCESS in 1h 29m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0ffdb75d061149b0a462d7976a326924 : SUCCESS in 32m 40s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dbab8a90eaf14ff6882bc488b39acb97 : SUCCESS in 59m 28s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f4520d8ad22f4bb398bce57c4fe10f18 : SUCCESS in 1h 12m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55d7e77016f24b02843b10e1cd566504 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de354546328d4e59b502ce739c8ba227 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7214fa5093084a97aab3333667c3de2c : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/809bfbd1433a41da8d2265ab06ecc7ea : SUCCESS in 7m 15s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e1ecc89af176413888581430130129b0 : SUCCESS in 33m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1ed022886d8a4c5d8214f0524274e17e : SUCCESS in 49m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/095b58a6e04c4e7fb3b90d9ad2fe6a08 : SUCCESS in 1h 09m 09s","accounts_in_message":[],"_revision_number":1},{"id":"2996ec31d7b805b6b24ba042761098fb844b8e87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 17:27:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ad57b2930fac842880883fe1317e14f923958c53","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 17:30:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/793deab5d39845b8ab323f7e795d6e08 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6738372ae6594884ced9123529e596263d25677a","revisions":{"6738372ae6594884ced9123529e596263d25677a":{"kind":"REWORK","_number":1,"created":"2021-10-28 10:58:07.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/24/815824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/815824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/815824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/815824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/815824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/815824/1"}}},"commit":{"parents":[{"commit":"4b3ad066d1a7cf4e93fe774a140b09dd5eedc6a8","subject":"CI: make Debian source jobs voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b3ad066d1a7cf4e93fe774a140b09dd5eedc6a8"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-28 10:55:37.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-28 10:57:48.000000000","tz":120},"subject":"base: drop commented repositories","message":"base: drop commented repositories\n\nWe moved additional Debian/Ubuntu repositories into\nkolla/template/repos.yaml quite a while ago. It is time to\ndrop commented out copies from APT sources.list file as no one keeps\nthem in sync.\n\nChange-Id: I2bb89a30a88c8b889c5af2fc609e236ed966f7ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6738372ae6594884ced9123529e596263d25677a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6738372ae6594884ced9123529e596263d25677a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
