)]}'
{"id":"openstack%2Fkolla~816170","triplet_id":"openstack%2Fkolla~master~Id24458cd655aeace0e745c512aa7890b3edda51c","project":"openstack/kolla","branch":"master","topic":"some-debian-cleanups","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2021-11-02 18:05:11.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-11-02 16:08:00.000000000","reason":"Mark Goddard replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-11-02 11:35:11.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"Id24458cd655aeace0e745c512aa7890b3edda51c","subject":"gnocchi: do not install cradox","status":"MERGED","created":"2021-11-01 13:29:43.000000000","updated":"2021-11-02 18:16:27.000000000","submitted":"2021-11-02 18:05:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"816170-some-debian-cleanups","meta_rev_id":"7f414e5391b6eddd69aaa555e68c205f41cedb78","_number":816170,"virtual_id_number":816170,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-02 18:05:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-11-02 16:08:00.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-11-02 16:26:24.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-11-02 16:26:24.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-01 14:56:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-02 09:56:50.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-11-02 11:06:10.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-11-02 16:26:24.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"f6933f459e65aa28269f59c6f9857d79f27f95c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-01 13:29:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7647a7234613c3e0e2e210fe6fbf65a21ae1129","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-01 14:18:15.000000000","message":"Topic set to some-debian-cleanups","accounts_in_message":[],"_revision_number":1},{"id":"39b00c4ad880206b43339ae925b8e9f023025957","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 14:56:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/afece3c72dfb452998b6664f2a756120 : SUCCESS in 29m 26s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/41db66efa3c34c269fdabb84672ec166 : SUCCESS in 37m 15s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5b75874ffed64ba18bf0c7123b02b453 : SUCCESS in 1h 05m 55s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6addaac0c5014e8da45df5b5fa2c9852 : SUCCESS in 1h 07m 14s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/979a446982ba484197ea3f74bbc82fab : SUCCESS in 1h 24m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b0930017d7544010b5f53ac0a85e821a : SUCCESS in 32m 39s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/83eb9d835451419eb12158ed6015683e : FAILURE in 18m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9ef2eb076833477298dc0091ab366357 : SUCCESS in 55m 32s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2e9eaa35d9cf4149ba749b9f69ec4373 : SUCCESS in 1h 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62a2645543b54ebbab1fb7860c65aabf : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b23890ab215494da683658bfa323822 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/996d9149a8f14f6daf0e7a19e0a50ce2 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24ec5c6d22704d6dad4636b168d49b70 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d398dfbf94441f2b69ac00da5e1dc3b : SUCCESS in 6m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fd6a5df55bd342df84ea5dabe68e8e74 : SUCCESS in 21m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/391c5bc17dcb455386969df54e6e2e75 : SUCCESS in 20m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b7e37fe718c6409eb8f8a19bddbff141 : SUCCESS in 48m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7c03f714f1394f5fa7bcaedcfdb4e8c2 : SUCCESS in 48m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7ff9edbe322140c2b3c9e6cbf60f8c19 : SUCCESS in 1h 13m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8b467e63a1f2114d6699f100b6046338294a922e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 15:11:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/614009d0805f435c83918730c3542944 : SUCCESS in 49m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/43c3769638bb4fd98d237542279f60ca : SUCCESS in 1h 20m 26s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/501a5c4c20cd44dfbbbd173bd3e1bc6c : SUCCESS in 1h 33m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a1ce9a785d34cbb9805a830ce54c1f5 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0fd39fd29fe44dfabdc51feb1518bad0 : SUCCESS in 5m 00s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f67080c9e2404f67943ffaf2eabd3d22 : SUCCESS in 42m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"197fca13e07bdd3802164d994aa586ee5aa5e8fd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-02 09:56:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e322ecb2e224eb21183dbae1c4a04e6a87ea9dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-02 11:04:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"714d5245d5c2917f1f6cc5baccfdc18955510b0d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-02 11:06:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2ed9fb9f8a354daa5fb62a544945350325ebcbba","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-02 11:17:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fe7ec4e590c66720a0c00811803d545fdd6c0df0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-02 11:35:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b8d5164a8db3c6c77e8b9727b62d9d21bf5952e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 12:36:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/359403af2d054650addd8157214894d9 : SUCCESS in 32m 52s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/176e391eb1cb4409aaddd21d7195a026 : SUCCESS in 35m 21s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c0b1f7adc2bf415089cf43e3527563a2 : SUCCESS in 1h 14m 36s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/df050464981a4508824caf9a15efd7a6 : SUCCESS in 1h 00m 50s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/21bbf1caa7224925a2db0703610f0e0f : SUCCESS in 1h 26m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/df7ed88866824d7680dee936992cd56e : SUCCESS in 29m 05s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ab80bf97a7494e0ca059fb915741a5b9 : FAILURE in 18m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/10a0170f31ce405a9d1b08977a7ff278 : SUCCESS in 50m 53s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/fceac871b3ea4f049b53d3cfc60fb05a : SUCCESS in 1h 15m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/819105dd9ecc4eef884cd6d34be36d38 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd3a0ff2836442cbb472d09763a63474 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f60d56de39a944dc828bc56d66219203 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b42dd6c716834654adfcf3a839f1ec76 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adc6e363fe444d35b5e267d456771a81 : SUCCESS in 7m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/022d4cc6b8354e4b98270a4ab90108c7 : SUCCESS in 19m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8b60dd0176dd482cb94df60a0b9f3497 : SUCCESS in 36m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b2768b0e44bc4febb58f80dcb25570a5 : SUCCESS in 50m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6c79706ffab34390a95a31c1c2d59fbd : SUCCESS in 54m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9cbe633179354927aeda3a5c8e5cdce8 : SUCCESS in 1h 11m 10s","accounts_in_message":[],"_revision_number":2},{"id":"8e60ef260a9ad1c27364b26233eb610692415132","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 12:38:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/07cd887148714e82a917c8065f0155da : SUCCESS in 56m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3c9d0792210f4347b16fe08d804fcf80 : SUCCESS in 53m 58s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/73d4c15b4fa04516b018848c0cfa28a2 : SUCCESS in 1h 23m 26s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3bd0eb857b9a4ca0b60a8d5c192441a3 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e3030b50d504595bfa2c1205742e6eb : SUCCESS in 5m 14s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b296783d800c484d906d2aa29c0dc749 : SUCCESS in 53m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"81d58b87e335b95849906c415e55fd654084d77f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-02 16:08:00.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"acbf05173381030c3b2c47c66f7e81b66e5dd710","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-02 16:26:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ab2a1b520c6aa83fd3493f280750fc8b7df1e35c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 16:28:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1920199eebf67e093f52c9c9cefc7871e7ea9c10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 18:05:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/86972fd89f724c28b24fcf19f27a4106 : SUCCESS in 41m 34s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/470bd8029b2148ed83d85a94425ad6b9 : SUCCESS in 1h 18m 59s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a22ee492766f410885d0e451637acb25 : SUCCESS in 1h 30m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fdf805ae1435454094bc2209776ef50a : SUCCESS in 33m 13s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7895aad8b31341b7a3a8fee5d29a5f9c : SUCCESS in 1h 01m 58s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8dcd4cca3a804990bb9b9571c496c676 : SUCCESS in 1h 12m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c245529727d240feacb4779d5229d690 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96e5d1fab81f4f21860f9c1424b040a9 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6414ca510bc24bc8877d3a9f9207dd14 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2080b949abaa475a98b0f303d8ec1ca0 : SUCCESS in 5m 39s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c7358f0ba92341e8bb5ce053aee87674 : SUCCESS in 36m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/42d86826f8a34177bf8af3b1f506e7b0 : SUCCESS in 55m 25s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/327b493eef1b45b9a3dd21d7652482a5 : SUCCESS in 1h 03m 47s","accounts_in_message":[],"_revision_number":2},{"id":"2aacd58a0080c3e90d4eb1850ba7c1a3346de3d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 18:05:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7f414e5391b6eddd69aaa555e68c205f41cedb78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 18:16:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ebc573d5d4c463690db60cbc8d06b54 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1b41f2e682f15dd5f3eb0fb09d61c35f18607bb8","revisions":{"556f304268cb7f576a261c8ddfd9c7d0a20a78fd":{"kind":"REWORK","_number":1,"created":"2021-11-01 13:29:43.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/70/816170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/816170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/816170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/816170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/816170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/816170/1"}}},"commit":{"parents":[{"commit":"896480988843dc22303e42b9170808d20829970b","subject":"Merge \"cinder-volume: Install binary python libs only in binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/896480988843dc22303e42b9170808d20829970b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-11-01 13:28:19.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-11-01 13:28:19.000000000","tz":60},"subject":"gnocchi: do not install cradox","message":"gnocchi: do not install cradox\n\nProject description says:\n\n/!DON’T USE IT, but use upstream python-rados, except you really can’t /!\n\nAlso installing it created errors which are silently ignored.\n\nChange-Id: Id24458cd655aeace0e745c512aa7890b3edda51c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/556f304268cb7f576a261c8ddfd9c7d0a20a78fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/556f304268cb7f576a261c8ddfd9c7d0a20a78fd"}]},"branch":"refs/heads/master"},"1b41f2e682f15dd5f3eb0fb09d61c35f18607bb8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-11-02 11:04:50.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/70/816170/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/816170/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/816170/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/816170/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/816170/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/816170/2"}}},"commit":{"parents":[{"commit":"eff681063adc62faef55cde0ffa6df8de1a04055","subject":"neutron-openvswitch-agent: openvswitch is already installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eff681063adc62faef55cde0ffa6df8de1a04055"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-11-01 13:28:19.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-11-02 11:04:50.000000000","tz":0},"subject":"gnocchi: do not install cradox","message":"gnocchi: do not install cradox\n\nProject description says:\n\n/!DON’T USE IT, but use upstream python-rados, except you really can’t /!\n\nAlso installing it created errors which are silently ignored.\n\nChange-Id: Id24458cd655aeace0e745c512aa7890b3edda51c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b41f2e682f15dd5f3eb0fb09d61c35f18607bb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b41f2e682f15dd5f3eb0fb09d61c35f18607bb8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
