)]}'
{"id":"openstack%2Fkolla~816175","triplet_id":"openstack%2Fkolla~master~Id578cbc49c9a432414e5c758e0c853e59f79b039","project":"openstack/kolla","branch":"master","topic":"some-debian-cleanups","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2021-11-02 12:33:53.000000000","reason":"Change was submitted"}},"hashtags":["cleanup","debuntu","trivial"],"change_id":"Id578cbc49c9a432414e5c758e0c853e59f79b039","subject":"neutron-openvswitch-agent: openvswitch is already installed","status":"MERGED","created":"2021-11-01 14:16:58.000000000","updated":"2021-11-02 12:36:35.000000000","submitted":"2021-11-02 12:33:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816175-some-debian-cleanups","meta_rev_id":"eb447918ab4cd98eb4bb8f861bc92a6aba065111","_number":816175,"virtual_id_number":816175,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-02 12:33:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-11-02 11:05:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-11-02 09:50:38.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-11-02 11:06:29.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-11-02 11:05:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-01 15:31:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-02 09:50:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-11-02 11:05:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-11-02 11:06:29.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"073176110cdf911a806025763a40b0e14ffcdec0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-01 14:16:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9099f1633ed92059c69076481e75e69b00c5a3b","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-01 14:17:28.000000000","message":"Hashtag added: trivial","accounts_in_message":[],"_revision_number":1},{"id":"d04127ff8a8e23cd3b5c6262813dc6375cb1c351","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-01 14:17:34.000000000","message":"Hashtag added: cleanup","accounts_in_message":[],"_revision_number":1},{"id":"5766673b56de02e6b8031dff9d66bdfd316131a6","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-01 14:17:42.000000000","message":"Hashtag added: debuntu","accounts_in_message":[],"_revision_number":1},{"id":"253e8590c6fd6c6c6b56089e03dd663f1ee17b41","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-01 14:17:56.000000000","message":"Topic set to some-debian-cleanups","accounts_in_message":[],"_revision_number":1},{"id":"2bc449725f2563f59fe9fd4de5ff0f7eb346464f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 15:31:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6cc20609fc2645f4a636a16f4c96b7d2 : SUCCESS in 25m 12s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6476db1afeb747d5a1352adb71ccef56 : SUCCESS in 33m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/85a5264cbfc84c60b2516a33acb3a83f : SUCCESS in 57m 15s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0083690cfcf1484a8c35188e8bb1ba89 : SUCCESS in 1h 09m 46s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ec190d2470de4ea486003df94af607d1 : SUCCESS in 1h 14m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d67686e2a551451995441b191e84a4c4 : SUCCESS in 27m 38s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2b6bb1e4bd244f728cc0619db1fdabd6 : FAILURE in 15m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a452550b70234ecaa38246ef97c0507e : SUCCESS in 45m 47s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c472135113574d0fb4cea462b6602e88 : SUCCESS in 1h 08m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f29bc70195d34ba7b2b833ddbad07334 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0798cd15c90e4f5c865ab3604bdcebf3 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c64b2b9b39d45e69b7b98996726525f : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f4de6d8308f42f2b6ea59d40048a3f6 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b4c03b5e2fb428d8c9751fbab35d186 : SUCCESS in 5m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7aaa24638fec4146a9a46030453af44d : SUCCESS in 20m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f0b1bfdebe46445abb5db84a365fe5e2 : SUCCESS in 23m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/92e53fcd09724f90a35b7a2bb4c8b252 : SUCCESS in 50m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e5cc626a6dce4c2fbb3be7a6fc0d04c0 : SUCCESS in 51m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cb43173e5be949f6a39d8ba2227cfa0e : SUCCESS in 1h 03m 09s","accounts_in_message":[],"_revision_number":1},{"id":"90ac9350d14ce688a11114fdac92439459690334","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 16:07:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6b14d11177fa4ff68bc4c07f6e2cd572 : SUCCESS in 1h 22m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dbc47b4e37784fe69d9e5b04423b1b69 : SUCCESS in 1h 37m 24s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/66dcad1c3c24442b89557fdb0f2b344e : SUCCESS in 1h 23m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/58205b5b13474593930e2567d08caa56 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c92635de78434187b92e5d9ff3dccc9a : SUCCESS in 5m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/141a0d3e0f3d4ea0bfc2e8b7b046a19a : SUCCESS in 35m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"71a6e15fbe35612946fb8f7ad530b789eab73371","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-02 09:50:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"72a6311c0b0b1554be3adf8ca2278f09aa352b22","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-02 11:05:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b9829b71a42bbcfa6275218b7acb66b53621a05f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-02 11:06:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1d287792f87c187212c1daa6e599a89aaa27c63b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 11:08:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"095448b82f069cbe6c6d663375c393f98d85ed0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 12:33:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b13f267ced714aadb2ccf9a00490cd1c : SUCCESS in 38m 43s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3ee05413440a4775a99a9ddd4087eca5 : SUCCESS in 1h 05m 12s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/cbd8f583944d4740bc4aaba63b7124d9 : SUCCESS in 1h 22m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9cee816feab44c8291b2e6b9cd685649 : SUCCESS in 28m 16s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/79c65c26075a4867b47f082754c9e25b : SUCCESS in 50m 10s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8bdb608c82fd4b04bc7c1799d91791ec : SUCCESS in 1h 05m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/769ea588858f4c678faddd5d69ed3f2f : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41ec3d71032b4f99b988bbf64272dc56 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70dd42754dba46df8f7c99e80f4c6fe9 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc46445f3e7a48fe9096139ab4341b6c : SUCCESS in 6m 29s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ee2556c272a944d29e1641ac3f26804c : SUCCESS in 28m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9e5d20c376fc4003b51b45256d7e0746 : SUCCESS in 51m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5b6cc1b921934db7b9f6fe7ab9289139 : SUCCESS in 1h 06m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6a9d27846cf532d4e60de240cde58dafa6fd2850","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 12:33:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eb447918ab4cd98eb4bb8f861bc92a6aba065111","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 12:36:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3c47a23d40a4105bd516822f554d64b : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eff681063adc62faef55cde0ffa6df8de1a04055","revisions":{"eff681063adc62faef55cde0ffa6df8de1a04055":{"kind":"REWORK","_number":1,"created":"2021-11-01 14:16:58.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/75/816175/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/816175/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/816175/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/816175/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/816175/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/816175/1"}}},"commit":{"parents":[{"commit":"896480988843dc22303e42b9170808d20829970b","subject":"Merge \"cinder-volume: Install binary python libs only in binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/896480988843dc22303e42b9170808d20829970b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-11-01 14:15:18.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-11-01 14:16:51.000000000","tz":60},"subject":"neutron-openvswitch-agent: openvswitch is already installed","message":"neutron-openvswitch-agent: openvswitch is already installed\n\nWe install openvswitch package in neutron-base image already so there is\nno need of repeating it.\n\nChange-Id: Id578cbc49c9a432414e5c758e0c853e59f79b039\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eff681063adc62faef55cde0ffa6df8de1a04055"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eff681063adc62faef55cde0ffa6df8de1a04055"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
