)]}'
{"id":"openstack%2Fkolla~818378","triplet_id":"openstack%2Fkolla~master~I62148b95e226711cbc70ab2a411774114f9dd7b1","project":"openstack/kolla","branch":"master","topic":"yoga-deprecations","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2021-12-01 21:06:19.000000000","reason":"Dr. Jens Harbott (frickler) replied on the change"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2021-12-05 16:09:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I62148b95e226711cbc70ab2a411774114f9dd7b1","subject":"build: 5s delay + note when binary images are used","status":"MERGED","created":"2021-11-18 13:00:29.000000000","updated":"2021-12-05 16:10:35.000000000","submitted":"2021-12-05 16:09:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"818378-yoga-deprecations","meta_rev_id":"4aebd0227a70dd255440e9def9e18153c47fa2b2","_number":818378,"virtual_id_number":818378,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-05 16:09:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2021-12-01 21:06:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2021-12-03 13:26:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-12-02 20:04:57.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2021-12-03 13:26:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-12-05 16:10:35.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-18 14:36:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-19 10:11:07.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2021-12-01 21:06:19.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-12-02 07:55:42.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-12-02 20:04:57.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-12-03 13:26:19.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"a3785ec811fcb94d34a8d51aa2ad571e56f5ee2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-18 13:00:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11c8578587abd05d239cdc51399b6fbf13e2e65a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 14:36:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d745664c5bca4022b7b6ad916c51cc6a : SUCCESS in 33m 02s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c0a8659abcd34c84ae50197eae779f28 : SUCCESS in 46m 01s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1f83fb4e9cf54872a682c70116b2feca : SUCCESS in 1h 09m 21s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/13888e7de79746ae990ffd8813e27305 : SUCCESS in 1h 12m 17s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5b99ffbd961f408891d0e30edb9ff117 : SUCCESS in 1h 30m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cf278b7bee0445de9210482253fc1929 : SUCCESS in 21m 56s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7adb8bddd74747aeb58105fb70809e82 : SUCCESS in 22m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/748fb05442d84e03bd6aab6fb5665335 : SUCCESS in 58m 19s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/60e7e65436f64204b4c23586261a55e5 : SUCCESS in 1h 15m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/645fdad1b88549babccfde8fadc63367 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8eaf9a369c6040a3829e31e5a731b531 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5eb099839ea34eaea6fe1de931adf345 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d79911ee03e849ecaeef47fd393cb15b : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c945fdf2ded4d2c8f5376d96f063c55 : SUCCESS in 5m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d0f5c39098124e0498b87fdabc265d15 : SUCCESS in 21m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4e27e530940d42c0a44af61b4c853ce7 : SUCCESS in 30m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3cb5eff351b04261899c30e25eca25db : SUCCESS in 51m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fc4106509e4a4b1c83dd6d0617206777 : SUCCESS in 55m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7b0533329bb14aadbeea6ef7ea2c74f8 : SUCCESS in 1h 12m 03s","accounts_in_message":[],"_revision_number":1},{"id":"4425e0a21c749bfaefcc3ac4aa2483831c52e2da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 14:42:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/35868b1a5d8e48cdbdeb8d9abcbe712d : SUCCESS in 1h 16m 49s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cc29fbf5d2244393b8279ea8eae4da31 : SUCCESS in 32m 53s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4d5179cca2d546388bf0f26a2db364bf : SUCCESS in 1h 37m 18s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d834f514785b4730a16cd33ec4e8b747 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b71a241c1aa6436aa4219fc1891a9fa1 : SUCCESS in 5m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8060d3820989475b9f3419156c807bef : SUCCESS in 32m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7462da687053b68467cb80b7dea684729cb37526","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-19 10:11:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e96ac08d11a7155fefe55a0fdc2ff4871e4a3aa","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-19 10:16:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aa12ff8f25b2aaf2e866b70f4fb3deee94d5a69b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-19 17:50:04.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e063a65770023cad5544114fad60859dc6e7aca1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 19:13:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d5945b042d3641058a6de1125634dc84 : SUCCESS in 51m 50s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/82caa868910b469282e0eb521342ea67 : SUCCESS in 44m 58s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a8bf07b24e334b16805c42b553412e25 : SUCCESS in 1h 16m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/787b8f27e41f472fab4ee43b67f29936 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df583cc848d442e183844ef7f3a5a498 : SUCCESS in 4m 32s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e217be5d28a74b88bd85efd40e12ff0a : SUCCESS in 37m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ac4d2ecc2a5c246375cacb4a79c9d3b5a04a2a33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 19:19:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7db902c8bd34487b8636ff682ae4c92f : SUCCESS in 30m 58s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a6b4056dd7174a789116bb0deacd72b2 : SUCCESS in 37m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c28cfded9f064f80823361ef8c6e5386 : SUCCESS in 1h 09m 55s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/cda3f6d3d5204c4b969f71f4e4bb7310 : SUCCESS in 1h 09m 20s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6b7b5dea0e584cb8a6da10047ed62663 : SUCCESS in 1h 24m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ccbbe350f2be485bbf5ad6d76b9151b6 : SUCCESS in 36m 17s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ce93efbe487a497faba5985106090809 : SUCCESS in 22m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/23f516291ca144e7af0b947e4613dc3a : SUCCESS in 57m 58s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/bf5cecd21bc041d4a6d9c5a329d4bcc2 : SUCCESS in 1h 14m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10e3a099530b4fcdb64572f3c976c126 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/749ee12a704f45a0b0e68f8479c59266 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aee535e2cd95410e88a107368f47aab0 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f0ab918ee99431caa19bd252ffba3bc : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a5e62566d5f41019f13254458243ae5 : SUCCESS in 7m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fecd7f83026e4ddf82b2c92d2f34287b : SUCCESS in 20m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b8dc2fa5dd7644df8cd3182ec21e61b5 : SUCCESS in 29m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/81091bde27214ac7b388a548e3e0a223 : SUCCESS in 57m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0afacae491ec43f6bd4d17c186674b55 : SUCCESS in 53m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/46f39d909e69430a8c96f08cb451f685 : SUCCESS in 1h 10m 20s","accounts_in_message":[],"_revision_number":2},{"id":"85e366442c530b6b5911fc8ff58c34cc18749322","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-01 17:27:16.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"0da2b42950d6c6212848c4698d08479b34eb5f81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-01 17:27:27.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"3f0d2bd49536efdac54990084b8942db0e26acc3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-01 17:27:51.000000000","message":"Topic set to yoga-deprecations","accounts_in_message":[],"_revision_number":3},{"id":"3d76ec39efbc8e3c38ad7150bdc0d1bc8c00dda5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 19:22:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/465d9903c7d640be9e84afd834eee8f6 : SUCCESS in 28m 27s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c7e71770486f49afa9d9c63ae67d3af4 : SUCCESS in 34m 29s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5b6d8ff0d51143b5819d62d1c2d0d978 : SUCCESS in 1h 13m 40s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7a52485d421b46e5a4de926dd13a6a3b : SUCCESS in 1h 08m 43s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/36722716730d46ce96fd42b5035de2d4 : SUCCESS in 1h 26m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d38634c592ce4fc0b5fc381a646dd0c3 : SUCCESS in 25m 54s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3d72fbd0beb5404792f9ad4ad2154490 : SUCCESS in 23m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/19e38c183bb34493961891f07f59f6a5 : SUCCESS in 49m 22s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5885ffc5c1e34277b746e365759f4e03 : SUCCESS in 1h 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15bb1952e32447a9bc3132925d23883f : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/414422dd19be40e382ac3129ea99725b : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79c1a57912314404bed1095a329b455f : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1439f388fdd47248bdbc4fea3dcff54 : SUCCESS in 6m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/68816dcf4f6240ec81e14c6e29cd0e0e : SUCCESS in 15m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2de4b4288b7e4a92b04c7de8e4cff17d : SUCCESS in 27m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a7658a6428504d698f45dc7fa62f5301 : SUCCESS in 45m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0a46495ff1b3485496a2f6eb60e0485f : SUCCESS in 52m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f5d6a50f80dd402fb0621bbb06c1b2ba : SUCCESS in 56m 52s","accounts_in_message":[],"_revision_number":3},{"id":"08d05a8e79ecb59e53c43c13c2f7326ad4728bcb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 19:44:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/dab7674d6a524d9d9ffeff68d7120462 : SUCCESS in 1h 02m 11s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/51e89f2b9307463b83080498a535271f : SUCCESS in 50m 00s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/03da67290a5641d99cdfc18619f11611 : SUCCESS in 1h 33m 30s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/daea97f1a1444aa19590016d4abef9ef : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e161d7faf0c4700b5a7df76c0607bf7 : SUCCESS in 5m 38s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/079211ce1766490eb4d6f8055f33e940 : SUCCESS in 58m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"004c384bcd68289138868c0b05a629cb93b630ee","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-01 21:06:19.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"891682b1f0517f8b9d90ead432a982d6a7bd9810","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-02 07:55:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e730aae9f5cff53998998f526ec164a65666d1a6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-02 07:57:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"af2de80674b7ce97502d9094ed51ac2e4810fe5a","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 11:07:32.000000000","message":"Patch Set 3:\n\nBuild failed (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/e12398aedfc349d781403246126a8d83 : TIMED_OUT in 3h 10m 11s\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/07693ca8645b4196b8863ec4d790518f : SUCCESS in 29m 54s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/4095d09fd584448b948bf3309463250e : SUCCESS in 28m 42s","accounts_in_message":[],"_revision_number":3},{"id":"5faba08b7c996865a62d871ecc91faa83cc9cf79","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-02 20:04:57.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3d345eddfca7bb73991f87864e9bcc12d4297a64","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-03 13:26:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"54a1310221fe890670c7f3d1992bf3d49e00d4c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 13:26:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"304d9efce75558019ef2874bee18228d79cd6f6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 15:01:27.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6b026e1bbf194227ad6ed57a018618c0 : SUCCESS in 44m 19s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/816127aad5a7496cb28e7450fa6fc808 : FAILURE in 26m 09s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/322f8ffa834643ef84448c4a783f97b9 : FAILURE in 25m 54s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/340dcdf5c67e42f6a652233a2b899d1d : FAILURE in 26m 00s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/70b121ca7a3847ee825ab0661dca3aec : FAILURE in 18m 53s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/84ce80780dc94ce28beb9e4ae8703476 : FAILURE in 21m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74d67b33bf6241c3bf7005287bddb43f : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ea379d5c0314d6e9eb4b1488fc9f173 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f06b256712e43da9ea6e1faab051fd3 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa91bd0fde4f4c2d987307915dc6c24b : SUCCESS in 6m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/574d8044d2bd4b6b87cdbf812c37390e : SUCCESS in 27m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/726003bf8efe46d4b32eeeb56e036d46 : SUCCESS in 54m 07s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0d116fe0b6a647b5b4751c0b3acddaec : SUCCESS in 1h 12m 26s","accounts_in_message":[],"_revision_number":3},{"id":"5590fe4b040640d6202513ec0f76cfc7105ac5c8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-03 22:48:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e458fb031988ac067339126c203a670d6042f77f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 23:57:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e583e834aaec447aa4c402a389f4ff77 : SUCCESS in 29m 33s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7426a70caf0c49aab8c58f61650b4492 : SUCCESS in 40m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2fb856f59fab4a1086252dccbbdf8529 : SUCCESS in 1h 02m 48s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2b9ac8486e714af39dbeb2a6b50e7925 : SUCCESS in 54m 20s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ed77bbc2cee349718d405805baebb882 : SUCCESS in 1h 07m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/86ec77e2e1c14c7282a2000295768bcc : FAILURE in 17m 54s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dda19531564e433bb5ceb5dee6b8d6f2 : FAILURE in 14m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3a08ed0d158d4feaa9da1e614aeb9496 : FAILURE in 15m 51s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d625bb4e57ed4fb4869ac26c752aa89c : FAILURE in 24m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d6c248e42484826a7859a913815e150 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d850833bbbc4b86a42f63ed5af0c26e : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d858151944ca4927b2847800c0b14468 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce9722e31ec8458dae7eabb68a4f9a05 : SUCCESS in 5m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6a2f617004ce4d909602b6e9cba49359 : SUCCESS in 16m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8699c5fea2c14a4caa04566651e2ce51 : SUCCESS in 37m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/16995b727a354ce082c308716e601ad1 : SUCCESS in 48m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f34eb7e900db4594972da4f9d778c951 : SUCCESS in 35m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/92b5447b49ea4d7db788274176e06071 : SUCCESS in 56m 55s","accounts_in_message":[],"_revision_number":3},{"id":"28d437b2c16578c419f3980ec7c49f908ae4a369","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 01:04:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c3070a2131664fb5bf8abd3a6268f369 : SUCCESS in 1h 42m 38s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5133553c91744936af2e2d6755288311 : SUCCESS in 1h 31m 15s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/051cc141100847f883ab0e0cac641e5d : SUCCESS in 2h 14m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c77aec90a97d4edd9c6b2b18222f9647 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cc3a50dd7a48486cb7ac2ac6c0124628 : SUCCESS in 5m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ee4664dba7254cec88b8ad00f5eaf05c : SUCCESS in 34m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b68d25ff2f0fd74ccf9ad62312a29aab2514d5ff","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-04 18:38:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6210e4407bd6620a4d7182792e6ad62f14b4449a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 20:21:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/959ef1a163dc43eca9dcca0409e1ef58 : SUCCESS in 1h 04m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bde83ccff8a44df4a61a412fe02671b3 : SUCCESS in 53m 37s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a4fdc9f5402a41b9bbc48ac06232afad : SUCCESS in 1h 25m 28s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd9de5758bea4d54bbeb52f1bc76c3f2 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ae54b173053f4fc5aa8d7446ff20e64f : SUCCESS in 8m 16s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9227666d32f94daea5bd6e7c1cff4bdb : SUCCESS in 1h 08m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"61ea56dd8eb04d07d1970d9c3c020e230cc05171","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 20:35:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4fcb20e0c6494494a2fa5e77db98b9a2 : SUCCESS in 29m 02s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/06866b8be46044e3b1069a2089c31828 : SUCCESS in 35m 44s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/275262188c44454f8467d315973e6407 : SUCCESS in 1h 13m 40s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/281248b38aa14ec78b6c08c1d3ec6de1 : SUCCESS in 1h 03m 32s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/17058d1f9d494cf4ac3a1d6c9e99be84 : SUCCESS in 1h 51m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1989873ca72b452f86f313f987391804 : SUCCESS in 24m 44s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/63ef9a9fec8247708f0a346d1a74867c : SUCCESS in 15m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3bc7dba4dc6c4da88f6d8d73386b6348 : SUCCESS in 56m 37s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b50d2211d8ee45978b36967b9cfd254b : SUCCESS in 1h 06m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57317695f2cf4aaeacbc8457d9999d62 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8afc040ba9b0433ebde600143a3e038d : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/635f6af09edf406c96639c00aaca8edf : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddc4fdba7d9d4c6090e0c779e5f19c44 : SUCCESS in 5m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/49e3dbf62be248328cfab62429f021a4 : SUCCESS in 17m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0aca39e5afb642cf88330d0a978863a1 : SUCCESS in 28m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/635244c9dda64e6bb38950de1b7b13ec : SUCCESS in 49m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/354d40cb2eee464089325c45f9ea8c1e : SUCCESS in 48m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/be96e65cee9546e3be657b758200f113 : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":3},{"id":"7b700f1a4b91fa09a7f02d9515f8fac9d9753552","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 20:36:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"18e7715b99c173f6fc3acbd3cead3c9ca9725904","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 23:38:13.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/262487254bb6400186546fab345144c6 : SUCCESS in 36m 35s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/42e1a4fec5b149bc9fdffd29dee9a940 : SUCCESS in 54m 53s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/56277f035ca2444b9281b5f54665b39c : TIMED_OUT in 3h 01m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9ca67d82cd894c6b9b2b618ac6237805 : SUCCESS in 24m 33s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c68d4b594fb046a789c247b6d26aebfe : SUCCESS in 1h 00m 54s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/429efeea4d334118b012e3445c7752f1 : SUCCESS in 55m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73acee9e6b4947b1ba0529f510ac5a5f : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed98262c296d4ad28254d2dd0bb15973 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7975d8cb64ca4ad58ccfe640d1b1e4d7 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7a0823fa8b34b3a882e63d70786c073 : SUCCESS in 6m 52s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/96b75d0e5f634d7293ed6f239c61bb9b : SUCCESS in 28m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7d81785a6e914eeaaa8a077819aadd1d : SUCCESS in 43m 25s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b0112479613c44d89c69314e75d3f922 : SUCCESS in 1h 05m 57s","accounts_in_message":[],"_revision_number":3},{"id":"e6a82cb9524c47b86a7938106105aff86f937e15","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-05 13:17:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a16cd68c596a1cef500c21e865883b8a588c7d00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 14:45:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ba3b9b40a6bb44f5add4acf551b3e8e8 : SUCCESS in 30m 57s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c80b0fe4c2b046ccaa43f6cfbbf4444f : SUCCESS in 37m 06s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6ce8610011374d778a6fc3cffca93963 : SUCCESS in 1h 18m 41s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7fe1dd0904f14cdc81cc21b86dd392b8 : SUCCESS in 1h 09m 46s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f51cc1bf4d7e450ab1ce32426ea15a80 : SUCCESS in 1h 23m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b448af13d4fb4d42b1b48a2c15854ccd : SUCCESS in 23m 24s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a485e90fb23c40bfa9c52a46d33c26d3 : SUCCESS in 13m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d82ebba5c35e4d5a94fd893a85782950 : SUCCESS in 44m 43s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/01184049b80d4cadb446b8f4c58c3c0e : SUCCESS in 54m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/410b95eb7a794e2386e26e24409f8cb6 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63f9c520f5b74bba9bb4f754a0d134fa : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbca1b7cbcb9434ab39283bada2dcf9b : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7155b02f884945709be766c6e1942835 : SUCCESS in 5m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/96d30107eace4475866dc06729672a0a : SUCCESS in 14m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/daa98bc57b1b482aaf45ce50d42034aa : SUCCESS in 35m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bfcf9cb72d1444ccb77aef6d18598f28 : SUCCESS in 54m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/baa0449f8c974d52bcd6ba2d51f6c259 : SUCCESS in 39m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2a1fcfa0e3194343b4ee304237972cd7 : SUCCESS in 47m 12s","accounts_in_message":[],"_revision_number":3},{"id":"ff10c6fe5565202b21b8bfbff3fa742dcc542fbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 14:45:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7d51621c7bf4c40537032b53bdcd336fdd4df1bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 14:51:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1d1aea5e649844849be27cf282f839dc : SUCCESS in 46m 55s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0c52cb16a1b24f15b75ea386444428e0 : SUCCESS in 38m 26s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/266224b844f34b9d9a41ac4fd1c36006 : SUCCESS in 1h 27m 42s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f095a39d1c984afbb7dd760f069adb1d : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1c8850bb3ac94d5c9b326d7adcdf189d : SUCCESS in 6m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/fef7ab5d852248cd9121efd829a324e0 : SUCCESS in 39m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"55195542e75d6b26c01d4266ea3a63dae1ceb099","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 16:09:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b0da9e683a764f2c9de8ef9b571968dd : SUCCESS in 42m 05s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1b62dfd3ff664b40854674d36e8bf402 : SUCCESS in 1h 16m 46s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8ce7687633f44a00a0f7ff348ed51ca0 : SUCCESS in 1h 19m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/66df87bdc5474e1aa60589c7aed10c4d : SUCCESS in 27m 30s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3a0996a57f6e43ee96ab71e5ffb0972a : SUCCESS in 59m 39s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d4e7adc4411e45f490e447bff5717777 : SUCCESS in 1h 08m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff648857ba5142b49061ed2e674115c8 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/782579b910b5495688c17a57778de8c4 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ce057d1f6b644b2bff0945999fa3f26 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2217f8943dc4796b39b9158746ac78a : SUCCESS in 7m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b5acf0c168b84d669315bb04bbf1c753 : SUCCESS in 27m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1279cace87ba4e5aaa54ec0c04d792f2 : SUCCESS in 1h 06m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7f86f95bfe604f1ca80ed49540213441 : SUCCESS in 1h 06m 39s","accounts_in_message":[],"_revision_number":3},{"id":"e8ca2dc3928004285cd795ea825678ff5725fcf4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 16:09:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f0ceb5e5ea7227581ab9ef6fef86cc3eaa48f8c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 16:10:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72b404b510f14c039942b099f2e73431 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3239812c3d92c60f241477b1d7f0232324d424f7","revisions":{"9655770cfa42037c8aabd542e691bb5fd24800cc":{"kind":"REWORK","_number":1,"created":"2021-11-18 13:00:29.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/78/818378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/818378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/818378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/818378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/818378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/818378/1"}}},"commit":{"parents":[{"commit":"8794afa8c61e01c2821e8309bedb6c17259aa0c1","subject":"document that binary images are now deprecated","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8794afa8c61e01c2821e8309bedb6c17259aa0c1"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-11-18 12:58:27.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-11-18 12:58:27.000000000","tz":60},"subject":"build: 5s delay + note when binary images are used","message":"build: 5s delay + note when binary images are used\n\nWe deprecate binary images during Yoga cycle and drop in Z one.\n\nReminding users on each run to make sure they notice.\n\nChange-Id: I62148b95e226711cbc70ab2a411774114f9dd7b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9655770cfa42037c8aabd542e691bb5fd24800cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9655770cfa42037c8aabd542e691bb5fd24800cc"}]},"branch":"refs/heads/master"},"9c75a78fde0baea3b5c72fc1fb4d717b7c02e321":{"kind":"REWORK","_number":2,"created":"2021-11-19 17:50:04.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/78/818378/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/818378/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/818378/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/818378/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/818378/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/818378/2"}}},"commit":{"parents":[{"commit":"8794afa8c61e01c2821e8309bedb6c17259aa0c1","subject":"document that binary images are now deprecated","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8794afa8c61e01c2821e8309bedb6c17259aa0c1"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-11-18 12:58:27.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-11-19 17:50:04.000000000","tz":0},"subject":"build: 5s delay + note when binary images are used","message":"build: 5s delay + note when binary images are used\n\nWe deprecate binary images during Yoga cycle and drop in Z one.\n\nReminding users on each run to make sure they notice.\n\nChange-Id: I62148b95e226711cbc70ab2a411774114f9dd7b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c75a78fde0baea3b5c72fc1fb4d717b7c02e321"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c75a78fde0baea3b5c72fc1fb4d717b7c02e321"}]},"branch":"refs/heads/master"},"3239812c3d92c60f241477b1d7f0232324d424f7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-12-01 17:27:27.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/78/818378/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/818378/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/818378/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/818378/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/818378/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/818378/3"}}},"commit":{"parents":[{"commit":"dc683fa1ccba2eacf7af9be6c9cb520943027a33","subject":"document that binary images are now deprecated","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc683fa1ccba2eacf7af9be6c9cb520943027a33"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-11-18 12:58:27.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-12-01 17:27:27.000000000","tz":0},"subject":"build: 5s delay + note when binary images are used","message":"build: 5s delay + note when binary images are used\n\nWe deprecate binary images during Yoga cycle and drop in Z one.\n\nReminding users on each run to make sure they notice.\n\nChange-Id: I62148b95e226711cbc70ab2a411774114f9dd7b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3239812c3d92c60f241477b1d7f0232324d424f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3239812c3d92c60f241477b1d7f0232324d424f7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
