)]}'
{"id":"openstack%2Fkolla~818893","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~Id3f824b741a5aa98efc7162fb7d49657e86a2bee","project":"openstack/kolla","branch":"stable/wallaby","topic":"add-swift-lock-path-and-set-permissions-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"30609":{"account":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"last_update":"2021-12-04 15:29:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id3f824b741a5aa98efc7162fb7d49657e86a2bee","subject":"Add Swift lock path in Swift containers","status":"MERGED","created":"2021-11-23 13:14:30.000000000","updated":"2021-12-04 15:31:19.000000000","submitted":"2021-12-04 15:29:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818893-add-swift-lock-path-and-set-permissions-stable/xena-stable/wallaby","cherry_pick_of_change":818892,"cherry_pick_of_patch_set":1,"meta_rev_id":"a873beba000ef739f848857a5c5c86f2b60332d7","_number":818893,"virtual_id_number":818893,"owner":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-04 15:29:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-11-24 09:01:36.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-12-04 14:00:09.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-12-04 14:00:09.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-23 14:48:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-24 09:01:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-12-04 14:00:09.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"2a5ea7bb60ba6820e794c1acf82309f0dfb4d4a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2021-11-23 13:14:30.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"a2b86a3f886b90b8fe2130bd30f8f7c5915f8710","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 14:48:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/36296c62e6ec4940b10a0616b7c115a2 : SUCCESS in 42m 22s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/214de1f2bde04297a1e92f819acfeb5c : SUCCESS in 55m 16s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8d8abb3efd904d9ab0a481fcfbe2e150 : SUCCESS in 1h 19m 43s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/223854fddb354ce1ba1e3709bb8680d4 : SUCCESS in 1h 24m 01s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/34b9acb7898d4485b01af724469fc12a : SUCCESS in 1h 24m 23s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/05db9ec9df1b4b688e2a0f8dc8b464f3 : SUCCESS in 42m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6b4a3fb35f5f4cc38979b9ee7e25854e : SUCCESS in 20m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f2bdbad8e3b0431b92c4776924289f5c : SUCCESS in 58m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7805c6f6a5a544ce921732dc76c6d2a8 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e57395e2f824440a9f775521893b779c : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f65f6c16d69493d9163f84503c87a9e : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c793e288c608480ebad01f0d46f671b3 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/919c535e04224279aed87b73d0828ccf : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f985d83004ff4a45a2510cf81ac3e077 : SUCCESS in 5m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fa7fdb95e753469dbd9648330f3cb3d2 : SUCCESS in 22m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/933c046f6f004136b8ec39ffd4c1dfaf : SUCCESS in 35m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/df6f356619be4ee4b7c4b301621f5614 : SUCCESS in 51m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1cf852863fd54fe58e82d8e7ba70006c : SUCCESS in 51m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f08bf6eafc594daf8173422444d13d09 : FAILURE in 1h 06m 03s","accounts_in_message":[],"_revision_number":1},{"id":"9abf56040a29e99e3c5eeef36d638a77f693b917","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2021-11-23 14:50:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"11f56cdd7ed7c249200bbd1c074125346ec1b36b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 15:53:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c9f3aade060a48a7b97de22e60fdc82f : SUCCESS in 1h 29m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4c37842db7e04106a6a67213c782d460 : SUCCESS in 2h 08m 08s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/75d6485fd05c49b39af0b76a93f4fb44 : SUCCESS in 2h 19m 41s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a9f228b47cbe4421ac1a4fa209fe9502 : SUCCESS in 1h 11m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a81c5892e25c9e739229425334bdada4832f6f72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 16:23:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3dcece94129845a2b8b875df0b859ac5 : SUCCESS in 22m 41s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f644edf942b54c4ab893148abb102762 : SUCCESS in 1h 00m 49s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/09b3f5efd833469da75794eb11a61dd1 : SUCCESS in 1h 14m 50s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b1c446b953c444aca7f15463591c49e9 : SUCCESS in 1h 19m 27s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e560f5248f6a4ffe87b13152d3ac60cf : SUCCESS in 1h 05m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e85ac093b9154fb4b822b65f70b45874 : SUCCESS in 34m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ec0bade6555e4b1eb565a49d21344290 : SUCCESS in 20m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/27f6116fcd0c49e38922fad31eeb4835 : SUCCESS in 58m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4bcd416902543bfa4dbd059c63fa2d5 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a5ab0e128e34e55883aba5947c5cd97 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2eb86c325580433c99d85cc96bc40150 : SUCCESS in 7m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/901b68e106904c228cde4a523300448a : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9e1b94b177d4eaab131643c68afe82a : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6096566d7b8644d2abbc5559629d99a5 : SUCCESS in 5m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/88ce879af4ea432b9613aa7455b1bbd0 : SUCCESS in 20m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f6cd49243cbe4ccab0635fd600bf842a : SUCCESS in 41m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d7d23a1ffdc74547b817dea6ed21663b : SUCCESS in 1h 08m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4df3aa24c1df4f99bed8d4a6f176f1c5 : SUCCESS in 50m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/323db9c590784a3bb988fa7c75918540 : SUCCESS in 56m 28s","accounts_in_message":[],"_revision_number":1},{"id":"0472e69a44c16838c5e5cc3399e699cfeb1b30c5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-24 09:01:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4ed2239916faf81bb2ef7c73781d26abb3795668","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-04 14:00:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b14ddb3b691e4708a6e3735331b4cc80452d5798","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 14:00:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01b736ec9cb9042418b92b2d61ac63b4f930eb9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 15:29:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d094c147496e4bd6ab16aeaf638ed224 : SUCCESS in 45m 23s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7bc710b1f7b347858a56a4017bea3d3c : SUCCESS in 1h 14m 19s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/79ea6110818e4df295d54d5f51053a9d : SUCCESS in 1h 25m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8b2b97f6c4f49db9e471e1b059f46b8 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/829da7c22be94c178bd52a6af6e62a48 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09d0b34cdbe542d091306392804f19fa : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/752a085abcbd4e0eab1efa01cd81a853 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c21de279a83443c39a5a98cb977754dc : SUCCESS in 3m 55s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1af4f0a2f3fc4dc6aa5685317c6f88da : SUCCESS in 32m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/23895c2965fa4364b1d29a68bc4c14b2 : SUCCESS in 54m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/26c3e2f66ecf44a1ad937d6669d15185 : SUCCESS in 1h 02m 37s","accounts_in_message":[],"_revision_number":1},{"id":"db3dd6fa922287db487aebc157648b168e9977d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 15:29:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a873beba000ef739f848857a5c5c86f2b60332d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 15:31:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe583413e6724172b7e005441ed03d62 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/938daf88422c43bfb9bafa00d8856d9f : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f81e7939fdff00cb738a49d7aa83ba1445211c8a","revisions":{"f81e7939fdff00cb738a49d7aa83ba1445211c8a":{"kind":"REWORK","_number":1,"created":"2021-11-23 13:14:30.000000000","uploader":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"ref":"refs/changes/93/818893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/818893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/818893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/818893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/818893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/818893/1"}}},"commit":{"parents":[{"commit":"6448d3d21e13f1c70ceb20acb89583c58d594e52","subject":"cinder-volume/ubuntu: add lsscsi and nvme","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6448d3d21e13f1c70ceb20acb89583c58d594e52"}]}],"author":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-04-26 20:06:58.000000000","tz":-240},"committer":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-11-23 13:14:30.000000000","tz":0},"subject":"Add Swift lock path in Swift containers","message":"Add Swift lock path in Swift containers\n\nSwift-recon-cron requires rw access to the lock path\nspecified in in object-server.conf file. Currently it\ndoesn\u0027t exist in Swift containers\n\nCloses-Bug: #1926203\nChange-Id: Id3f824b741a5aa98efc7162fb7d49657e86a2bee\n(cherry picked from commit ff25b500d23155485dc57d10aed9070a4cba88b4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f81e7939fdff00cb738a49d7aa83ba1445211c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f81e7939fdff00cb738a49d7aa83ba1445211c8a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
