)]}'
{"id":"openstack%2Fkolla~818894","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~Id3f824b741a5aa98efc7162fb7d49657e86a2bee","project":"openstack/kolla","branch":"stable/victoria","topic":"add-swift-lock-path-and-set-permissions-stable/xena-stable/wallaby-stable/victoria","attention_set":{},"removed_from_attention_set":{"30609":{"account":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"last_update":"2021-12-04 15:31:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id3f824b741a5aa98efc7162fb7d49657e86a2bee","subject":"Add Swift lock path in Swift containers","status":"MERGED","created":"2021-11-23 13:14:44.000000000","updated":"2021-12-04 15:33:10.000000000","submitted":"2021-12-04 15:31:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818894-add-swift-lock-path-and-set-permissions-stable/xena-stable/wallaby-stable/victoria","cherry_pick_of_change":818893,"cherry_pick_of_patch_set":1,"meta_rev_id":"d599a9d2f9897b94044b69c9b113c8edd9a357a1","_number":818894,"virtual_id_number":818894,"owner":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-04 15:31:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-11-24 09:01:45.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-12-04 14:00:16.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-12-04 14:00:16.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-23 14:29:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-24 09:01:45.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-12-04 14:00:16.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"271c5545116a122918ce78e1db85deb5e77b67b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2021-11-23 13:14:44.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"5175a6ca7a1253e57483de31538f98cd59d1320d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 14:29:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8a3aa74b1092480bb464169e80429efc : SUCCESS in 23m 20s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2b4bb313f0ef472da2c652c6b3cb7529 : SUCCESS in 33m 06s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/977e6ab4874e491baaf166d481651f12 : SUCCESS in 29m 40s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/129e2281e9bf4df8a20561ad8a88a487 : SUCCESS in 29m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a4e752184e7a43a9bdfc4d243ee0ab1b : SUCCESS in 1h 00m 11s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c1d4b63fffe341ccb49d818c114991c3 : SUCCESS in 58m 50s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6d82441699934d85bf6391b4886c6bfd : SUCCESS in 1h 07m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fa1c8a5e9af64eb3b9cde3b3391c483a : SUCCESS in 1h 09m 11s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f9f71751d8584635a26d25361f2ecbd2 : SUCCESS in 58m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3c64cf06f1264800a9f62d777dbaa164 : SUCCESS in 1h 08m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/12739c98f950469db62ba65318ee12a3 : SUCCESS in 20m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/55f81d1b52794c12adf59bbcb93733ef : SUCCESS in 19m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c96f2d4a7f3740dcafeb019a22cadda9 : SUCCESS in 1h 01m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58ab8ef1994e411bb686126082799a7f : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/642a8077d99b48d98310e6900945eef6 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c02cdc16b5444b984a47d6345b0c306 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0531f2036f4c9a9e26f3217028954b : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92b9723e5a354fea85c8cb3918cdea22 : SUCCESS in 4m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a04a54aff08543679635b9836881abc3 : SUCCESS in 25m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7c3c18c6e6354d2a9c425eb471f9cc6f : SUCCESS in 28m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ea4b36d1adc047fc98e9f8012b0d34c2 : SUCCESS in 1h 05m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bba75e6db405457995890dd99916704b : SUCCESS in 51m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6f720a9295ba477b9e61ec06f723b384 : SUCCESS in 59m 08s","accounts_in_message":[],"_revision_number":1},{"id":"75e1ea8791ff228bf572626ae0940a3b261fabaf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 15:57:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/802fdee882c542a99531d2532e515fe4 : SUCCESS in 2h 19m 18s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/68d7c18cee48447a8d88331d24acbfe3 : SUCCESS in 1h 16m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/05be987b3e1b47a4b3f42df99f573593 : SUCCESS in 1h 30m 02s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5ecd45b392a6417482aa20a3dae1e627 : SUCCESS in 2h 08m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/550034f1011d49febd60d020f7bc8fd1 : SUCCESS in 1h 03m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"222617f92a4e7adc557f7fe945c06a4356fe9687","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-24 09:01:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9bc749dede5918fa63d32e7f32082d6622d4154c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-04 14:00:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b5ecfcec5197e2803714e3582b2226115e275b57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 14:00:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"371a3e07392789f7ef7646a59265f30765637414","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 15:31:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/efe36c98449b45709bac7dc9dead7767 : SUCCESS in 40m 30s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7f552a74967d4edb9d25241a493c1535 : SUCCESS in 42m 25s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4943651cf32f4b43b59e21f0e0e888ec : SUCCESS in 1h 25m 59s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a8d0a6bd2fa148c4bf4d4799244385d8 : SUCCESS in 1h 09m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b5f3b90fb3654a168a549f2a023b7b43 : SUCCESS in 1h 02m 24s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/17b53aab8d4e4201ba09306131c695fb : SUCCESS in 1h 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed60dfa7bc2c423db07a4fbcc1be8139 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99decb9130924e51b72d53d4ee442390 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c82351d9d1e4de08d224c826f53b739 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f265ee2587d424aad9421b146f35475 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14d573dfb7c2445691ab32f557541595 : SUCCESS in 3m 49s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/04ea64155ccd4181881c1c10beed301a : SUCCESS in 27m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/523dca5774df4dfb8c31d47619d99d66 : SUCCESS in 46m 27s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fc19b39c5aef480da47ee68eac40c319 : SUCCESS in 59m 38s","accounts_in_message":[],"_revision_number":1},{"id":"99b6f5f88768acb6f650fbc2045e74383f97a30e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 15:31:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d599a9d2f9897b94044b69c9b113c8edd9a357a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 15:33:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a810e0c84201474e9a905f21d77dbefa : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bb1e4d2a8234a5197767d54ef90c78f : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0bc8f2a335b28f9aabc796f6ebddaa8424b6fb61","revisions":{"0bc8f2a335b28f9aabc796f6ebddaa8424b6fb61":{"kind":"REWORK","_number":1,"created":"2021-11-23 13:14:44.000000000","uploader":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"ref":"refs/changes/94/818894/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/818894/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/818894/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/818894/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/818894/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/818894/1"}}},"commit":{"parents":[{"commit":"49a5b7e55381b4b1ef7aa0bb23178f13a924ee62","subject":"cinder-volume/ubuntu: add lsscsi and nvme","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49a5b7e55381b4b1ef7aa0bb23178f13a924ee62"}]}],"author":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-04-26 20:06:58.000000000","tz":-240},"committer":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-11-23 13:14:44.000000000","tz":0},"subject":"Add Swift lock path in Swift containers","message":"Add Swift lock path in Swift containers\n\nSwift-recon-cron requires rw access to the lock path\nspecified in in object-server.conf file. Currently it\ndoesn\u0027t exist in Swift containers\n\nCloses-Bug: #1926203\nChange-Id: Id3f824b741a5aa98efc7162fb7d49657e86a2bee\n(cherry picked from commit ff25b500d23155485dc57d10aed9070a4cba88b4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0bc8f2a335b28f9aabc796f6ebddaa8424b6fb61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0bc8f2a335b28f9aabc796f6ebddaa8424b6fb61"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
