)]}'
{"id":"openstack%2Fkolla~818895","triplet_id":"openstack%2Fkolla~stable%2Fussuri~Id3f824b741a5aa98efc7162fb7d49657e86a2bee","project":"openstack/kolla","branch":"stable/ussuri","topic":"add-swift-lock-path-and-set-permissions-stable/xena-stable/wallaby-stable/victoria-stable/ussuri","attention_set":{},"removed_from_attention_set":{"30609":{"account":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"last_update":"2021-12-20 11:33:18.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2021-12-20 11:33:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id3f824b741a5aa98efc7162fb7d49657e86a2bee","subject":"Add Swift lock path in Swift containers","status":"MERGED","created":"2021-11-23 13:14:58.000000000","updated":"2021-12-20 11:35:56.000000000","submitted":"2021-12-20 11:33:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818895-add-swift-lock-path-and-set-permissions-stable/xena-stable/wallaby-stable/victoria-stable/ussuri","cherry_pick_of_change":818894,"cherry_pick_of_patch_set":1,"meta_rev_id":"6eea1a3a22b77eeafc94fe7add365aa76ae3f759","_number":818895,"virtual_id_number":818895,"owner":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-20 11:33:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2021-12-16 09:07:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2021-12-15 16:46:30.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-12-20 09:59:28.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-12-20 09:59:28.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-23 14:31:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-07 10:58:21.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2021-12-15 16:46:30.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-12-16 09:07:45.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2021-12-20 09:59:28.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"a004c3b9cd633ed1dfeb129425c1e86a8d41614a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2021-11-23 13:14:58.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"bb761c9af50789c61c9b64c245b18cd44480e271","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 14:31:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/567a34f0c1a84ade93fbece6cfe62277 : FAILURE in 19m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/40ad16618368427a868fe9d096886cce : FAILURE in 30m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5d9475051d7e4d2385fac07365b556ee : SUCCESS in 1h 03m 34s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3c79e7a367194489bfad69b9e64da48b : SUCCESS in 1h 04m 38s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/46956ec0423949658c915c5690e4b0e0 : SUCCESS in 1h 11m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3d088deffc3c4b009ecf07292237adc5 : SUCCESS in 26m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/addaa3780ef147f2a9918f87d6c72e3c : SUCCESS in 21m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/65c51ca163194f7ebf8f75f8c40d9517 : SUCCESS in 1h 07m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e275992f57334a4483ad73ed1f15cf54 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1f185ecbdc34f38a9a17b855b52043e : SUCCESS in 6m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d5f9a6816f0498596ed3763caade2fd : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea7022ba1de84fce974a4c8a268c93f8 : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25089ef0c46f4080ad196d92fe31fd1a : SUCCESS in 5m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/74e6d71e078b482c83f2f88c6cbc4957 : SUCCESS in 23m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/07c85539c64a44a18b2e19d2e741acab : SUCCESS in 30m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cf603452a4ab4db9a8c2c96110d26ffe : SUCCESS in 53m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/71b3cb297ac6447b88f681f903ba8eb9 : SUCCESS in 54m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/db45655abbed456dbd40f9451445dbee : SUCCESS in 56m 51s","accounts_in_message":[],"_revision_number":1},{"id":"5085fcd2c8d87b14164f1859f96b5bb230195023","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 16:29:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/7f9822269d744750850c906a29119745 : SUCCESS in 2h 34m 17s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cbc4a92410f045ef856398ffc18b6d22 : SUCCESS in 1h 30m 59s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1d6a4c7009284c81b021280a5256fe92 : SUCCESS in 2h 16m 46s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c417a3e1f9a247c49992051b1c117b66 : SUCCESS in 52m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0de79bf06ddd8131194326ffc80bb6a5392b6231","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2021-11-23 17:46:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf32fb863789f71011dffab7bc3f27b6521aa184","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 19:06:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/fca6228310fa45fea3aac5320859213d : FAILURE in 27m 41s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6c80cb4cce6a4137a2fcc19b159b5b72 : FAILURE in 31m 28s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bf8c8e992bd24789ba7c21343f0899dd : SUCCESS in 1h 06m 55s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0fac7be30a48495aa89cce623b383646 : SUCCESS in 1h 01m 43s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/dbc6050aa2da4045a2d27d05c19dd4ec : SUCCESS in 1h 15m 31s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ea1c53e91b7b40a48610530fce491f5c : SUCCESS in 27m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d2e1fc7db4004e1f9ee2b8ad08da469e : SUCCESS in 19m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/686e43c288b24ccfbd8c73f52a5deef7 : SUCCESS in 48m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1042f4ac2dd94918b5943928ac03b36c : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47a02a9b03fe4e56afc079735d7e18ec : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/82caf1beed0f48158c1032efb33e07c9 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3633344e8b154791bbeb59481ee5687c : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a32935ec0b5a449a9dacab613eb37c6e : SUCCESS in 5m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3bf37b756121401fb199f3782c182aaa : SUCCESS in 28m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc3d2e75bd9c4f9abad87eb63dd2e0a9 : SUCCESS in 26m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/17837cc4418240929b0254b202221847 : SUCCESS in 45m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/280e62dfbbe44cbe9d198d2011572997 : SUCCESS in 54m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/59d9c0465c004aadafcbdb3abaf51d93 : SUCCESS in 1h 05m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a80a32033e00028b9f53df2da0b92b693a45dfef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 19:13:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/bf7a9528a2994b8e8218f29cf8f60194 : SUCCESS in 48m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8ac27c8809e648978fdcfc36bfdd087c : SUCCESS in 39m 43s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/86ee2a5d5da446f282886f796a1bf50f : SUCCESS in 1h 20m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1cbadd33a3ee43d092911c0d8307e279 : SUCCESS in 39m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bde4724e3fefd75ae4084c369d1f54de4a9de416","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2021-11-23 21:16:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"296f4d9b125703d1d1ff5a64a12b8f74af96182e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 22:37:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7fdac29934d741cf9b7e23fbdefaedfb : FAILURE in 24m 27s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2692b01cb2a44633b5c32d220a3095e3 : FAILURE in 29m 17s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/065eb411d92d4b258d0f82ddcf2beb32 : SUCCESS in 1h 03m 42s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/df6ae9760d1b4738aa42b6745026dab5 : SUCCESS in 56m 56s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/4e9dc76e50fb450daac3c090bafc87d2 : SUCCESS in 1h 14m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b4497f5f7c464c229c9f38e3840bc1ac : SUCCESS in 29m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a9cda895512a434db142146020bbd4df : SUCCESS in 27m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8542fab6f5fd47c4b9c670519385b5cd : SUCCESS in 1h 06m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8f5a30e83a14936a7c1df4b151117d5 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/929611cc66cc4ceca73a1471c65e778a : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9808b31119b04eb6aed43046c6cb35e2 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cd48cd2d6a14f89948f2e9d38f73aef : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b946d8abbe224ccc88118149c60af312 : SUCCESS in 5m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7cb49387b5e24893a33ef1cf8638d14e : SUCCESS in 24m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d85e6f27fef54c85a4831082b8dba310 : SUCCESS in 28m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/06d5a1a98fa04ec5bc84618c227db7f8 : SUCCESS in 57m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7daa97576cae4fceb22dea8f25445673 : SUCCESS in 51m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/20d2b13d717f46e09f63942de7df7e83 : SUCCESS in 1h 06m 00s","accounts_in_message":[],"_revision_number":1},{"id":"81285add05c6367fdae9589203f7a336910985cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 00:43:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/969fccec6cef47d28de89363f8585856 : SUCCESS in 38m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5886d0547cd045dd8d354187fe5ab986 : SUCCESS in 1h 07m 51s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2d37fb19703a465abda53307ce0c7683 : SUCCESS in 1h 52m 07s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/97851c28e94d44dbb62ac4e99c4d7a56 : SUCCESS in 1h 05m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65ed39cf08291e6d6cd823a33e144475ae38d176","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2021-11-24 14:19:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"85757570288c2a808f5da2537e3d54a25881a8d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 15:42:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3bedf70d4e384185880dd9250ec35019 : FAILURE in 19m 23s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/45ff4c2e9dcf4a89b24c7d969aca7d4a : FAILURE in 31m 13s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f95fbd934a8343a186738fa68da69b4b : SUCCESS in 1h 01m 33s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ad7dbdeb227d41f2a9a4f859e33f1fe3 : SUCCESS in 1h 00m 28s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ac26ff3cd12d4ccf83d122410e8546f4 : SUCCESS in 1h 16m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6bec1f6e4060420795490b08698715b0 : SUCCESS in 22m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a50d6197fcb44083bc92ea3599d9cb8e : SUCCESS in 18m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3e891b7d14c0454ba42674de7df3150b : SUCCESS in 59m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e3238a0c3c3430fa3775c8d2e7cfa76 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4bb8694b6ee4754b438c382c7ae0084 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/227c9918afd84563b95fe106dffd1fd7 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f30aaaaa0de849daba90453c00f8f6b9 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/724007da05b442ba819739237a466166 : SUCCESS in 4m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4ad49869ebc24c9ea4ece8aad409d12e : SUCCESS in 22m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d5716f600b9a4f67b5ff2f9924b92d78 : SUCCESS in 25m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc26e59cd31f438d8c5b8dd11515e2f2 : SUCCESS in 53m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0b1ddd07b6ce4a0fb09b2d051da826b9 : SUCCESS in 48m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/95470f65db4d4ce085aa1f896bc9a168 : SUCCESS in 1h 07m 23s","accounts_in_message":[],"_revision_number":1},{"id":"9104fa831c0d35d7d20ec1b2a0df34a6c9768c24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 16:18:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/64d53666c73c492ba79b126d69c5243e : SUCCESS in 1h 37m 09s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/74d9b11f4aaf43d799c91bb40c910b85 : SUCCESS in 46m 13s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1bd6f7029b244a33a38d1dbed73d7c84 : SUCCESS in 1h 56m 58s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/bb3a837af9284be9a17c7dc6031f81d1 : SUCCESS in 44m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32ed8167701d654ec26a92bb2b51cc225bcc2610","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-07 10:58:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a831a8b8bcbd535ab7c9f5b76bbb9178db3f291","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 13:08:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/76be38764ddc4f71940908ab43cbc7fd : FAILURE in 45m 31s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/34c166f394b544548f701d60999e59e2 : FAILURE in 46m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bc9debb19f7b43f88785d92143bf2855 : SUCCESS in 1h 06m 05s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/97b329e38c9e43c2913822c7bcc3171a : SUCCESS in 1h 02m 55s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/edf669793378432f84bf35e8c459735c : SUCCESS in 1h 13m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6d793d35954349469140afeb2d49242a : SUCCESS in 23m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/83e0966a0c2745979fe923ee71e36e3c : SUCCESS in 19m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b5b146194bb44893a103c75e4b30a2dc : SUCCESS in 58m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51fcd7fdd3f84eac87163e2c60abc806 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f29f33ce35c845a39f7f28c1fcec9aed : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dfd2fb3f62d944988e0d4497cfdf88f9 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48bf11508e6c4f31b3ccd9bf4e207b42 : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a400bdaa2c34bb69a4f40f42f0b7b15 : SUCCESS in 4m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/86f24ac649d4496187c4000c7313616b : SUCCESS in 30m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d599305f91bb44b1b82eb7b9de266b07 : SUCCESS in 29m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ad77ccdc75ac492f8057da184d6ab332 : SUCCESS in 1h 04m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7cd608230e494a1bb5a431d0021d5bd2 : SUCCESS in 54m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8c8d24e50eba4b49994f26966706a52e : SUCCESS in 1h 08m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c9d526f06a774aee1d754641a97b857bb7074749","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 21:24:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/d7a8fcb91ebd407ea1a5975b2b9e8202 : SUCCESS in 2h 02m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/18fea56e9677452db207c79343098bca : SUCCESS in 2h 11m 32s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a61f13770447420a9b95f76fefa95256 : TIMED_OUT in 3h 03m 47s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b0be4b56627c41e78f3ae4e88b187cf2 : FAILURE in 36m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"059e0b6b590fa9d744baf2187ad6928557f3a601","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-14 08:22:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ddaba3b8a56fb3d1cdc9f0ddcc7b0ad4fb5b4328","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 09:32:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f9f28c56bdb64417b0343ee726282e91 : SUCCESS in 26m 15s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6be864f44ed64381a3b418a4cbd2a172 : SUCCESS in 29m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e29b9580cdee470ea24a71ad0ad27dd8 : SUCCESS in 58m 43s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f9d43ea70e944b2c9d962ba59ca98d8c : SUCCESS in 57m 33s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/45bf9653154943b895d39e90fcd8e85b : SUCCESS in 1h 05m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/14dcca7fcdd04b22a1b213affdd7ad73 : SUCCESS in 22m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/93012c3db9a0482a8497bb526e479cc7 : SUCCESS in 19m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ce057d01d08441cd992b90a7bce667be : SUCCESS in 54m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2acd8fb3681248d6a0254335d88c242e : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4ee475f80ff468eb60dfe6e9338b32f : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf6f1f33a47b4a69b41cc4149c304962 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53ffa577aee246299d4716b85bcbbac4 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/378cf7fa9c834b56a31dcceafb643bfb : SUCCESS in 4m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5f2eb779c6f64daf80a5443f8b5b81fe : SUCCESS in 22m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a443d2fa891f4596a03305beb081b357 : SUCCESS in 25m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/38f8815b04d94296b1f5a59f90fe8da1 : SUCCESS in 57m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8a12eb1f155d4da68e17e1a18e2032e9 : SUCCESS in 50m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ec726e661dd04b4db9cd3a4b6a90d123 : SUCCESS in 1h 00m 36s","accounts_in_message":[],"_revision_number":2},{"id":"ff7da5baf60e1bec955acf8d0ba47eb273eac5b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 09:37:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/4ea06f9617d749208d6d5e5dc17a9e9d : SUCCESS in 51m 21s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bf41eb4187704f499f98beaab0cf4158 : SUCCESS in 44m 22s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/89ca0505c0064a78a74d338c2f8f0705 : FAILURE in 1h 08m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0308c0c2be8c41868e31db2768822ea6 : SUCCESS in 46m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"55991a78f1a9e51bc97d6aa8ebf112610b06cf46","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-15 16:46:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7b8efc7528ec17df60eb08d631097b581a5110c7","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2021-12-16 09:07:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2a2c404e8708b10a3c810e11c9d1563d45295074","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-20 09:59:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f1c865b66715ad0e38042c634b2c1b09c9c9205b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 10:00:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7c1380527b59d16580f97fad9f84b093d81e2736","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 11:33:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/25e62c35859b4a2d98e2cc767ad21d75 : SUCCESS in 28m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e607accd785e4b5b9f12621066dc981d : SUCCESS in 54m 49s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/30c6ee1dc1cd4f3d954707bc347eaf87 : SUCCESS in 1h 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02244569a3f74bf2b442536e4b62c022 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0930d267847421381bc2864a476d594 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/711534d1e6374747b489dfb1206ef35d : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fb7068dba9d41ffa37a357196b3b91b : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57e4beded6084394b9b850dd7f96b641 : SUCCESS in 5m 06s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a93c7c35545a48e99ad7bebcecb974fe : SUCCESS in 30m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fed0f091674746f3a94f97999959deb8 : SUCCESS in 48m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/12b0c22e59534535adb35bf3a4999dfb : SUCCESS in 1h 03m 23s","accounts_in_message":[],"_revision_number":2},{"id":"e0fa90bbce57bd47009b473510867800645c95f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 11:33:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6eea1a3a22b77eeafc94fe7add365aa76ae3f759","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 11:35:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/388e5dc32bc047dca9b424fd09191df6 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3b2583c6415446996f3ffca1added6c : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2287fb5226c2e44093dc0e5d7c6ed3fbc88f85dd","revisions":{"df18e8d91a5a4aa2d0def7a8c91be71d92808065":{"kind":"REWORK","_number":1,"created":"2021-11-23 13:14:58.000000000","uploader":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"ref":"refs/changes/95/818895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/818895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/818895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/818895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/818895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/818895/1"}}},"commit":{"parents":[{"commit":"5445c1e54a1e96d568bc9f45b066e479640e3895","subject":"cinder-volume/ubuntu: add lsscsi and nvme","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5445c1e54a1e96d568bc9f45b066e479640e3895"}]}],"author":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-04-26 20:06:58.000000000","tz":-240},"committer":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-11-23 13:14:58.000000000","tz":0},"subject":"Add Swift lock path in Swift containers","message":"Add Swift lock path in Swift containers\n\nSwift-recon-cron requires rw access to the lock path\nspecified in in object-server.conf file. Currently it\ndoesn\u0027t exist in Swift containers\n\nCloses-Bug: #1926203\nChange-Id: Id3f824b741a5aa98efc7162fb7d49657e86a2bee\n(cherry picked from commit ff25b500d23155485dc57d10aed9070a4cba88b4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df18e8d91a5a4aa2d0def7a8c91be71d92808065"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df18e8d91a5a4aa2d0def7a8c91be71d92808065"}]},"branch":"refs/heads/stable/ussuri"},"2287fb5226c2e44093dc0e5d7c6ed3fbc88f85dd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-12-14 08:22:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/95/818895/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/818895/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/818895/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/818895/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/818895/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/818895/2"}}},"commit":{"parents":[{"commit":"3af23da7ccd07ace68ae5ec6a76b110783fac627","subject":"monasca-grafana: Fix build errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3af23da7ccd07ace68ae5ec6a76b110783fac627"}]}],"author":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-04-26 20:06:58.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-12-14 08:22:32.000000000","tz":0},"subject":"Add Swift lock path in Swift containers","message":"Add Swift lock path in Swift containers\n\nSwift-recon-cron requires rw access to the lock path\nspecified in in object-server.conf file. Currently it\ndoesn\u0027t exist in Swift containers\n\nCloses-Bug: #1926203\nChange-Id: Id3f824b741a5aa98efc7162fb7d49657e86a2bee\n(cherry picked from commit ff25b500d23155485dc57d10aed9070a4cba88b4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2287fb5226c2e44093dc0e5d7c6ed3fbc88f85dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2287fb5226c2e44093dc0e5d7c6ed3fbc88f85dd"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
