)]}'
{"id":"openstack%2Fkolla~818896","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Id3f824b741a5aa98efc7162fb7d49657e86a2bee","project":"openstack/kolla","branch":"stable/train","topic":"add-swift-lock-path-and-set-permissions-stable/xena-stable/wallaby-stable/victoria-stable/ussuri-stable/train","attention_set":{},"removed_from_attention_set":{"30609":{"account":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"last_update":"2021-11-24 14:22:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id3f824b741a5aa98efc7162fb7d49657e86a2bee","subject":"Add Swift lock path in Swift containers","status":"ABANDONED","created":"2021-11-23 13:15:15.000000000","updated":"2021-11-24 14:22:55.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":818895,"cherry_pick_of_patch_set":1,"meta_rev_id":"5d18f444c164aa85288c724ec9c02168c4171ca4","_number":818896,"virtual_id_number":818896,"owner":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-11-23 14:33:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-23 14:33:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"88dad7ccfbdb6dbdd03ab69c33ea2a599e216b06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2021-11-23 13:15:15.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"8336208bc66c2c812d13d7886833a1de3f43f282","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 14:33:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c1c29bc343344b979e56f0568655b2fa : RETRY_LIMIT in 3m 17s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6c808c1c6d9b407e9e7b1c34bcc022cf : RETRY_LIMIT in 3m 52s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/43aa0ee198cf46d19cd0321988d80b1f : RETRY_LIMIT in 3m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3daa07c336f24fb4ae8607aeae026123 : RETRY_LIMIT in 4m 10s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/90332f08aedd4cfdbacff641e5dbc6eb : FAILURE in 21m 19s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/93e672711d8a4101b5eb54ffdfcc20d2 : SUCCESS in 1h 02m 14s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9a32e1cc0ffc4523b41ef5cff31ffbe4 : FAILURE in 21m 27s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8eb3207170a94ee1b7b728be6fbf7836 : SUCCESS in 1h 10m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1f8b26901f4f4955b6d193e649fe2e09 : FAILURE in 28m 05s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b73b8899c96e4196a706134b93b9dadd : SUCCESS in 1h 07m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b02f094d7b3e45db80bc65fa230fe55d : SUCCESS in 23m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/caa9bdbcbf14412bb69c55b77ce1d7a1 : SUCCESS in 30m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2b30e5f0899848d3b494ee6372bb95a7 : FAILURE in 9m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c5a21f729804f4a98463cb5819cd5e6 : SUCCESS in 6m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75925dd86910499eb9149dcb9854f5a8 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e5c61398cca4a53b91f62c1d738062c : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1107f16fe8e94fe89e9484755360a2e2 : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/807c6f8ea6104503bcd163200fa44ea5 : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/783ddece92df493fb4c258123577b55f : SUCCESS in 4m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/80013bf417be47bb9f6ab06ed4180f8d : SUCCESS in 23m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/22019eb2728a4d679c2677ea0092e718 : SUCCESS in 24m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ab2938cd514a41618b3ca9624bacdad4 : SUCCESS in 46m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8b41c089fe994d3299b85e2dc7f27e15 : SUCCESS in 48m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/92dcd216842e4d9782b35c95127c9a04 : SUCCESS in 1h 02m 13s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"63ff2abfa49879c57c37990c48246e8b4ad88722","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2021-11-23 17:46:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46248b506644ab27877d00957bee7fcce9f09429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 18:59:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1d4cdc0fe2684e00b1e01cdeeced93a5 : RETRY_LIMIT in 2m 47s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/bbd3d980837342d690ac7957cfff7958 : RETRY_LIMIT in 3m 36s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/43efba6966a54462b0c86ec34d57cd71 : RETRY_LIMIT in 5m 17s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b2dcc1bd96af43f29e94d768a7ffde46 : RETRY_LIMIT in 3m 00s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/89da991fd51d4df69993d2ab16108f97 : FAILURE in 21m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0be62fa1b69b4194b07f706eeb66494f : SUCCESS in 1h 04m 38s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/87e17ba5a6e4437f9715222d01c73d04 : FAILURE in 24m 17s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/fb149efd8f984d19a22edbba5bf72cdf : SUCCESS in 1h 04m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/88c5d9680c8c4e29b5b27a53b0e419ea : FAILURE in 21m 28s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/bcc4070cdd39468781fb3007c66feba1 : SUCCESS in 56m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2ab5c8f2c7134ac282d5bb3f7655490b : SUCCESS in 23m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/258244db417e44ad8ae2554a084183de : SUCCESS in 21m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4327dc5d5c95428a80386fbaa3693b75 : FAILURE in 6m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/695b0446eb0941a4b3de565115760c15 : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0bcf24f23aef4f65a98a205586967083 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de2ea867e557419ab9e7acaee2735197 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d37f9938c18d48a68dcb9aabe7c3a29e : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a36443afe1d54ecabbd448b6351a5e81 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a0da444c46e4432863b749cc6090a36 : SUCCESS in 4m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dd08c4b28b7c4b3d81e0602f3030889a : SUCCESS in 28m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8ae3bf2d48b445dcb24eb8c4129cc810 : SUCCESS in 25m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c5f7e84d8dc14fafa65a8f920fb94f73 : SUCCESS in 48m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1f2c8d2a4ee64d00a6855b02aa29bcfc : SUCCESS in 45m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/249532fb75a34cc69746c6dd09488429 : SUCCESS in 1h 03m 04s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"5d18f444c164aa85288c724ec9c02168c4171ca4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2021-11-24 14:22:55.000000000","message":"Abandoned\n\nstable/train is going EOL","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7642a769fd4c688ca1f2c2cf346e947229edeb5","revisions":{"a7642a769fd4c688ca1f2c2cf346e947229edeb5":{"kind":"REWORK","_number":1,"created":"2021-11-23 13:15:15.000000000","uploader":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"ref":"refs/changes/96/818896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/818896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/818896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/818896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/818896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/818896/1"}}},"commit":{"parents":[{"commit":"fa53a60c149026ff6641e1e1d9bae240dd888f3d","subject":"Fix _get_images_dir when pip-installed with --user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa53a60c149026ff6641e1e1d9bae240dd888f3d"}]}],"author":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-04-26 20:06:58.000000000","tz":-240},"committer":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-11-23 13:15:15.000000000","tz":0},"subject":"Add Swift lock path in Swift containers","message":"Add Swift lock path in Swift containers\n\nSwift-recon-cron requires rw access to the lock path\nspecified in in object-server.conf file. Currently it\ndoesn\u0027t exist in Swift containers\n\nCloses-Bug: #1926203\nChange-Id: Id3f824b741a5aa98efc7162fb7d49657e86a2bee\n(cherry picked from commit ff25b500d23155485dc57d10aed9070a4cba88b4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7642a769fd4c688ca1f2c2cf346e947229edeb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7642a769fd4c688ca1f2c2cf346e947229edeb5"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
