)]}'
{"id":"openstack%2Fkolla~818898","triplet_id":"openstack%2Fkolla~stable%2Frocky~Id3f824b741a5aa98efc7162fb7d49657e86a2bee","project":"openstack/kolla","branch":"stable/rocky","topic":"add-swift-lock-path-and-set-permissions-stable/xena-stable/wallaby-stable/victoria-stable/ussuri-stable/train-stable/stein-stable/rocky","attention_set":{},"removed_from_attention_set":{"30609":{"account":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"last_update":"2021-11-24 10:10:04.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id3f824b741a5aa98efc7162fb7d49657e86a2bee","subject":"Add Swift lock path in Swift containers","status":"ABANDONED","created":"2021-11-23 13:15:37.000000000","updated":"2021-11-24 10:10:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":818897,"cherry_pick_of_patch_set":1,"meta_rev_id":"1b1e6b9eac144814e63e4f785015becf9404fcaa","_number":818898,"virtual_id_number":818898,"owner":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-11-23 14:03:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-23 14:03:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f1ea33736bcce41fb8ed7786ea6485c9a8572f6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2021-11-23 13:15:37.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"9f19445f4d34007c3abba88f48264e54338989f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 14:03:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/55244e00f9f74a9da4e46583318f0d4a : RETRY_LIMIT in 3m 06s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/acbf860751d24317be0d743343a58fd1 : RETRY_LIMIT in 4m 04s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/9ae700282beb4d5eaa0398ec7caf2c03 : RETRY_LIMIT in 4m 46s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/df927b367afc41018a762f181b32760b : RETRY_LIMIT in 4m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7688670e83814ce0897bb47d07c55879 : FAILURE in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55b01e41a6104ea18c6f1e1ddb7d3bb5 : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/77d56e7534e54cf9b6d5aee50800eb9f : FAILURE in 4m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b425976af2894a5e858ee82f77670edd : SUCCESS in 8m 03s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/860718174c7841728f3a8c8b5354b381 : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1eaa409f2904088adef5eb5ad0505f9 : SUCCESS in 5m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4d4470483dca41f481c74e453805b9fe : RETRY_LIMIT in 3m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1d852c2aadae42c2b500141ee905cf29 : RETRY_LIMIT in 3m 30s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"1b1e6b9eac144814e63e4f785015becf9404fcaa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-24 10:10:04.000000000","message":"Abandoned\n\nstable/rocky is going EOL","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81122918f25583b73fbc7d6b873d336908db4e65","revisions":{"81122918f25583b73fbc7d6b873d336908db4e65":{"kind":"REWORK","_number":1,"created":"2021-11-23 13:15:37.000000000","uploader":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"ref":"refs/changes/98/818898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/818898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/818898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/818898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/818898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/818898/1"}}},"commit":{"parents":[{"commit":"dd0069addbfba430d1a8cf15e91005d0f0279c11","subject":"Add /var/lib/kolla/venv/bin to Zun exec_dirs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd0069addbfba430d1a8cf15e91005d0f0279c11"}]}],"author":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-04-26 20:06:58.000000000","tz":-240},"committer":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-11-23 13:15:37.000000000","tz":0},"subject":"Add Swift lock path in Swift containers","message":"Add Swift lock path in Swift containers\n\nSwift-recon-cron requires rw access to the lock path\nspecified in in object-server.conf file. Currently it\ndoesn\u0027t exist in Swift containers\n\nCloses-Bug: #1926203\nChange-Id: Id3f824b741a5aa98efc7162fb7d49657e86a2bee\n(cherry picked from commit ff25b500d23155485dc57d10aed9070a4cba88b4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/81122918f25583b73fbc7d6b873d336908db4e65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/81122918f25583b73fbc7d6b873d336908db4e65"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
