)]}'
{"id":"openstack%2Fkolla~819351","triplet_id":"openstack%2Fkolla~master~I98cc19c95fcec07dd4e494c14c09938d754f1de0","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2021-12-04 21:06:27.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-12-04 19:42:50.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I98cc19c95fcec07dd4e494c14c09938d754f1de0","subject":"nova-compute: trim image a bit on CentOS","status":"MERGED","created":"2021-11-25 15:54:25.000000000","updated":"2021-12-04 21:08:15.000000000","submitted":"2021-12-04 21:06:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819351","meta_rev_id":"a0b3a9ca342abec1cc07af18bd222a8407aca0f2","_number":819351,"virtual_id_number":819351,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-04 21:06:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-12-03 13:21:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-12-04 19:42:50.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-12-04 19:42:50.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-12-01 17:28:34.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-25 16:05:52.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-11-25 16:05:59.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-11-25 17:55:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-03 13:21:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"907c044fce33c3e9659ff8dea6cbed1712d6784d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-25 15:54:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37987f7cac6a6ce436619c368986f06fd05e0a3e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-25 16:05:52.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cded5f876c35bd813fc435c4b74286432def8522","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-25 16:05:59.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"8a4e370c09354afa2779109397ba197413dbd6b5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-25 16:15:40.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2eef0ee078bc9ea78ea85f1a79f7bfbe665b1c4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-25 16:15:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a7645362f8885da2f4bdc93adb57eebb0ab8d940","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 17:55:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8476a91de1f34e54bac8757033db2c5a : SUCCESS in 26m 53s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ae81b24e92d544719056f73e59a352df : SUCCESS in 36m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ca7bfb9155d44a0f89c12ee46334285e : SUCCESS in 1h 15m 48s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/20838564bea340ab9cb53c0c2db4c165 : SUCCESS in 1h 10m 32s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ca8aa51802284780a5923b7a6fae6d8c : SUCCESS in 1h 29m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6781a73bb7c34bd0aebeaabed9ef02cc : SUCCESS in 31m 07s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cd30a9118121491dbcfe02d9cf85412b : SUCCESS in 19m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3524c670c5be4cbfa3fc512289b90686 : SUCCESS in 1h 01m 16s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6c059bf927de4f8f9c6a3587b49bc0c4 : SUCCESS in 1h 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cf2afedafc4448cbe0a70929d2872a9 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecfbc079eeaf440bbc685326a702a498 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e310db0958c4411d9591396b7b759c77 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/499492f33c074fe889fbbca14078a69a : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/772c762b86ff47c4b3f2347452eb7258 : SUCCESS in 4m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/034730d1a144467aac1b342d1ccf1300 : SUCCESS in 17m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/44f33c2a512f4a3f830c143452dfba2a : SUCCESS in 27m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c52430f1ec854aa9b8370a20bc03b0b4 : SUCCESS in 48m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0617c3d4b9464a72b0d1fc4bb2c80ad8 : FAILURE in 47m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/606dcff272604872b0e25f13fe7b73a1 : SUCCESS in 1h 07m 55s","accounts_in_message":[],"_revision_number":2},{"id":"f86fc56cfd7c299175357a2cba07d73a8b0f88a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 18:50:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/bfec01d9b3774b6fae88320bb294803a : SUCCESS in 2h 04m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/20148c20abf348c6b0ec3dcfe9cb95ef : SUCCESS in 1h 49m 17s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/337ddcaa2d6f4802a656e95b2f16f092 : SUCCESS in 2h 29m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4bddf6ba7f442c08dea4a72afb816a1 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/12a51a6eb5e74deca924a9495043793e : SUCCESS in 5m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7480d51ae36b4eba82041ac8f9c8bb9d : SUCCESS in 31m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7269083db728b2082882ea307566a4cdc51d1ac1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-25 20:35:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0aec13dc808b405e67dfd030eb9ccf0a8a856d67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-26 12:10:32.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"788825d9c480af73d0d0bb2d14c3186d4b719b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 13:49:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a1312e91ece149e792233335feb51cec : SUCCESS in 47m 58s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5923c63cb16a4347b3029498cab9aac4 : SUCCESS in 51m 55s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/10d004586cdf4d738e7696eec910e74d : SUCCESS in 1h 23m 44s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/57d8a710b2a849a49c7387afac2c404c : SUCCESS in 1h 25m 09s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/28272d10369d427198f62cac59c631d7 : SUCCESS in 1h 36m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/80266ef2c93c4bee8c143454e0433238 : SUCCESS in 28m 26s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/163a5d62236c4353872e2d3d329b98a9 : SUCCESS in 17m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/68ff537c089a40bbb13cc93c4104a2d7 : SUCCESS in 56m 14s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/af6448c31fd7438b82ffc4e32f784e23 : SUCCESS in 1h 07m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95624f032e8a4cd68313e88555709833 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50ca470105e84870b2539e5d401ee8e1 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09dcf1c8cbb046088dec04a427098356 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b67ec35b04bb4ca995e336a6b135d0f6 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1379cbfe8fe94b50987ed1f67cf57c8f : SUCCESS in 4m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fcf35b205c174b42baed3a6120e7bf2e : SUCCESS in 21m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/91d71daa39ed4eab9d4fbb080be8cb2b : SUCCESS in 23m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/41d75e244a184079a854b84ad7dff2db : SUCCESS in 57m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fabd6f8cbaee4be9834378504fb53bdf : SUCCESS in 52m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bb533fc224074084985bec7c11f27fa6 : SUCCESS in 1h 20m 52s","accounts_in_message":[],"_revision_number":3},{"id":"2690f467ed64e42396db747e3668df3b3df35528","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 14:33:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c8ad0cfee8ff49898d37a053257a7f6b : SUCCESS in 54m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fb9ad3f61dd24dbc87cf5ae6c9d9e0fd : SUCCESS in 1h 33m 51s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/118ff7b1048c45b081cfc9437f0a707f : SUCCESS in 2h 20m 49s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/434f3285204d4f8a89e4222476b5e814 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9d87edff51c046238c82aa7bceaf71f6 : SUCCESS in 5m 15s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/07cf23f59e9340a3b12215cee8f57ae0 : SUCCESS in 1h 42m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"51a5bcd4d35f53e9c52fc88af158ac8b8e019a8d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-01 17:27:59.000000000","message":"Topic source-upgrade-pip removed","accounts_in_message":[],"_revision_number":3},{"id":"68eb4e3ec7fd8103568e44dd00b3dd6a5a702afa","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-01 17:28:34.000000000","message":"Patch Set 3: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2e56ce6e80a5bb7af1aca1a5991244e17085da81","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-03 13:21:41.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"967cd1ba5777ddc1812534cc2eb385f68073cbda","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-04 19:42:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ccb363d6cb59c20ef2e08a80b4a6a458894358ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 19:43:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5c8d260b490bad4a3ab989a4c847904d8bb95c57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 21:06:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d90c78c356004a248ea10a60d4cc2ea2 : SUCCESS in 38m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/dbf36cb8a12348fa8beb2a9703e22d41 : SUCCESS in 1h 04m 59s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/193d87ff80324fb7a3f8ecf4bdeced79 : SUCCESS in 1h 21m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9aefe1b100ff48a28cd1ed76f172c66f : SUCCESS in 23m 22s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/943d017118144573aa2233899c11b052 : SUCCESS in 56m 23s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1a08279a0a5f4e73b423b034c4a4e65e : SUCCESS in 1h 13m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcb66f09827b42fcbdf6b590efcb21a0 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1d6f6c6eef74022ad7eea0c726c8830 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bd772cfec8746e9949a9f0d9c762513 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eee5158d57e2472bb3ff5d55e07d1579 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b478607b3df04ea49894795b6ae6a8a2 : SUCCESS in 3m 42s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8b9b85e653b240159e3b4e9e8c8f71d1 : SUCCESS in 31m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2d2048b802b14a849bf06d66e194797a : SUCCESS in 38m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/68d2cf89a72c468fa2e2ddfd25dc8e3d : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":3},{"id":"13bfac5566527f6fa7a7b9a8955daee113e38a63","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 21:06:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a0b3a9ca342abec1cc07af18bd222a8407aca0f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 21:08:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c4e446ddb854f828bd92d033b5de745 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2c9ec95e2aa46f9b7ec271be38166ca : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bbef9846c3eff5c96d4e81ffd3b4cf3446dd1eb2","revisions":{"cef787859f52ee2e9a2e3fc26a0a2a8790bd95da":{"kind":"REWORK","_number":1,"created":"2021-11-25 15:54:25.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/51/819351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/819351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/819351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/819351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/819351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/819351/1"}}},"commit":{"parents":[{"commit":"1a9d5a1a42c86b447af371ab5e0864f2b452f6ad","subject":"Merge \"Add Swift lock path in Swift containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a9d5a1a42c86b447af371ab5e0864f2b452f6ad"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-11-25 15:51:22.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-11-25 15:54:15.000000000","tz":60},"subject":"nova-compute: trim image a bit on CentOS","message":"nova-compute: trim image a bit on CentOS\n\nlibguestfs package fetched kernel-core one which fetched linux-firmware.\nWe remove the last one and save ~500MB of space:\n\nbefore/centos-binary-nova-compute: 3.3GB\nafter1/centos-binary-nova-compute: 2.71GB\n\nCloses-Bug: #1946801\nChange-Id: I98cc19c95fcec07dd4e494c14c09938d754f1de0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cef787859f52ee2e9a2e3fc26a0a2a8790bd95da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cef787859f52ee2e9a2e3fc26a0a2a8790bd95da"}]},"branch":"refs/heads/master"},"53cabd6610c469a3239ffd2546ac8e5015d46b5d":{"kind":"REWORK","_number":2,"created":"2021-11-25 16:15:54.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/51/819351/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/819351/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/819351/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/819351/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/819351/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/819351/2"}}},"commit":{"parents":[{"commit":"1a9d5a1a42c86b447af371ab5e0864f2b452f6ad","subject":"Merge \"Add Swift lock path in Swift containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a9d5a1a42c86b447af371ab5e0864f2b452f6ad"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-11-25 15:51:22.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-11-25 16:15:47.000000000","tz":60},"subject":"nova-compute: trim image a bit on CentOS","message":"nova-compute: trim image a bit on CentOS\n\nlibguestfs package fetched kernel-core one which fetched linux-firmware.\nWe remove the last one and save ~500MB of space:\n\nbefore/centos-binary-nova-compute: 3.3GB\nafter1/centos-binary-nova-compute: 2.71GB\n\nCloses-Bug: #1946801\nChange-Id: I98cc19c95fcec07dd4e494c14c09938d754f1de0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53cabd6610c469a3239ffd2546ac8e5015d46b5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53cabd6610c469a3239ffd2546ac8e5015d46b5d"}]},"branch":"refs/heads/master"},"bbef9846c3eff5c96d4e81ffd3b4cf3446dd1eb2":{"kind":"REWORK","_number":3,"created":"2021-11-26 12:10:32.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/51/819351/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/819351/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/819351/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/819351/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/819351/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/819351/3"}}},"commit":{"parents":[{"commit":"1a9d5a1a42c86b447af371ab5e0864f2b452f6ad","subject":"Merge \"Add Swift lock path in Swift containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a9d5a1a42c86b447af371ab5e0864f2b452f6ad"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-11-25 15:51:22.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-11-26 12:10:32.000000000","tz":0},"subject":"nova-compute: trim image a bit on CentOS","message":"nova-compute: trim image a bit on CentOS\n\nlibguestfs package fetched kernel-core one which fetched linux-firmware.\nWe remove the last one and save ~500MB of space:\n\nbefore/centos-binary-nova-compute: 3.3GB\nafter1/centos-binary-nova-compute: 2.71GB\n\nCloses-Bug: #1946801\nChange-Id: I98cc19c95fcec07dd4e494c14c09938d754f1de0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbef9846c3eff5c96d4e81ffd3b4cf3446dd1eb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbef9846c3eff5c96d4e81ffd3b4cf3446dd1eb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
