)]}'
{"id":"openstack%2Fkolla~820042","triplet_id":"openstack%2Fkolla~master~Ife9385c93239e910db2e4405ec4661f667357bc0","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2021-12-04 21:06:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ife9385c93239e910db2e4405ec4661f667357bc0","subject":"nova: drop pypowervm dependency","status":"MERGED","created":"2021-12-01 15:52:24.000000000","updated":"2022-02-01 12:37:19.000000000","submitted":"2021-12-04 21:06:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"820042","meta_rev_id":"1134b115ff2a92a1fd7d02758cb926775ee6d564","_number":820042,"virtual_id_number":820042,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-04 21:06:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2021-12-02 06:47:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2021-12-03 13:22:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-12-02 11:01:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2021-12-03 13:22:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-12-02 11:01:55.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"CC":[{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-01 17:35:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-02 06:47:52.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-12-02 07:55:22.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-12-02 07:57:25.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-12-03 13:22:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-12-03 20:22:43.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"CC"}],"messages":[{"id":"254c49d56c18b5c5aa12893b076a4647e5c894b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-01 15:52:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e5b9bddc6ae57ead2ceeb269f183376327a3aa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 17:35:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f5c5e06d4fdc4a65a07f7b95695edec0 : SUCCESS in 48m 20s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/57195e87447e452f8d0a4f173ce5f6a4 : SUCCESS in 55m 49s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2bab3a260e0c45aaa7f35afb49191c52 : SUCCESS in 1h 18m 55s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e72c57c93a34473eaecfe78077f09b7f : SUCCESS in 1h 16m 57s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4dd4d57b0d9c4deeb0c501aa8567a20f : SUCCESS in 1h 40m 04s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/da14f36699c646429f5554a5111a3535 : SUCCESS in 30m 14s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/35c8899cf5d049aa8bcdeb43d2ff6ac7 : SUCCESS in 19m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/18f5940331374408b73f86bec3e92ad2 : SUCCESS in 59m 58s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/72646ef172d24ffc80ca616262abd46e : SUCCESS in 1h 09m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be82cdda09724a49b247ea2378b82726 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a94dda4e027043a8abbc5ff2fa14411a : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e770ab3d8bf488dbe9d21ce86eaebe7 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d03628b6b4b482d80df8f6c4a3d59d1 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81ab657a828b4299b6696482c66fbf65 : SUCCESS in 4m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/594cb2597acf4fd09cec931be1b2bdfe : SUCCESS in 18m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6fb5337aa5b34725b108f99d7c9ac59b : SUCCESS in 30m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1141e6f4363a4daf9100092f20c1cb94 : SUCCESS in 48m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9ad27423ae8f4a7fa7279e060328d4a7 : SUCCESS in 54m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5ba0a262662b4bff8ecaad85477792db : SUCCESS in 1h 06m 32s","accounts_in_message":[],"_revision_number":1},{"id":"2c698de3097eb67537cb7023e37f61f10bba88bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 18:15:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ecea450da5094ab38f146ed3e75c7c9c : SUCCESS in 2h 00m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/53fcc0c42d36486aa851066a2e834ef6 : SUCCESS in 1h 42m 25s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7ea979f5a9564db981a3b112137770b1 : SUCCESS in 2h 20m 41s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7efa23dbf3a4f6c91235659a280f2c6 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/988d53f2fe3a417fbd2e6aaa2865f237 : SUCCESS in 5m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/394372da79b04edebba0205f76b7bf2d : SUCCESS in 1h 24m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f881e2803174b0b5315f85f051c62181311aa4b2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-02 06:47:52.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b5bc2983b5645fdfee6f428517d6d598e45aa54","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-02 07:55:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fd4a56fe76c4588598b94cde95b5a154b0848d2a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-02 07:57:25.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ef30681da53ac90e60b2b6975e7a8e9d6e1e150c","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 09:01:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/63137e349c954013aa67b1c71978e6ba : SUCCESS in 43m 19s\n- kolla-ansible-centos8s-source-kvm https://zuul.opendev.org/t/openstack/build/1ff81fce452e4c79a2288e735afa4673 : SUCCESS in 1h 04m 22s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/0c2a44647cb147c0ba5bdde2d7e502d7 : SUCCESS in 30m 04s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/c7882332c1794225844efcf0c5d7d549 : SUCCESS in 33m 00s\n- kolla-ansible-ubuntu-source-kvm https://zuul.opendev.org/t/openstack/build/5921989ad8664974af577c0d5ba0079b : SUCCESS in 45m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c56546e037515668f3b6faa1996b003fa0b42b8c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-02 11:01:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f17bb2b679865bae752074f9a3db1ba987ee44db","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-02 11:01:55.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"f7b2628f0203da2b3862bc3573c62ed8b64f38e7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-02 11:03:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"92503f14e1ce2e64e055803ceb1f6e6afe2a99ed","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-03 13:22:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bb7fc1a50b371779fad531bc6bbfb89c8c1c7d25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 13:23:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32fb11e1e3519fee67415a6f951c73d217d26e6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 14:52:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b0ff9f1132a84e8da18c89d98c773748 : SUCCESS in 43m 13s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fddf55f2b9cb4cf3bfd149f109f697af : SUCCESS in 1h 14m 58s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/745127aaa2254d4c9c46e4f3a3a89005 : SUCCESS in 1h 28m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/12639301445e4301bebebef098841d21 : SUCCESS in 27m 40s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/961982ac9cbf485fbcb5f6c8d50397de : FAILURE in 19m 58s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/4f24bb8e46e94fe78808faec5823e3ce : FAILURE in 19m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f00013247fb046c5bd90b9a69b373fcb : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3057ddcc3fc4b538fb840f55b850cc9 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e8802597ca648b0b7f1b1d65ea0c505 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b6ce7db56474a2189da8e61a252fdae : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37e9d9ad2b3f49e1968a1e02e117bc72 : SUCCESS in 4m 50s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/03a588cf48a349339e4efb165f437df4 : SUCCESS in 32m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/24a0d1a802704710bdc968b1803d9c5b : SUCCESS in 55m 24s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/789e15e2a2694f8abe13032b76df46f0 : SUCCESS in 1h 07m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9edc0643814f45118d1ad740e859de6aa43bb314","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-12-03 20:22:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac2e3f618af7e2161ceabf5e56a73c1a7e78b2c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 21:48:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/54c2673c88a24737817113bdec675a78 : SUCCESS in 1h 23m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c393720fe0ca462d8f4d799cccd29ef4 : FAILURE in 37m 21s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b6760762d86b45129ea8fb095859a546 : FAILURE in 26m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0412d2fc2f394592b894b15f63b8ce30 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b17edca25da5480f9d9de75e9933a5a7 : SUCCESS in 4m 21s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d2debb5038614f4394e15231b70d1199 : SUCCESS in 32m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"651c03ac1e3c86c16b1333391ab42569ce3198dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 21:51:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/eeea5a43a48740ec8159b9468446183b : SUCCESS in 27m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/864406a5ecff416abd8f83b427a769a9 : SUCCESS in 37m 51s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/036797c1a14b4542b0349e056b029563 : SUCCESS in 1h 22m 26s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/99ed238bc0194b41ab970b470e510e21 : SUCCESS in 1h 06m 58s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9bf1b8f39dff4608b246fc085cac05c9 : SUCCESS in 1h 03m 23s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/aa43dba74e7e485794cf3adf50de3b71 : FAILURE in 30m 38s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/57b4cfa4120e44099256ef4dcfd8d395 : FAILURE in 19m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bdc6b74e0ea34d5982b26405836aa201 : FAILURE in 20m 02s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/22c17c4b32774bb3b5692175f9f540f5 : FAILURE in 24m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39422eb0621240ac89b42e333efef2c7 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e59be376a8834c908144a9f694386557 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1149b21cb8a4e2a87166fc528f4ffbf : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f193fb5191648f88d6c1035e5ad9589 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c9fc5d2623b4202b8e19a83c454fbf7 : SUCCESS in 5m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2ee0dfe426664f579821c4418bb5f678 : SUCCESS in 14m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0dd28bf4d3644c02a28097d71b99ad17 : FAILURE in 28m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9086235458cf400d93a4232456a3e8d3 : FAILURE in 21m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1bf7694282ef4c6baf3061372c243276 : FAILURE in 18m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5c715156655947039e7e6134376d3bc5 : FAILURE in 26m 44s","accounts_in_message":[],"_revision_number":1},{"id":"508d48aef4a3cb49ae453a6a9bc80e68a90bd27d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-03 22:47:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32ff205ca35e066266f80a938c47aeef56626a30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 00:06:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8b7fef1d7aac429d93b8f8a991a88976 : SUCCESS in 29m 54s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/083e79250f974aedaef5c6600b9a0960 : SUCCESS in 33m 48s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/da54c0ec194a49e596799a148dced2b5 : SUCCESS in 1h 12m 01s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6098bbff53f64bcf8c49eab5d5908b64 : SUCCESS in 1h 03m 52s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ad8c547ab37d4c2a85faa273f5cecbcb : SUCCESS in 1h 17m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/46548ea3d0d141b8b41dd1f67feb1d8b : FAILURE in 29m 54s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9be95c5dbf28442298c529eee413686a : FAILURE in 21m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a5b408557970478c9d9f5927ff0ec457 : FAILURE in 17m 15s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/110d75020e494da4a41e993d5582256d : FAILURE in 18m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a03eddc27290478398072e7e684ec9a9 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff8b15b4fa8b46b7b0294a0690209460 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/080c03eee51043868ce1c3bc7eadffdd : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e764e89126dc4c44931d64e103f06173 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d8bb92a681c43b0be7a89716efc762d : SUCCESS in 3m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6674b8383e064792837652ca6707f3cf : FAILURE in 16m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fc51030e2d1244fcb44a08f76aea370b : SUCCESS in 30m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/832d850745c5418b95bdf3a7df17f209 : SUCCESS in 42m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b45b92919444497298f8777e00c6dd1c : SUCCESS in 47m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/87a4540f91f64e0198a73c0a4902028b : SUCCESS in 1h 05m 21s","accounts_in_message":[],"_revision_number":1},{"id":"bde79cbab314d4922a8f4a5cbdf030e2bf6d6c70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 01:36:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0a9749af982242c08589274c726ac70c : SUCCESS in 1h 39m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5293bca8d2e140debe91e90c708f387c : SUCCESS in 1h 12m 46s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/43798b4bc2c9431988d16f6b6a604a6b : SUCCESS in 2h 46m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c5b21a8d1444a55887afa63cff2d921 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f0da44caf934611983d620aaa0b161b : SUCCESS in 5m 32s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2fb44d9960d64d8988c2d44920f8885a : SUCCESS in 1h 20m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8bf0950b768a2471768d8773d030612c02565a8c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-04 18:36:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b46908ad3405cb2ccc24ae6c46554f145aa1aa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 19:51:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b69752c146324d8f82b0fba2c2ed8736 : SUCCESS in 28m 45s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/433a0430fd1f4e25982b4d274060cd8e : SUCCESS in 50m 54s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ec197f27c9684f62b555fa6c96ab20d2 : SUCCESS in 59m 56s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f8ee48895607470cb2513f683dc0c40f : SUCCESS in 1h 05m 22s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/fe69149d13434afbb93f1249d1f4adca : SUCCESS in 1h 13m 31s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c517ce39d76549588678940e39951ea2 : SUCCESS in 23m 57s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/49d9ba5767c54e3c87a91e153eb9b9bd : SUCCESS in 17m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/264047d5499647f18687465d2106249c : SUCCESS in 53m 46s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b22df6445c7c4065a365fec61add1c5d : SUCCESS in 57m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe5ceef036694842b44dde288f49d329 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13ee776b646b461c8df2414859d385ce : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec9187d56d4c40a2bab69449dc2d30f3 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c5675bd9d57467bb3a75fde37605ac7 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1482e55faa5545378dbb83b463bf1d01 : SUCCESS in 4m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a04f2d8ca1e24baeb1485fc36da2c025 : SUCCESS in 30m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0b64f4d93ddc45c6904faa7a52d8af8f : SUCCESS in 36m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/489a112b000d4019869fbf2bd38c398a : SUCCESS in 53m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d12359aabbf247e2bc6182761195a259 : SUCCESS in 51m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/92a75526f6334afca24a3871ea9bd8d8 : SUCCESS in 1h 02m 54s","accounts_in_message":[],"_revision_number":1},{"id":"ab98908054f50e7733f77a43a09dee36f466314c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 19:52:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b55fabe9fb70cfb192f47967cfd01e469eb72ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 20:28:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/45b543d7eb0048059031ddde8e622175 : SUCCESS in 1h 09m 29s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/703ab3f43a474e13aeb6fa91aad6f7a5 : SUCCESS in 52m 32s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e2a470deaa8b4c93a21e42564b860ef7 : SUCCESS in 1h 42m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5b3a60bb7d444ad8ba752917ccaa967 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7cb334c6ab3641b59837842093d4f199 : SUCCESS in 5m 22s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/331e6e71e4504d49a8f501f3e6d852c9 : SUCCESS in 50m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ebc3b9ef7d2c4c46ab6162c9fdf291d33ba6343","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 21:06:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4d5c5cf6af824acda26244ca0b1eca9e : SUCCESS in 31m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fe85799b82d04dda9dcc9234bdc304da : SUCCESS in 1h 12m 31s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2c726fbc4dd6410b9796c3a45aa5df53 : SUCCESS in 1h 14m 05s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/80a7c9179f78475abeb8b5617b0768d1 : SUCCESS in 30m 15s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a01cf22e11f742b2808788288ba225cc : SUCCESS in 1h 02m 17s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c6212a506f024dba86cdffa8b080c330 : SUCCESS in 1h 08m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1cf97067f474f49acb10bd5d805a9ea : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d6336771e354108903ed5712b023f39 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1245e654371040d98f027e3f26cdcd2a : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48bbc4052f794b829e40c29397bbec92 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1231a6e7543e4f48a680741fbdfb9988 : SUCCESS in 4m 29s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e21f5aec2c9b4c78acf8b7924a337d81 : SUCCESS in 25m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/32aceeb653a04d5daa573c3efe58faba : SUCCESS in 55m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1efa673f86934044b2c19efa058c407b : SUCCESS in 1h 06m 19s","accounts_in_message":[],"_revision_number":1},{"id":"91525f1bf68102c955a586b5c8b177db8d128548","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 21:06:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ef963955bc872d99e53cacd17b3bae4797bfbc72","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 21:09:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1ae0ee3f9ac487c8e4d3e449a4f514b : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f06d2e28791c439893fdb1bef014d0ba : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1},{"id":"0267f8ba093c7f03111f15f1cbec6ac5f288058d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-05 19:00:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1134b115ff2a92a1fd7d02758cb926775ee6d564","tag":"autogenerated:gerrit:revert","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-01 12:37:19.000000000","message":"Created a revert of this change as I8bb6c4c07c30f108e77fbb8cdc38d634b42b900f","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7de91fd60355104cbd650e0fe6f680b2eb0f2ffa","revisions":{"7de91fd60355104cbd650e0fe6f680b2eb0f2ffa":{"kind":"REWORK","_number":1,"created":"2021-12-01 15:52:24.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/42/820042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/820042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/820042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/820042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/820042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/820042/1"}}},"commit":{"parents":[{"commit":"1a9d5a1a42c86b447af371ab5e0864f2b452f6ad","subject":"Merge \"Add Swift lock path in Swift containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a9d5a1a42c86b447af371ab5e0864f2b452f6ad"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-12-01 15:48:51.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-12-01 15:52:08.000000000","tz":60},"subject":"nova: drop pypowervm dependency","message":"nova: drop pypowervm dependency\n\nNova depends on pypowervm for POWER architecture support. But it is\nunmaintained upstream and breaks CentOS builds (wants to install Py2\nonly \u0027futures\u0027 package).\n\nChange-Id: Ife9385c93239e910db2e4405ec4661f667357bc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7de91fd60355104cbd650e0fe6f680b2eb0f2ffa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7de91fd60355104cbd650e0fe6f680b2eb0f2ffa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
