)]}'
{"id":"openstack%2Fkolla~820043","triplet_id":"openstack%2Fkolla~master~Ia4cbc70f1dc54a5654ccf770c9a866dcb6fef52a","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2021-12-08 16:38:57.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-12-08 16:38:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia4cbc70f1dc54a5654ccf770c9a866dcb6fef52a","subject":"switch source images to follow master branches","status":"MERGED","created":"2021-12-01 15:54:13.000000000","updated":"2021-12-08 16:40:50.000000000","submitted":"2021-12-08 16:38:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820043","meta_rev_id":"e2c20fc88fdfeb2dfe57fa676685093b57a70156","_number":820043,"virtual_id_number":820043,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-08 16:38:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-12-08 15:13:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-12-05 19:04:06.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-12-08 15:13:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-01 17:30:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-01 19:59:05.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2021-12-05 19:02:58.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-12-05 19:04:06.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-12-08 15:13:50.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"65d74ee256499d6e50c32281d666c96f66bc0a0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-01 15:54:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e47bbd2ef956dfb05e1b19e4304c911275827ffe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 17:30:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8316d2511c3147cbae71745d6918775e : SUCCESS in 41m 28s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/925fef9aef4a40619cdcca6f7b8569df : FAILURE in 39m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4e9610e7f62547bda1a95dd297352f01 : SUCCESS in 1h 26m 35s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c9264c128224459592b93b7bbe95dbbf : SUCCESS in 59m 33s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7291659991424ac1b4257add95c13612 : FAILURE in 1h 33m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/89b0c81fa0e644f5834f93cbbd8ec170 : FAILURE in 31m 03s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9982ccae44f44929acc007e6e437db18 : SUCCESS in 15m 53s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/da38d2fec468475093e21977864b1e0c : SUCCESS in 1h 00m 52s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/924a893eb51e48a0a1b562bc6b389e5b : FAILURE in 1h 22m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75897a4f28aa4c24a7e3d5c4a1a1ac16 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ba6a82ede1a48da8648a968b2427959 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/233ad0dc97c14793a8bab063df509f12 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d00ab2a5459497cba0cc5f4b9e2a278 : SUCCESS in 5m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2ff1dff702074ba488e1d1d1e3a542c3 : SUCCESS in 17m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d929083e8c6841529e9769725d753bef : FAILURE in 30m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b3bd7a866fed43d8bfb77cbd680a7374 : SUCCESS in 48m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a52ad445885e4d668ec6ae6720a1e9c2 : SUCCESS in 50m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5e35db6bfd4b462dacfbf8acec62bc15 : FAILURE in 49m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ab1a02d2040c6c128c290128767f8decbd16a35c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 17:36:10.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c98d8fc2370a4708ba4bdfd86c8248d8 : FAILURE in 57m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ca24755b6e2d4a59aefd0768c876a545 : FAILURE in 42m 35s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7e174e78b91847e4a2080ddfb72b4411 : SUCCESS in 1h 31m 09s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2689a04993614f9ba0550d9e13d986d6 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05395316cb044489b858a7acfce4f7e5 : SUCCESS in 4m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/194fda9683ac4b4da97ad38a35cfc29a : FAILURE in 46m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf21aa3bce61d2ff5eda755f30b88a8b4a49a037","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-01 18:15:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5dd70c39d52916ce80788da6caed923c6d7d8c89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 19:53:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/12c41b2bdf0145baa4572a124e733f57 : SUCCESS in 52m 03s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/de3118e8761545b9a05459221ccbeb28 : SUCCESS in 42m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/bfeddd0904624f30b7c3c652cee5faf8 : SUCCESS in 1h 08m 02s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/04bfe4a3be6b4bfc85dfccb4f8ac8ef7 : SUCCESS in 1h 10m 35s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e9465e453b1e412c99a3fb0d3aaed90d : FAILURE in 1h 29m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/df9c2c4a2013420ca9ba320168d57856 : SUCCESS in 25m 21s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5c1131419ef54924b5e78d34ac77b60c : SUCCESS in 20m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/65daa28300a349ceb08068c08c2dc7b2 : SUCCESS in 57m 28s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1faed35c31dd4808ab8f73eec80bd05c : FAILURE in 1h 12m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5510e074f854e01b465f3af55847936 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63819005a2244072b242543f923c2bc4 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/028b0e9e13b64bcc98dd3fcb6a0626e1 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95a0a3de5a384772a55b576969f2a854 : SUCCESS in 5m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c101f654147542ccb8c9c219caad3483 : SUCCESS in 18m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c31712c05bb24edab467c25059fdb71c : SUCCESS in 31m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc7b106542214869b9a088045537cf6d : SUCCESS in 57m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/51eb3d9cd7794d65993816fb44f99277 : SUCCESS in 49m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c07f173270034259935316631f6bce56 : FAILURE in 1h 09m 01s","accounts_in_message":[],"_revision_number":2},{"id":"f4bffcb4a28116345ae2d1ed8c6b2b3b3cd860aa","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-01 19:59:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b9a196bda402e152653d040e977422239d516398","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-01 20:02:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4e6a5aad782b16163b18025673b053672dfb8a39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 20:07:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b00374694c32404195bec90198113109 : SUCCESS in 1h 46m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/85db745d2e4c4846b07bbc982fc656e9 : SUCCESS in 1h 11m 34s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/364a69005ac4436491a67e53c992ea73 : SUCCESS in 1h 37m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b998dfb2c7854dbcab158fea1acf4e5d : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a7f04d37751248bcb72ac30a28d49246 : SUCCESS in 5m 14s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1104883124624567ab6daa414dec48b3 : SUCCESS in 47m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ff251e18f185deeff72d501291f66ac95fb8aed9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-01 20:35:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a82a4689a73617b1b5d4740af8d5d4b0463f5487","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 20:40:19.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/d5aa3ce6af7b4d33b5b82f7f97d04e85 : FAILURE in 39m 50s\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/540b8aeb60a14c27a4a562a35dfbca37 : SUCCESS in 30m 07s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/fde17d3156c34d308554282ebe11bad1 : SUCCESS in 27m 49s","accounts_in_message":[],"_revision_number":2},{"id":"60701f464c142479f8958f23122528f18b301aed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-02 09:13:10.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ad27e70658e5699f22a019ada3225883d3e3359c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 10:43:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a5879b4ed85744a9870540db4657e89e : SUCCESS in 36m 01s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b46276e979e74b4fbb614ce4d252ca10 : SUCCESS in 42m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ddd22458f9ec426898749130cc5cfd27 : SUCCESS in 1h 19m 33s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e4ef5f9b09aa458eb12211670cff95b9 : SUCCESS in 1h 17m 45s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/cd3b4a4531ce4512a151681e1782ac09 : FAILURE in 1h 25m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a4a2a9c0da7c4ad58ae62dc5a381128f : SUCCESS in 28m 53s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8d1399f4c9d94914b230419fd9ec452f : SUCCESS in 15m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fb268e949a0740e5b5963398f69ca106 : SUCCESS in 59m 19s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0c644ce3e4044f2b8548f0cdee4eb362 : FAILURE in 1h 12m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d8d99b8846d46f0b1231adf1f134795 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cdb98f4472542b1b18062942e4d14db : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bd1747a3783413883e194e056921226 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5ba427c14fc4021a399c51584adb791 : SUCCESS in 5m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1d72218002564a4586eae5e409acb658 : SUCCESS in 21m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7add0f32d7394e1bb24097f6d0af7aab : SUCCESS in 25m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b6b11b85cec5418996b77ac1e5b1be7b : SUCCESS in 53m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/37afe146b22146d58d4bc82e2794ed8f : SUCCESS in 50m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/574ed76f0efd4ba0b52b5417ad621a7d : FAILURE in 1h 03m 10s","accounts_in_message":[],"_revision_number":3},{"id":"3cccd4532805813a5fb0df6e7bed9923c0c563b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 11:00:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/69cc104b3e4343a7b63bc9b422a601c8 : SUCCESS in 56m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/09650ebba69f4420a2832484ddc05098 : SUCCESS in 1h 12m 03s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c89505ab936143638c5caa8a59b96234 : SUCCESS in 1h 41m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/486fb6b7bb564d41a4e7f9d32ccfc7bd : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e149cad8070a488f8543b180ff322a21 : SUCCESS in 5m 38s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3da631505f7345b79029a0907234dcea : SUCCESS in 1h 01m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4aa4bed85308fcd2c39fdd51d34416a1dab98faa","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-05 19:02:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a811987e941030f2aa53c0097f781d62cc18ee44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-05 19:03:10.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"075cc9adbe6ad80e2f744a35bbd0e9a56ed8fcd2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-05 19:04:06.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7c3ea0ffa167157a382381a32968f7ae868d4670","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-05 19:05:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0193853ca4c77c4c2c54dcdbde9d178fb118c41c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-05 19:06:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bfd737bfbd56ce039a12eea388434a29336401f5","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 19:49:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/accbfa31f8014dfaba0b7370c37d0571 : SUCCESS in 38m 45s\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/be92c45856544d73a35abf8e31eb96df : SUCCESS in 26m 26s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/b4c17a4f11d048b786c077f4c783991c : SUCCESS in 30m 10s","accounts_in_message":[],"_revision_number":4},{"id":"9d81df49fbd5684df4d611e00ee986673cfdc493","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 20:36:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4583971cb27c4ad2b3f938fb372cd8de : SUCCESS in 24m 52s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5b42a84a27ad45a580059f4d4ee0a0ab : SUCCESS in 34m 27s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/840bbbb54c764320a7eb6dbbce2fbdab : SUCCESS in 59m 57s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b5ec0bd2ed614bb58ae178e7a5d68de0 : SUCCESS in 1h 04m 17s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ed65849fb3984d569e08eea0b079c173 : SUCCESS in 1h 32m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ee499bb249a74211ab8397e74dbfc53b : SUCCESS in 26m 56s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6e43df37694841aeaab94b8f5e75dbc9 : SUCCESS in 12m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3ea82c20faf942be9c80d0ad18932b9c : SUCCESS in 51m 31s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3b1bdfe11b4d41068f202f415bbe024d : SUCCESS in 1h 00m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d7ed507d96c4a8b9aa67b80eb3db4eb : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f0076d694a74ac99687b1cc6487b528 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80cb7bd720814fb8aa41d4cc1c58cb1c : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8128d85131984be885c51b58346a59f1 : SUCCESS in 5m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4a38c67a313c4045bbfbf976766965ce : SUCCESS in 19m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/337ca0ae7ebe49b690ea57bae73f81fa : SUCCESS in 28m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c7f64d3fbefe42b08dbda2eca0b5f048 : SUCCESS in 55m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f30b080391bc429c943e651a4e7fb1ea : SUCCESS in 58m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/788478906aa24ceabb0fb7fc7f92c1c7 : SUCCESS in 1h 08m 41s","accounts_in_message":[],"_revision_number":4},{"id":"15b9f0a4835c4998b34e4024e6d361f8ef29af23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 21:48:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b7bd56c2af04447f886b2f7ef9c958fc : SUCCESS in 2h 15m 15s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ec131da37e33433d93988b81b0766635 : SUCCESS in 1h 55m 49s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/74c152ad1f094e88a8c65971ebcc5cfc : SUCCESS in 1h 30m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a6b62ff3da5a4718aac0ffd9f6d0ce14 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fac27a9dee5343ada7b08faf1140ba30 : SUCCESS in 4m 50s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9671d6f680664dd6817cc2bc03e25f00 : SUCCESS in 37m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"52c6d89e405f103d5d93e8c75424bd7cf3463d59","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-08 15:13:50.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"17838b6e5606111d7fd7a7193c4a2df8b9235330","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 15:14:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"eb0bc3233fdeee4b11b54fa1429d938d34e6e760","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 16:38:57.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c40fe8ae8fc34c869c0d252579ab60a1 : SUCCESS in 37m 39s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/21392232fe5b4ceea7b634d00fe14e0e : SUCCESS in 1h 10m 20s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/13a46ed52453411bacff6619640c6e8e : SUCCESS in 1h 20m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b667a7b93557435e9592c173eb7eb284 : SUCCESS in 34m 46s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0c4b7adac89d45a49db9ee000c9d174d : SUCCESS in 57m 21s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c7a13de0c52f41609ae48ca264196718 : SUCCESS in 1h 09m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71b3cd91ced04ba89a07ca97f7a833b0 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e38d97aa23ce428fb944727a41e6347a : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33f5a562076b453c84ba6a582ee4c9d7 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/740bdf9cec7e4583953085a2fff84e40 : SUCCESS in 6m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/70438800f81942de80c05533682f6aff : SUCCESS in 46m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/976a459d221544998ac9fa2b0ba99c71 : SUCCESS in 52m 07s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c751aca85f094628be340a23c15cf8b2 : SUCCESS in 1h 11m 18s","accounts_in_message":[],"_revision_number":4},{"id":"c1add5f0a414afb7d36c6a418d35665215c0de4c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 16:38:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e2c20fc88fdfeb2dfe57fa676685093b57a70156","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 16:40:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43bcb67b7daf487dbaac8e01007f70c5 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8cf23a6c312c5318c05c6e6b97f1a3f7ca34f0e4","revisions":{"13de905c8da05f00bf857d68b389470384957165":{"kind":"REWORK","_number":1,"created":"2021-12-01 15:54:13.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/43/820043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/820043/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/820043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/820043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/820043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/820043/1"}}},"commit":{"parents":[{"commit":"1a9d5a1a42c86b447af371ab5e0864f2b452f6ad","subject":"Merge \"Add Swift lock path in Swift containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a9d5a1a42c86b447af371ab5e0864f2b452f6ad"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-12-01 15:53:33.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-12-01 15:53:33.000000000","tz":60},"subject":"switch source images to follow master branches","message":"switch source images to follow master branches\n\nWe passed R-17 point.\n\nChange-Id: Ia4cbc70f1dc54a5654ccf770c9a866dcb6fef52a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13de905c8da05f00bf857d68b389470384957165"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13de905c8da05f00bf857d68b389470384957165"}]},"branch":"refs/heads/master"},"7cc7b30dcd23170a35b0ff00aa6f8e7e1efd56c4":{"kind":"REWORK","_number":2,"created":"2021-12-01 18:15:48.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/43/820043/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/820043/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/820043/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/820043/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/820043/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/820043/2"}}},"commit":{"parents":[{"commit":"1a9d5a1a42c86b447af371ab5e0864f2b452f6ad","subject":"Merge \"Add Swift lock path in Swift containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a9d5a1a42c86b447af371ab5e0864f2b452f6ad"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-12-01 15:53:33.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-12-01 18:15:17.000000000","tz":60},"subject":"switch source images to follow master branches","message":"switch source images to follow master branches\n\nWe passed R-17 point.\n\nWatcher and vmtp are disabled due to PrettyTable version conflict.\n\nChange-Id: Ia4cbc70f1dc54a5654ccf770c9a866dcb6fef52a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cc7b30dcd23170a35b0ff00aa6f8e7e1efd56c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cc7b30dcd23170a35b0ff00aa6f8e7e1efd56c4"}]},"branch":"refs/heads/master"},"fbacecac01cde3a66fca3bb80c81df50e31a9e0d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-12-02 09:13:10.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/43/820043/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/820043/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/820043/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/820043/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/820043/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/820043/3"}}},"commit":{"parents":[{"commit":"7de91fd60355104cbd650e0fe6f680b2eb0f2ffa","subject":"nova: drop pypowervm dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7de91fd60355104cbd650e0fe6f680b2eb0f2ffa"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-12-01 15:53:33.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-12-02 09:13:10.000000000","tz":0},"subject":"switch source images to follow master branches","message":"switch source images to follow master branches\n\nWe passed R-17 point.\n\nWatcher and vmtp are disabled due to PrettyTable version conflict.\n\nChange-Id: Ia4cbc70f1dc54a5654ccf770c9a866dcb6fef52a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fbacecac01cde3a66fca3bb80c81df50e31a9e0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fbacecac01cde3a66fca3bb80c81df50e31a9e0d"}]},"branch":"refs/heads/master","description":"Rebase"},"8cf23a6c312c5318c05c6e6b97f1a3f7ca34f0e4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-12-05 19:03:10.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/43/820043/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/820043/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/820043/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/820043/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/820043/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/820043/4"}}},"commit":{"parents":[{"commit":"0443cee2232bb7cf82e7cc68364b78c71810503c","subject":"Merge \"build: 5s delay + note when binary images are used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0443cee2232bb7cf82e7cc68364b78c71810503c"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-12-01 15:53:33.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-12-05 19:03:10.000000000","tz":0},"subject":"switch source images to follow master branches","message":"switch source images to follow master branches\n\nWe passed R-17 point.\n\nWatcher and vmtp are disabled due to PrettyTable version conflict.\n\nChange-Id: Ia4cbc70f1dc54a5654ccf770c9a866dcb6fef52a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cf23a6c312c5318c05c6e6b97f1a3f7ca34f0e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cf23a6c312c5318c05c6e6b97f1a3f7ca34f0e4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
