)]}'
{"id":"openstack%2Fkolla~820345","triplet_id":"openstack%2Fkolla~stable%2Fxena~I98cc19c95fcec07dd4e494c14c09938d754f1de0","project":"openstack/kolla","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2021-12-06 15:21:56.000000000","reason":"Marcin Juszkiewicz replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-12-06 16:37:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I98cc19c95fcec07dd4e494c14c09938d754f1de0","subject":"nova-compute: trim image a bit on CentOS","status":"MERGED","created":"2021-12-05 18:57:49.000000000","updated":"2021-12-06 16:39:59.000000000","submitted":"2021-12-06 16:37:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820345","cherry_pick_of_change":819351,"cherry_pick_of_patch_set":3,"meta_rev_id":"649b0fe23c63002a644348ac4304679dc41a1ed3","_number":820345,"virtual_id_number":820345,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-06 16:37:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-12-06 15:21:56.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-12-05 18:57:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-12-06 15:21:56.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-05 18:57:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-12-05 20:32:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc49c78ac3bf36a27a72fd717e9fc031f2df3fca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-05 18:57:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8a958d5bf40c97010b5513a7d64719f6703ba21b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-05 18:57:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8f6aa52c4a74d6a92027fc764b965ef9fb917796","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 20:32:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/07a074f7c5c543169355a541485bc49d : SUCCESS in 24m 13s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/61643f94c4d54ac082b25d9234ef1d75 : SUCCESS in 38m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1ed18c1da58246a4ad0889bb87dea67c : SUCCESS in 1h 03m 15s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/dab34304f4a64de3acddfb8ec67976ee : SUCCESS in 1h 13m 55s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7757fb5f32f742fbb0956def022693af : SUCCESS in 1h 33m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8343361448864ac4b3a9a56880ec466a : SUCCESS in 21m 06s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ec8a6439bc654ca28eb97d660c9a73ac : SUCCESS in 19m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b0642c1e41c34ff592e980054cdca279 : SUCCESS in 1h 11m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7707efab4e3143ef82cc5cfda939d0af : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21a6f571d72642fb964fa89d7d297778 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/795b5497d20446409cc1ff764d7c4e4f : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfc20ebc311f45a7b1b9b96518c253c7 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55c68ad1236a4452a499780e9aa3714a : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf905fffe3fb45f0afe695b5cbf023f6 : SUCCESS in 5m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bcc84a1c34224c25a4734c0ade599f2e : SUCCESS in 26m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9fc54261d1a349c7b76df0993cd455ab : SUCCESS in 30m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/52e9786a7e9e4105be51ee49daec78a6 : SUCCESS in 50m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b277a85b6f3d4d9287a4e75eec4d3aa7 : SUCCESS in 53m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3d52c642fb374b85b17733c372349d56 : SUCCESS in 1h 08m 33s","accounts_in_message":[],"_revision_number":1},{"id":"20c40e648b43d743c888cc6c93866bfa0ee20cf1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 21:26:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c395e71395084c11b0a050bed1758c33 : SUCCESS in 2h 20m 16s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c3dc356cf01c4cf2937c47fe004c2efe : SUCCESS in 42m 25s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ae5dee3df9de406f96aee30d07e64c82 : SUCCESS in 2h 26m 40s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c5206b93dd54acdb1e17e8793abc6b9 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe2228d734784a1e81dcd5d5270d2013 : SUCCESS in 6m 47s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c3cf682bddf452eaa9bcb626cefa9d6 : SUCCESS in 1h 46m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"490224a45be8e67007a23cb397eaba6f78bc31fe","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-06 15:21:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d1fe8628d692aaa212526c985d929705e932d8fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 15:22:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"80ee500a995e81b12c9129ddcfecbf768abaa1c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 16:37:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b063313987c04e04b4bf874517bb58b6 : SUCCESS in 57m 51s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2974d401c30e499e9a6e3b42cc3e4ad3 : SUCCESS in 1h 11m 27s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6171f94d4ba2494f8ff283de51c5e851 : SUCCESS in 1h 14m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/962ba203935d489fa9e0f4f268ad79c2 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a27d64eacfd64aa59b7aaf5707d03861 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0d7c2b88e754bf39c7141b0fb4aef07 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4235c86999524c68840d81291e4ad317 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eba09f24d47a421c87b431e664171d3c : SUCCESS in 4m 49s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f8416dbba0a34b26abd5f4065ee6e7cb : SUCCESS in 30m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c407e8447e2404099fa6207047be9b1 : SUCCESS in 53m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0b656fbdaa80453b8578a4c60e4aa535 : SUCCESS in 58m 42s","accounts_in_message":[],"_revision_number":1},{"id":"326344b1f846d025c74c6d65e97c9219693b9df1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 16:37:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"649b0fe23c63002a644348ac4304679dc41a1ed3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 16:39:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb7bfc1d4cc248d0962bc5b12644aa75 : SUCCESS in 1m 24s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ac095e0552e4870b5d961f71c7cead0 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c02ec3f8ee0af23a2e9ea29b9427cbfbe1912984","revisions":{"c02ec3f8ee0af23a2e9ea29b9427cbfbe1912984":{"kind":"REWORK","_number":1,"created":"2021-12-05 18:57:49.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/45/820345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/820345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/820345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/820345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/820345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/820345/1"}}},"commit":{"parents":[{"commit":"bad1c1a4722d9738c90524fe276ff08843920364","subject":"reno: polish release notes for Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bad1c1a4722d9738c90524fe276ff08843920364"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-11-25 15:51:22.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-12-05 18:57:49.000000000","tz":0},"subject":"nova-compute: trim image a bit on CentOS","message":"nova-compute: trim image a bit on CentOS\n\nlibguestfs package fetched kernel-core one which fetched linux-firmware.\nWe remove the last one and save ~500MB of space:\n\nbefore/centos-binary-nova-compute: 3.3GB\nafter1/centos-binary-nova-compute: 2.71GB\n\nCloses-Bug: #1946801\nChange-Id: I98cc19c95fcec07dd4e494c14c09938d754f1de0\n(cherry picked from commit bbef9846c3eff5c96d4e81ffd3b4cf3446dd1eb2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c02ec3f8ee0af23a2e9ea29b9427cbfbe1912984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c02ec3f8ee0af23a2e9ea29b9427cbfbe1912984"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
