)]}'
{"id":"openstack%2Fkolla~820346","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I98cc19c95fcec07dd4e494c14c09938d754f1de0","project":"openstack/kolla","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2021-12-05 20:25:36.000000000","reason":"Marcin Juszkiewicz replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-12-05 22:01:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I98cc19c95fcec07dd4e494c14c09938d754f1de0","subject":"nova-compute: trim image a bit on CentOS","status":"MERGED","created":"2021-12-05 18:58:06.000000000","updated":"2021-12-05 22:03:08.000000000","submitted":"2021-12-05 22:01:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820346","cherry_pick_of_change":820345,"cherry_pick_of_patch_set":1,"meta_rev_id":"0738763621de6083c2268a2f29ff0a2a711ec926","_number":820346,"virtual_id_number":820346,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-05 22:01:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-12-05 20:25:36.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-12-05 18:58:12.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-12-05 20:25:36.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-05 18:58:06.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-12-05 20:15:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5c970e7eabb8de2070a5d641b4770c4f2d3ddd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-05 18:58:06.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"4a179ceb2e277a630e8786e788fa49a4993860ef","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-05 18:58:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"838061cd44e47437289a346857a84fbc30b109b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 20:15:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7a024f68dd66428fb90989956911ea99 : SUCCESS in 43m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/72eed0a7007942a3aed7db6382021617 : SUCCESS in 42m 09s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b18778e350304976afec818f87077872 : SUCCESS in 1h 06m 08s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f3010435344447c38f714d963e48bd36 : SUCCESS in 1h 00m 20s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7e91b5a7965e48b0954ccb0c2c78ee5f : SUCCESS in 1h 16m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/940d45cfd4e641928d8db8fe34d937bf : SUCCESS in 33m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6aeaf03d9a0b47ebb59bebc3d748f280 : SUCCESS in 19m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/354edec323eb40c99595b738699cd03c : SUCCESS in 52m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ec49a3a11224ce7a99bc58b8922e44a : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4a4700095ac47e59f996fccbf10378b : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ffdfa59f06a4dd99ee5e832893fda04 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea84c1349dcd4dd4b4a3758920fbd669 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eec7762afea4d36b5498c68d251b7f5 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5fe1a1d2379459b9f393296d33e2b24 : SUCCESS in 4m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fe1b471e22d145e1b5bf73e5ee69a44d : SUCCESS in 21m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c86f7dda16d44c5e951f4cfa70a82943 : SUCCESS in 44m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c09eada9e2e6417f99cb77a6ab23fc2e : SUCCESS in 53m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/97a30307c1df41ab9f9bd69bc2dca7cb : SUCCESS in 54m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/971100d0d2a1442a948ec059d237e7da : SUCCESS in 1h 04m 12s","accounts_in_message":[],"_revision_number":1},{"id":"b96a0709106025f2a076177aba02d45ed2c8d9fc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-05 20:25:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"382ae6da87ae5bea6649d28e4fcdb49848812dc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 20:25:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da376eed53ec69955ef072673c9e9d28c4577294","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 21:19:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9ed261e0abcb43af87dda956b863fc5e : SUCCESS in 50m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/de5e08316f3d431d9c55f8a96aad5269 : SUCCESS in 1h 48m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/afcefad06bab4d0784d1e7030eb876fa : SUCCESS in 2h 19m 04s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/99a9c8a64b92446aa65e3999d0c00ae6 : SUCCESS in 2h 00m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"097e92ef98ffcfacabfa0bd9682ef6c161f5bdfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 22:01:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/55a7f51f0fca4451b90b53e2c1546fa1 : SUCCESS in 41m 06s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2b419b02e6ed4e4abd64a0260a39609a : SUCCESS in 1h 30m 15s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b245169304654f039c62a98905b9823f : SUCCESS in 1h 17m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9077e1e24962497ab13f689a63d340ae : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/910ef8558cff4796bf9824443ed1dab7 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f2cafec009f455594f04c2bf8bb47cc : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f62c9720bbc1462ab9b786b990d7a909 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df31e56f39bc41cf920ad9dfe3fb67d3 : SUCCESS in 3m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/20559009bc004747b4bc51addb06003b : SUCCESS in 30m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6ccde08723d4c8e8aa8e169c0cde351 : SUCCESS in 54m 00s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cd769c16d1e14baa91ced6587835903b : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":1},{"id":"b5b72a96d769f1658f6fd8fa6a530d58e53adb2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 22:01:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0738763621de6083c2268a2f29ff0a2a711ec926","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 22:03:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccd7fcb352db4d83b4d2f17ce7f09b3a : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6f4aad0cdc64887bbe2feffb1db658d : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c33f878c1c6a774c6da37e8da0538bdc7add60e","revisions":{"3c33f878c1c6a774c6da37e8da0538bdc7add60e":{"kind":"REWORK","_number":1,"created":"2021-12-05 18:58:06.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/46/820346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/820346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/820346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/820346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/820346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/820346/1"}}},"commit":{"parents":[{"commit":"f81e7939fdff00cb738a49d7aa83ba1445211c8a","subject":"Add Swift lock path in Swift containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f81e7939fdff00cb738a49d7aa83ba1445211c8a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-11-25 15:51:22.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-12-05 18:58:06.000000000","tz":0},"subject":"nova-compute: trim image a bit on CentOS","message":"nova-compute: trim image a bit on CentOS\n\nlibguestfs package fetched kernel-core one which fetched linux-firmware.\nWe remove the last one and save ~500MB of space:\n\nbefore/centos-binary-nova-compute: 3.3GB\nafter1/centos-binary-nova-compute: 2.71GB\n\nCloses-Bug: #1946801\nChange-Id: I98cc19c95fcec07dd4e494c14c09938d754f1de0\n(cherry picked from commit bbef9846c3eff5c96d4e81ffd3b4cf3446dd1eb2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c33f878c1c6a774c6da37e8da0538bdc7add60e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c33f878c1c6a774c6da37e8da0538bdc7add60e"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
