)]}'
{"id":"openstack%2Fkolla~820347","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I98cc19c95fcec07dd4e494c14c09938d754f1de0","project":"openstack/kolla","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2021-12-06 15:22:06.000000000","reason":"Marcin Juszkiewicz replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-12-07 18:39:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I98cc19c95fcec07dd4e494c14c09938d754f1de0","subject":"nova-compute: trim image a bit on CentOS","status":"MERGED","created":"2021-12-05 18:58:19.000000000","updated":"2021-12-07 18:42:58.000000000","submitted":"2021-12-07 18:39:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820347","cherry_pick_of_change":820346,"cherry_pick_of_patch_set":1,"meta_rev_id":"01fcbe3ead52771b5c5019ea4ebb3b8b9ea26d00","_number":820347,"virtual_id_number":820347,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-07 18:39:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-12-06 15:22:06.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-12-05 18:58:25.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-12-06 15:22:06.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-05 18:58:19.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-12-05 22:05:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f316a2c6aa6832a123d50c2742f76be114a32164","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-05 18:58:19.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"d8661f2772996b94440b532fd1a1d1bc4011c160","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-05 18:58:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1e69acc07d79ed4c3c4c482a9a88060d20528228","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 22:05:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e8f1c2bf90ca43b1b0f1ce7f26b24545 : SUCCESS in 33m 48s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ceab3d7a87fa4d6ab36886a8fc8c53a4 : SUCCESS in 32m 03s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ad8ba0e0710c4b52a5df943b635098f8 : TIMED_OUT in 3h 01m 29s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1284856238fa470b8236c56ea0bd159e : SUCCESS in 27m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9a7d24b3504444a3b9b4092686661f51 : SUCCESS in 1h 12m 04s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f4766441a14f498c9c99cb64b0baf55c : SUCCESS in 1h 00m 16s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/edb514e08f2a463cbf765d1107b072e5 : SUCCESS in 1h 12m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7ecde50f657d4bb4ab4fc3e37567c69e : SUCCESS in 58m 35s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/34ed5d2bb53842c0bf9ba29a3cb4b466 : SUCCESS in 58m 53s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/421f294eff9e4882982817425de29637 : SUCCESS in 1h 10m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1509a7b074c44d8bb88d277347008135 : SUCCESS in 19m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/85925900efc54656b558c5dec968916a : SUCCESS in 19m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/925682e67c80439596a7c56369a82d0a : SUCCESS in 54m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cab0e0af3cc48e3ac1d32198dd5d20c : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4ed511371a147ab86f9889890e6a3d0 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/918ff30be268468993697f72cd473a05 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77d9900a737547cdbdd08be7b32c74e9 : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/129a7c894fff49489196010b9b7f1070 : SUCCESS in 3m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/58fd6fa7d67145ddaf065da63540500e : SUCCESS in 20m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7131a47ab47b40f688fe283a7a8d302f : SUCCESS in 20m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c270e99e14674db5a9fd5fc6f675273a : SUCCESS in 52m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/049efdec873a4abd88d4c22f3ce0a318 : SUCCESS in 54m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1a70a108b0a846cc8af55ddddc05e829 : SUCCESS in 1h 05m 27s","accounts_in_message":[],"_revision_number":1},{"id":"6a472cd5e9461dba4fb5c16fb2ac0169803bb457","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 22:38:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/cd60f5ecfcbf49c798ee4d2312736b3f : SUCCESS in 1h 52m 22s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/bce8ec5ef661422cb987d5eb13e25fb0 : SUCCESS in 2h 26m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ab2790800f3540f5b5c1bf4a93c7574e : SUCCESS in 1h 43m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/35bc51d37b5b4c2b94b368c1c9770ab5 : SUCCESS in 2h 38m 01s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5299f18657174710ae8772d92b028614 : SUCCESS in 2h 03m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f0531107b64335f30ec838825cfe172a3c9d14b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-06 15:22:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dcc2d2425c70a9dbb8d267459d956b2a96bc49e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 15:22:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"76bf39c423337e941c7770b5f2ec7e8850951cc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 16:37:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/06217037306d4718a8757b7d5f67ffda : FAILURE in 27m 26s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3ec211eac93f48c19019ea91e772b778 : SUCCESS in 30m 11s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3a2caff910de4b5ebc02d0318bb6975c : SUCCESS in 1h 04m 37s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/71edcb025b954087aa55ea3b8e54975d : SUCCESS in 1h 09m 05s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/71e8cf17dd2f470a8b74afc209071581 : SUCCESS in 1h 00m 47s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c825b37482b5478cb14c4e8c367cc6f3 : SUCCESS in 1h 09m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e1a7b37d397459287a8b662d71edf14 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a7f9484827e444290a3a2c4da802b41 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3af6c221a698488bb4f89f52c6faaa09 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d730f90bb48b4e60986affa7c2ea3bb5 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e24ef3d6466c4c1fb92a2b5a1a760462 : SUCCESS in 4m 35s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/877e955715064c67907b6000d3ddd576 : SUCCESS in 21m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/54f27b9feaea4030b74e50c5e800dbee : SUCCESS in 54m 24s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/09380d8c1b2047b893b90cc3add57081 : SUCCESS in 1h 00m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e089bbd70aee4b6c32052fd0c3eea4d8fa1e6fd8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-07 10:58:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8fea4bc26bc46da14d59a869e7adba4af600e139","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-07 10:58:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"004e7970249302127be56f4610afc8e93f828a14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 14:05:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a6bca16ff18e4d138536a9f0f3923cd3 : SUCCESS in 1h 31m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ec94b1da195f451e8c8f143b07d60699 : SUCCESS in 1h 32m 28s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/120f931cab2a4f1884b6de6a21fb407f : SUCCESS in 33m 18s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/31c6029c9ce744778fd9b78a2cc11ce9 : SUCCESS in 1h 35m 00s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7356e3c895734823948fc6f58b60e9e5 : SUCCESS in 1h 24m 40s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/512e0f21e14143cc904fc855d2241468 : SUCCESS in 1h 49m 33s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/512efe98f60f46e59a0fb9bdb3782e3a : SUCCESS in 1h 31m 45s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0c040b8db21a4ae5b9bac20496ec41d7 : SUCCESS in 1h 07m 24s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/556b2529f53444e2b030571d21d18907 : SUCCESS in 1h 06m 28s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/249cafd141a84f9aa230228c8d915f1e : SUCCESS in 1h 15m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ce3f1a3444024cb99a9bfce7d05fbde0 : SUCCESS in 36m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/45daea43ca204456ae37af7220577d06 : SUCCESS in 22m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/53544568227c426bb55102fcb773f538 : SUCCESS in 59m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4616e547f3a47fab8855883d55b8d07 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/996caa0eb24b46b193a56f15db765f2e : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f697247c2554be68d70c7c11d93bae2 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cf4f6aadb1546799137017fd4ab8109 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/571bad59025045d3b1a27016594f6359 : SUCCESS in 4m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/138cf97d60f5498d86f25f29114a91a4 : SUCCESS in 22m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/11b3f908bf5f4ab29f69d77095704ab6 : SUCCESS in 1h 14m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f4c3fd9c51445adad52848946cef722 : SUCCESS in 1h 13m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/240fca85e0a74cceb65bb4e3d57c41f0 : SUCCESS in 1h 34m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/04a4c7d3bfed4f508f7bd8404c5b46bd : SUCCESS in 1h 22m 26s","accounts_in_message":[],"_revision_number":1},{"id":"0c4d01da5283ae4fd4a181cf887e344aab08d131","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 15:11:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/157b17e5804c4c47b11eda543366b0fe : SUCCESS in 2h 28m 12s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c52945914c144e1985febb6b72c26835 : SUCCESS in 2h 30m 58s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5d5e6857be9d41fa9997e4c3a8581c8d : SUCCESS in 2h 10m 08s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b25c2bd2ab0240d09b03ca3fb61283fc : SUCCESS in 2h 33m 17s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e79e17572bd9405d9c35cf0f7eae6227 : SUCCESS in 1h 53m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08503f8e7702e1c8b0d9168ac0c704dc21b850ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 17:16:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4bedc78b6a283b12c3f4dae820bac9f6fe01027","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 18:39:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8efd53fc42db4c23a18c455a2eb69cb9 : SUCCESS in 30m 03s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/84bbcfda219946c082b2dbebaa87f60f : SUCCESS in 37m 33s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1a4322a09d344b82834dd62184b10857 : SUCCESS in 1h 00m 07s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9d708be196ba45478b6ed2b14fc27be2 : SUCCESS in 1h 16m 33s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6229f6daa9614076811c027f068013e8 : SUCCESS in 59m 35s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e1fdbbc449c844239c5ac13b4dddb5ea : SUCCESS in 1h 04m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47a559c818ec48abb76bb4b43c34c81d : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30774b644ce74ee7b22e8ea11de31bcf : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b973deaa0e7b48b4bd43967ae3c1046c : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cbdac6cda8d47e787d690ec04dd296c : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c87d8fa53294c349b6ba6108357f951 : SUCCESS in 5m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cb960e11dc5143818ab4165eaf05c21a : SUCCESS in 25m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3bc09d2e645148f6899e2aeb0890fd19 : SUCCESS in 50m 14s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d9ef996ef1bb41c8ad54653fd70d3d7a : SUCCESS in 1h 04m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a9302603a810cbd4af5793ec4a31fb5372196c04","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 18:39:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"01fcbe3ead52771b5c5019ea4ebb3b8b9ea26d00","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 18:42:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee2324874d454f7bba73036717d30665 : SUCCESS in 1m 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db111fb298a844fe9c7e9f7eb8eb63a8 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2774e59ca74b2131318024002eb754badbe05790","revisions":{"2774e59ca74b2131318024002eb754badbe05790":{"kind":"REWORK","_number":1,"created":"2021-12-05 18:58:19.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/47/820347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/820347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/820347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/820347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/820347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/820347/1"}}},"commit":{"parents":[{"commit":"0bc8f2a335b28f9aabc796f6ebddaa8424b6fb61","subject":"Add Swift lock path in Swift containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0bc8f2a335b28f9aabc796f6ebddaa8424b6fb61"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-11-25 15:51:22.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-12-05 18:58:19.000000000","tz":0},"subject":"nova-compute: trim image a bit on CentOS","message":"nova-compute: trim image a bit on CentOS\n\nlibguestfs package fetched kernel-core one which fetched linux-firmware.\nWe remove the last one and save ~500MB of space:\n\nbefore/centos-binary-nova-compute: 3.3GB\nafter1/centos-binary-nova-compute: 2.71GB\n\nCloses-Bug: #1946801\nChange-Id: I98cc19c95fcec07dd4e494c14c09938d754f1de0\n(cherry picked from commit bbef9846c3eff5c96d4e81ffd3b4cf3446dd1eb2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2774e59ca74b2131318024002eb754badbe05790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2774e59ca74b2131318024002eb754badbe05790"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
