)]}'
{"id":"openstack%2Fkolla~821044","triplet_id":"openstack%2Fkolla~stable%2Fxena~I754939da7636c57d2a8d5b83debb5d8a58e38432","project":"openstack/kolla","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-12-09 12:29:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I754939da7636c57d2a8d5b83debb5d8a58e38432","subject":"Ensure nvme-cli is present in nova-compute images","status":"MERGED","created":"2021-12-08 21:55:12.000000000","updated":"2021-12-09 12:31:37.000000000","submitted":"2021-12-09 12:29:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821044","cherry_pick_of_change":820892,"cherry_pick_of_patch_set":2,"meta_rev_id":"f8f7a463ee2f50f6c30f174c7ad5d5bfdd875083","_number":821044,"virtual_id_number":821044,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-09 12:29:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-12-09 11:01:49.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-12-09 07:33:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-12-09 11:01:49.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-08 23:34:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-09 07:33:58.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-12-09 11:01:49.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"3161f5a9c1220c226fe0fcd0b91604daaa2a8632","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-08 21:55:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6392158db8b1aa1ab0991e0d12e6bb8aac9677f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 23:34:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1bad9f0ffd8e4444b2b3bdc0616dfb5a : SUCCESS in 45m 31s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6e9a0be2c42f47058c23423e369dd02e : SUCCESS in 38m 22s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/92fba1c53ae54dbca536e33be6cd63ed : SUCCESS in 1h 12m 19s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b24db9df8b2b4865a4b6b7f437454195 : SUCCESS in 59m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/99b56ccb0169412ea666e9265e66c43d : SUCCESS in 1h 21m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f0d45480bf77444791c45a858785e475 : SUCCESS in 25m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/426305b14ea64324bb2bb913e33a595f : SUCCESS in 21m 56s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/469f9e4b761240528ce5c9980f1e391d : SUCCESS in 1h 03m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/654097ba5660451e87baeb741552b101 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5214fb040dbf46b78ceae475018ef6a4 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d2d5db759044fe6924012f8a2ccacf6 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a350a1346f2a40f2852f1e46c269beb1 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f24fb27351bd47fb9660d077eec5cb2f : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a80b4f73d67341a5ab94cd20c8d5aabb : SUCCESS in 3m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2deebdd535ae46c1b8d2baba691525fa : SUCCESS in 19m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5c60c1e489eb4775ba3e43afb1b4e724 : SUCCESS in 23m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/952c67e9b7ff40a4a1492fad21fc687d : SUCCESS in 49m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8fb3b5947d9a455e8ac6b9a647e422fc : SUCCESS in 39m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/91dab5aaa26a4fd1891a3e4ff6a9c164 : SUCCESS in 1h 36m 47s","accounts_in_message":[],"_revision_number":1},{"id":"eef6d2f15d41a3ae7aaf90352d5c20dfa36dbe59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 00:45:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d2ee035694e545119e559153eb5fdee3 : SUCCESS in 2h 47m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bf54ca261bdf4216ac7895b27c8aef9f : SUCCESS in 2h 02m 43s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/071294fd3f3244dd9811d20522d5d45d : SUCCESS in 2h 40m 53s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9de4591b54484accaac1d1e01536592f : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/76f9edd9b2244ed0a9c6c65365366b4d : SUCCESS in 8m 02s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e3ce85a495b944a7a951da12bb7f195d : SUCCESS in 2h 15m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d07d456a13ca679c440240d469b431f5d56fe218","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-09 07:33:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1f9c5e7aef3d751815d7a7fa8bf8c32a4af9febb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-09 11:01:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d8b93c960f6ec2ca8809ad511ccb368fca00c0ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 11:02:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c0123a4eb3d32aee6a385127ef6fdf400a9274a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 12:29:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f8f866784fae44c8909f5aa4b77f5a90 : SUCCESS in 28m 16s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/05d9b728c9704f9e8c721a36a382e5ed : SUCCESS in 1h 10m 37s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c4120366d66643fa8ce286d81977023b : SUCCESS in 1h 26m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b19a4d6c77b4100be5d3b2a86d8c273 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30229e07431b4614a670d22c8443ffa5 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7225e952dc634160ada0f1093c16e6b4 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/787dcb8997034fb5bda5de6ce13eb259 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09ba0057c5074c958845dab01a3ad389 : SUCCESS in 4m 28s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9795f670606047818853ae7d213fc88b : SUCCESS in 32m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3d42b6d38ae54d3dba0e35783d66ac1b : SUCCESS in 59m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0b224634f1cd48b88cd692638ea2d154 : SUCCESS in 1h 07m 24s","accounts_in_message":[],"_revision_number":1},{"id":"430c400398f5b742a11c3b06ff06ab7723b244de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 12:29:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f8f7a463ee2f50f6c30f174c7ad5d5bfdd875083","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 12:31:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13cd89bece6c4002b0750998f23c6066 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/565bcc6396924a18b652298b549c3376 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56d639283e358b731efb78283c319dd3d9f8cf03","revisions":{"56d639283e358b731efb78283c319dd3d9f8cf03":{"kind":"REWORK","_number":1,"created":"2021-12-08 21:55:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/821044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/821044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/821044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/821044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/821044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/821044/1"}}},"commit":{"parents":[{"commit":"fcd525d62b1b3d6ff191242888c9e49c1fce1da6","subject":"Merge \"Add Swift lock path in Swift containers\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fcd525d62b1b3d6ff191242888c9e49c1fce1da6"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-07 15:02:03.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-08 21:55:12.000000000","tz":0},"subject":"Ensure nvme-cli is present in nova-compute images","message":"Ensure nvme-cli is present in nova-compute images\n\nThe nvme-cli package is required to get the nvme command executed by\nos-brick, which is used by nova-compute.\n\nWe don\u0027t need to explicitely install it for centos binary images, as it\nis required by os-brick which is required by openstack-nova-common, but\nall other types of images were missing it.\n\nChange-Id: I754939da7636c57d2a8d5b83debb5d8a58e38432\nCloses-Bug: #1953509\n(cherry picked from commit 30eddf72f6d5380bb6078481581e47b0eec84982)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56d639283e358b731efb78283c319dd3d9f8cf03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56d639283e358b731efb78283c319dd3d9f8cf03"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
