)]}'
{"id":"openstack%2Fkolla~821533","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I231e0ace0822f55cfd7785e68078c577e9902481","project":"openstack/kolla","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2021-12-13 14:12:28.000000000","reason":"Michal Nasiadka replied on the change"}},"hashtags":[],"change_id":"I231e0ace0822f55cfd7785e68078c577e9902481","subject":"monasca-grafana: Fix build errors","status":"MERGED","created":"2021-12-13 07:12:11.000000000","updated":"2021-12-13 17:05:26.000000000","submitted":"2021-12-13 17:03:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821533","meta_rev_id":"f642adf601c8a6635a93460cfb1935d2e0d66b17","_number":821533,"virtual_id_number":821533,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-13 17:03:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2021-12-13 10:43:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2021-12-13 12:13:07.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-12-13 12:05:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2021-12-13 12:13:07.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2021-12-13 10:43:31.000000000","_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-13 10:08:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-13 10:33:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-13 10:43:31.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2021-12-13 12:05:01.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2021-12-13 12:13:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"9ab52b1a316835367dd81f233319273a29905e0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-13 07:12:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50bb8123d2bae2664094d5a5ecd64e55c2c04171","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-13 08:21:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"484c7ab2f2eedd4141929776e98d3270178d75e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-13 08:37:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c6f9ef66be9a50cb27e11f9b1734665afa5b0714","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 10:08:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/fe73f5d6f9ff4683ae262964d23cc9b8 : SUCCESS in 50m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9558e8ad021e4abca378ff059120d564 : SUCCESS in 42m 01s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9e451789008d4a449285455e31f6d70d : SUCCESS in 1h 25m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2f2971649a454ef087bf5caebc657f45 : SUCCESS in 1h 20m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f2144c391ca4398d97bfc7d6d180510898b18cd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 10:33:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e50186472aea4f4bb92898a909b459cf : SUCCESS in 39m 14s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3f4b6e803bb341b98ab99671cbb5ee50 : SUCCESS in 31m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ee817d9d251e4998adee8b54b885d2c3 : SUCCESS in 1h 00m 17s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d71c78ac65e244488435c3dc9e0a5468 : SUCCESS in 1h 02m 07s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fe92a36b12434da3bdfbc3b46ec62102 : SUCCESS in 59m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fc734d17814f40d38eb129892b6c77c2 : SUCCESS in 24m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/15a3d0b126764cd88651a04d97585bb4 : SUCCESS in 15m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f688793c51b248c18ef903767ad1e375 : SUCCESS in 55m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b2ea2177abd4f3fbf2e5458858e2243 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc94ecbf7b46462e8ad5f0e815de880e : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/003fcf5d0f7d4738b72560104ab9d387 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42eacd8ad9f5415a86a85e9e6dc19909 : SUCCESS in 8m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3e41060d16664bfda99077603a3f3996 : SUCCESS in 21m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4fb2485269d649b3b8fd93cc296919cb : SUCCESS in 22m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cfbb4abd20744c4db34384e6137e3424 : SUCCESS in 52m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/caac1149a7ce4939ac05f8abf0c543a1 : SUCCESS in 55m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5b5788d718a94ede8d106652c5d0ece9 : SUCCESS in 1h 53m 52s","accounts_in_message":[],"_revision_number":3},{"id":"583aa2e9b90a8ec5319cac2b5c953592ce29fa7b","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2021-12-13 10:43:31.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"448bd10e9bd15cb1b2a680de6890eb45a8aa7649","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-12-13 12:05:01.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"49815e7b141de3ce4872ebe295288d2d6747f39d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-13 12:13:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d6a071509c68eb2adc599f3ad61e06dd8fa76778","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 12:13:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"35496c2dfe5dba3344ae41321aba34f396c84924","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 13:19:28.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/656dc26231514d6e81102d5ca6ea958c : SUCCESS in 30m 55s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a03c3bd2ca9f4834a58f0007defa8f28 : SUCCESS in 1h 04m 28s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d42d17ede5d14a90b93f3db24d21f9f9 : FAILURE in 24m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caf88963f07d431987175877334e3126 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7a40d47bc374409be5e4c18461e9a75 : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47ca25eeb206415a9ed31db23998bc5c : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13fc3fbacb7441bf80b0b5fa8b4f62c4 : SUCCESS in 6m 40s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/74f8704d790f4332b4383e193d82a689 : SUCCESS in 25m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/debdba0edfa94508be8b8a3947e60397 : SUCCESS in 55m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bd34812c4ee247ae88a4218191741251 : FAILURE in 22m 51s","accounts_in_message":[],"_revision_number":3},{"id":"43a33182d00d0ba0ffd3a11d69b41ca625e4aa04","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-13 14:12:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d4a24e31adba9ce2a4e01d5d3f5c003141b15439","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 15:35:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/bb6c310a6e194adf9adf16b3c59390c9 : SUCCESS in 24m 37s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/76a432b552cc442fbb7600237b57580c : SUCCESS in 32m 00s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0bdf149e2ace4eaabdd4cca19f82f174 : SUCCESS in 1h 12m 50s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3681f2ebe2994eddb52715fd107f0c1d : SUCCESS in 58m 38s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a6e9c38c7df14453a1448235f1038fd6 : SUCCESS in 1h 19m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6f41e662e9534b1f9806883d5533e413 : SUCCESS in 26m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1903240c277b4c06bacc38a6e390e230 : SUCCESS in 22m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9c1200bf1d674c6e81a4923e4220a514 : SUCCESS in 1h 01m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae7147cb2c4a460b804d6452b61517b0 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0906e3f24a81491f86c4c9dd45f640cb : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84a6894a008e4fd098bdd492c6dd4d55 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/044b26f562244343822fc2ee69925276 : SUCCESS in 6m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3de7da49a87a42b2b0e72b67c82a9684 : SUCCESS in 32m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f02a8d99a6f8426ca3dcab1b5bd4de36 : SUCCESS in 27m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5088227c37904e52a50ad69e5c496cab : SUCCESS in 52m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/52d4622a6d654488b1d4f0e2a7971f25 : SUCCESS in 53m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5487aa04cc774e09814ca8869bbc43b6 : SUCCESS in 1h 10m 10s","accounts_in_message":[],"_revision_number":3},{"id":"6f679567f0cc4a2d24d761cd55267a119b070718","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 15:36:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"71872224fb1ad46157020e9c2ca4de9fefb1c996","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 16:16:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/fa0e2df4874d423f952bbc032753ac04 : SUCCESS in 43m 08s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/abedb05d58b947a1a15d0c3dba5b7ba6 : SUCCESS in 36m 01s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ff2c9f7c5ebf4e49a37e838d00429f82 : SUCCESS in 1h 59m 55s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/32fbec46c72d4166ab20f546bf977d79 : SUCCESS in 37m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c3de34bcf4a9acf0f8e3b53d9d32e300c980abd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 17:03:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7753d44703804114810f4dd8c3562869 : SUCCESS in 43m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/886f0a112a51402aabc4840618920207 : SUCCESS in 1h 03m 17s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/33bffe3896dc489e85756f7ce95beae3 : SUCCESS in 1h 26m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca62cbe15a0841769e59855581492149 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d48d1a683ee4e6199c41ecb0c7adc4e : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33dfd6ef74bd45f6a93055df07f7463a : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f73b27726bbd4d9da9b129ce763b87df : SUCCESS in 6m 35s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e53dda9b80c14d979e47ab060f48964c : SUCCESS in 28m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c0c12dcbea3c4824ad4bf5684414343d : SUCCESS in 1h 07m 19s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dec75d9ab431452d86fa642402db5523 : SUCCESS in 1h 16m 33s","accounts_in_message":[],"_revision_number":3},{"id":"ee512e0769ed0ca32616ea7796443f299c23789e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 17:03:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f642adf601c8a6635a93460cfb1935d2e0d66b17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 17:05:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/231eb8657586432c980d909a66645897 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3af23da7ccd07ace68ae5ec6a76b110783fac627","revisions":{"2dce4800d1c217b63eb8d1593c2f4e02866d0276":{"kind":"REWORK","_number":1,"created":"2021-12-13 07:12:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/33/821533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/821533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/821533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/821533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/821533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/821533/1"}}},"commit":{"parents":[{"commit":"5445c1e54a1e96d568bc9f45b066e479640e3895","subject":"cinder-volume/ubuntu: add lsscsi and nvme","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5445c1e54a1e96d568bc9f45b066e479640e3895"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-12-13 07:11:28.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-12-13 07:12:08.000000000","tz":60},"subject":"monasca-grafana: Fix build errors","message":"monasca-grafana: Fix build errors\n\nCloses-Bug: #1954621\nChange-Id: I231e0ace0822f55cfd7785e68078c577e9902481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2dce4800d1c217b63eb8d1593c2f4e02866d0276"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2dce4800d1c217b63eb8d1593c2f4e02866d0276"}]},"branch":"refs/heads/stable/ussuri"},"149639fa7abaa7d2faf6dbc078959888810f5029":{"kind":"REWORK","_number":2,"created":"2021-12-13 08:21:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/33/821533/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/821533/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/821533/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/821533/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/821533/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/821533/2"}}},"commit":{"parents":[{"commit":"5445c1e54a1e96d568bc9f45b066e479640e3895","subject":"cinder-volume/ubuntu: add lsscsi and nvme","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5445c1e54a1e96d568bc9f45b066e479640e3895"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-12-13 07:11:28.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-12-13 08:21:16.000000000","tz":60},"subject":"monasca-grafana: Fix build errors","message":"monasca-grafana: Fix build errors\n\nCloses-Bug: #1954621\nChange-Id: I231e0ace0822f55cfd7785e68078c577e9902481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/149639fa7abaa7d2faf6dbc078959888810f5029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/149639fa7abaa7d2faf6dbc078959888810f5029"}]},"branch":"refs/heads/stable/ussuri"},"3af23da7ccd07ace68ae5ec6a76b110783fac627":{"kind":"REWORK","_number":3,"created":"2021-12-13 08:37:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/33/821533/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/821533/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/821533/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/821533/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/821533/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/821533/3"}}},"commit":{"parents":[{"commit":"5445c1e54a1e96d568bc9f45b066e479640e3895","subject":"cinder-volume/ubuntu: add lsscsi and nvme","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5445c1e54a1e96d568bc9f45b066e479640e3895"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-12-13 07:11:28.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-12-13 08:37:16.000000000","tz":60},"subject":"monasca-grafana: Fix build errors","message":"monasca-grafana: Fix build errors\n\nCloses-Bug: #1954621\nChange-Id: I231e0ace0822f55cfd7785e68078c577e9902481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3af23da7ccd07ace68ae5ec6a76b110783fac627"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3af23da7ccd07ace68ae5ec6a76b110783fac627"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
