)]}'
{"id":"openstack%2Fkolla~821695","triplet_id":"openstack%2Fkolla~master~I3045786e863b098d7339d1066aef6c857aa5f97f","project":"openstack/kolla","branch":"master","topic":"fix-fluentd-es","attention_set":{},"removed_from_attention_set":{"33712":{"account":{"_account_id":33712,"name":"Daniel Meyerholt","email":"dxm523@gmail.com","username":"dxm523"},"last_update":"2022-01-27 13:16:43.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-01-27 13:16:43.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-01-27 13:16:43.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-01-27 13:16:43.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-01-27 11:33:00.000000000","reason":"Mark Goddard replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-01-27 13:16:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3045786e863b098d7339d1066aef6c857aa5f97f","subject":"Unpin td-agent and cap elasticsearch gem","status":"MERGED","created":"2021-12-14 12:12:28.000000000","updated":"2022-01-27 13:19:24.000000000","submitted":"2022-01-27 13:16:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":59,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821695-fix-fluentd-es","meta_rev_id":"264e4e888a1c03c95827dc5277fdeb3df1dad54e","_number":821695,"virtual_id_number":821695,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-27 13:16:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-01-27 10:47:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-01-27 10:58:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2022-01-27 11:33:00.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2022-01-27 11:33:00.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":33712,"name":"Daniel Meyerholt","email":"dxm523@gmail.com","username":"dxm523"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-14 13:41:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-16 09:08:09.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2021-12-16 11:09:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-12-16 11:17:21.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-12-18 09:43:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-12-20 15:28:30.000000000","updated_by":{"_account_id":33712,"name":"Daniel Meyerholt","email":"dxm523@gmail.com","username":"dxm523"},"reviewer":{"_account_id":33712,"name":"Daniel Meyerholt","email":"dxm523@gmail.com","username":"dxm523"},"state":"CC"},{"updated":"2021-12-28 09:36:04.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-01-20 17:50:30.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"6d3aa45fc0586986e52ed18c07242106b81b84e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-14 12:12:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bebe4582cd163bae4430a9065198b1cf2132716f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 13:41:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/c4f8c77a6a4547218d00999d068c2327 : SUCCESS in 28m 46s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0355b451a8d04503b327ae698a904018 : SUCCESS in 38m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/21faf47a2e67478aad0575c98b20674e : SUCCESS in 1h 17m 42s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/dba3fd2b1cac4601a196e82b36c632c3 : SUCCESS in 1h 11m 30s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a3116e41347c49b0845a1b519803ae6b : SUCCESS in 1h 26m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a1f5cfe2cf214ddfab4dc69db2b7dbc6 : SUCCESS in 32m 44s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/27c93e5ebb504d3dbe4327e22f9972ff : SUCCESS in 23m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dbab69547a7649368449ca60ce5d2d07 : SUCCESS in 1h 05m 34s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6f752f15c4894f8f878f9121c82d5ff0 : SUCCESS in 1h 21m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46c34bd174dd43568ecde7e8f629ffc0 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3286cd0631fd414fbd96453b2b98107f : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f20081a959f44e3993ee3e84bbe41f43 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdddb50ec2f34e1faae2a8a1987a581d : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe862bb412154b4e8e94fdf9a9a39ec0 : SUCCESS in 4m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3e40352392b94ccfbd895e1fab9b642e : SUCCESS in 27m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/48d3d7f61c9c4d7a9fd60852cb400dd1 : SUCCESS in 39m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c410bfd623e43b5bc30132efd810cd2 : SUCCESS in 54m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/714115b9f56047bb96cf3eafeb19017a : SUCCESS in 48m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6f3c473116b548bba8dfcc97a5a61b34 : SUCCESS in 1h 06m 58s","accounts_in_message":[],"_revision_number":1},{"id":"1e64a5c6bd8fb715170a2f2646c88b0dd8728196","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 14:22:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c4d8734bb60b4e9d8aa6c15ad967bf91 : SUCCESS in 1h 50m 04s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/741aa4c3b318457bbfa2e226e15b5959 : SUCCESS in 41m 07s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/de4a338984df493cba02c79c9b88567a : SUCCESS in 2h 07m 35s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8817c5dbec40464fbffc24ddfa8fdd70 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5fc3946aa4b3432aae7f66e200afe0a8 : SUCCESS in 6m 02s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/cb497f45881c473dbdfae3953361034d : SUCCESS in 1h 15m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4c6b1698accdcb01b0a6f05e2985f41914d2168","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2021-12-16 09:08:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5e9bcce5765ab3a6eec6411cac66464f166b8e32","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-16 11:09:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ac4d21d6097df88e9c3384256df3ce2e66e554bd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-16 11:17:21.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b49106d192709b2c417e026e5318e36ce857ec91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-16 13:19:40.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2c2f43a54abd0824e6af068259af74c633bba2b4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-16 13:19:53.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"14203ef58621d80bb58e309d6167fc3ffa7a7380","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 13:40:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4bd956b204bf45548d40aa5f06aaa6be : FAILURE in 9m 11s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4cdcf5bce61e4118a451c26de8a98d8e : FAILURE in 6m 45s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5f8d7802e5864ed9bcc67208847cd34e : FAILURE in 12m 23s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/81a1c0da5cb84b61ba2da7015ed0b731 : FAILURE in 11m 46s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7e9efb06d9bf4e2189f4c776ad9ac74b : FAILURE in 11m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d8a09f7e176549f4a2056a799ccaf7f8 : FAILURE in 4m 53s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4b8eb5625db149d688ade31f13efa13f : FAILURE in 5m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/725c23a913c048b7bcf199da09a0ef63 : FAILURE in 18m 48s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/eba9717b446b4373b7e9952bbd180e20 : FAILURE in 14m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7465dc75a004adb8f0c419fb6661e58 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64d3ee63dd6141fdb374f22a4af6cf2b : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d700978856704449ac58b70f6660f939 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f11b74841cd8462885283ac98eaf9d90 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/886399dbc0b044b59a91a24706e73856 : SUCCESS in 4m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5efb9814ec13489dbd378c09cfa56f0b : FAILURE in 4m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5a2240a5f606490093662a192a99a7ee : FAILURE in 4m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1875ad5a1fa8428d8dae8780211d5b07 : FAILURE in 10m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2f18fea0cc304424bed8e908979cf9cc : FAILURE in 13m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/98aaeb2f6390481288df703fee1f49cb : FAILURE in 9m 08s","accounts_in_message":[],"_revision_number":2},{"id":"9fbbc91b3931c956c3f0228a1a61fa9c3b87a987","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 13:44:48.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/63e69e91b32b4594ae67c4cbf8aa20b5 : FAILURE in 6m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8ddb0280065d4aaab1db5196ff254575 : FAILURE in 4m 23s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c38416e119354c11bc484adfa5f0ea59 : FAILURE in 18m 35s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e014c8309c67473c90ad817fa0c741f7 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/562f6cd7c5c6417199716c6f21d832ac : SUCCESS in 6m 49s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b7fdba68c8cc45efa2c5d874feadcd7b : FAILURE in 5m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2003742414da86529ef8d8bfadc9e9ec4863a542","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-16 15:03:00.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3389083296f6a5deaed2bca49d4dcc520e768b9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 16:53:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/c70ae3aa1f8747919253d4c8b6022cd6 : SUCCESS in 36m 02s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9209c97daaa24f088e080cf31e49296d : SUCCESS in 34m 16s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/80e360876a1340939fb3e2a800a18cbf : SUCCESS in 1h 06m 09s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/54663f82aae7457996747fb227ed5b11 : SUCCESS in 1h 05m 09s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9dfacb8577f9484595424ba30f90cdd9 : SUCCESS in 1h 40m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2528c14b7d5c452887372b23e76adfae : SUCCESS in 33m 36s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c097086bf5874886b0acee2cab868aa4 : SUCCESS in 22m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/714085728b7248fdb9df84a0d5841ae7 : SUCCESS in 52m 27s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/68f5d63ae7ae4f76afa703464f8af752 : SUCCESS in 1h 13m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/241f758b95804f9db1a48a79bb51bbaa : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/904fbec9a8384fdf8e68ef63b82eba49 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c906df821f0443d8d8b5f190d50abd3 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/336e4fb8fdc34aaca0f6f39fe584b6fc : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/123c1a4061904687bf444290225b6b13 : SUCCESS in 4m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6a24dfb6079c40a7ac20cbc39f88ad2a : SUCCESS in 20m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2284bb34c68142508ac52b574374a415 : SUCCESS in 34m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5c17bf7ddead4011a414fc6b81f9b3e8 : SUCCESS in 58m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0c468056259b40d69a1036272bb8e52d : SUCCESS in 43m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/32a8e184e1fa4b93bb6d474400bd6416 : SUCCESS in 1h 07m 34s","accounts_in_message":[],"_revision_number":3},{"id":"d0af03a93ce01ba41dad81690d97d4205a5c9197","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-16 17:34:46.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"715c1146fd3b2c20bd487f3bffdd571af15d2fcc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 17:41:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2cfd4a5a7d5a4519b7ef281282bf5c5c : SUCCESS in 1h 53m 16s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5733db52c4ad4fa1947b0829c1dacc38 : SUCCESS in 1h 29m 38s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/98eaa527239349b886c97d7b742791f5 : SUCCESS in 2h 30m 03s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2d1f45af6e54f289c27f4e5c1b91111 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a045eef311c2411f843da45e19d4e2b2 : SUCCESS in 7m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3a6da3430eec42bea54809915bbc4149 : SUCCESS in 1h 19m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"03627a571276085b27749f55378df4cb94e43811","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-16 18:27:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bdf85a1ed3f65f250bed7c9a5be3edd444a1f1cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-17 07:07:09.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"ebb4d8ad16148e53cf3326de4a39953908e426aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 08:21:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7cf87969134748b39693309a211379db : SUCCESS in 30m 08s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a9659609e83f45bf9b9643b874645b60 : SUCCESS in 40m 53s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5e2aede8d8cc4a5c8317a9d08e151eac : FAILURE in 8m 47s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b89c2bcd580246c697cb81e072e5f0e5 : FAILURE in 8m 53s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d93c45d2235649cf86423430a5aed017 : FAILURE in 56m 35s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c4257c85812a4b668f52829cc76765f7 : SUCCESS in 28m 14s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bccdd847fa6d4fbc9c445a39728e8c3c : POST_FAILURE in 39m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bdb256b6196c4c22a887702948749758 : SUCCESS in 48m 41s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/408a06af001644d6a004cd63d9b14dd9 : SUCCESS in 1h 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2163edff05a942e09ba248a639616d42 : FAILURE in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9c4c01557504a628f8e4757e3a3f271 : FAILURE in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1efe1ea089e04d4495491a0946841f81 : FAILURE in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/635e45ca77ec49a0a9b663cb81cc5903 : FAILURE in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfcfc915361d404298286cccb2a3ac78 : SUCCESS in 4m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6428a765773d4f07bf6afc3f9cebabae : SUCCESS in 18m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f6ff70c546b0488ebdab553aa89b82c3 : SUCCESS in 28m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/940d9557ba7a4e0da3df547f3437ee56 : FAILURE in 4m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0cb45f4e60a04e0aab6b278501e5b0a7 : FAILURE in 8m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7e137978ab6c4b0d8cd1e974bbaed922 : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":4},{"id":"4016b04db717a080a71cd21a91d79b15a363c07f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 08:45:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1e012555741c4cdcaa036ddcf512ecc3 : SUCCESS in 1h 35m 20s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e19fbd21690f46ff94d8522d0007788a : SUCCESS in 1h 30m 49s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/afce4413437141ee81fe7d68f35f9b17 : FAILURE in 12m 10s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c7086f82c7d4ec3ba54a7c0242d8ba4 : FAILURE in 4m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/162e3eb5a80e48cc946a2c1967d0ebdc : FAILURE in 4m 45s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/50a577e4fec641ddb06b58751a4e6212 : SUCCESS in 1h 29m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b7602a5e84c82b96561a667ce22c6aa476363c98","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-17 10:45:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bd0a4df841ee58413e0e468919a38a01a96aad53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 12:23:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9158214de56549f1acb5245240972e82 : SUCCESS in 45m 52s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5e6c4b819a01492e82cb3831ce1922aa : SUCCESS in 34m 57s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/61b802c8406a4e53bea571512557b122 : SUCCESS in 1h 32m 04s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d3cf0aeb50e64a4b95d5fa3eb4a9d71a : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7275ee3d8a284d689a294f20d3a7ab3d : SUCCESS in 6m 39s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8abf6efc6a1241fe8016a56cafdc48a9 : SUCCESS in 1h 34m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"71a4ff25b2530b9bfd89e2839a19acda5e107b38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 12:24:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6c2bb647c20e419f89863f5c378dcfba : SUCCESS in 27m 46s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/adc6469910a145439656cac37b410cf0 : SUCCESS in 41m 09s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/081fd93e028743e08000dd79acc80141 : SUCCESS in 1h 21m 46s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/5e89866ee78e4c80b4a64ba188243488 : SUCCESS in 1h 12m 10s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/466ef76d074a42b2b06cc9dee3e2fd7f : SUCCESS in 1h 33m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b2bb60de9d2f452ea2897a93b505ed0b : SUCCESS in 25m 32s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/79959757249b4af8ba4721719d9e036d : SUCCESS in 20m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aa9ff40f7680443aa723cd61f6b650f4 : SUCCESS in 1h 08m 14s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b7295f75934041f2a21d65d3b7064973 : SUCCESS in 1h 16m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1309691011714674bf5abe2e19954e41 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c9c937dd9f24a3ba526b1a6b1a2f8f4 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e07f334ad10444d6a9930729c3df16e9 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eeb0db312314147b95f19778fa3815b : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fc9174c36e246a6b95b71589c2788ed : SUCCESS in 3m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/56ed3e71c91640d4b130df77e27ba414 : SUCCESS in 18m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/68806348fddc49e6a04604ccaad7125c : SUCCESS in 33m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/572837f1c7e747a883b725de875467be : SUCCESS in 54m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/06d37509883440e28fde1b9effe587eb : SUCCESS in 54m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/04d9155364ca40948954fa01384f6faf : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":4},{"id":"1812099fa5c0d6e92837c6bb83ac419624b6a8f8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-18 09:43:41.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0dfed6337ccb4c809e1e0a53cbe0c01775d00008","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-18 09:44:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c608e55fcf366c8ebdfb4fabd94220265c3f24b4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-20 12:11:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"413d29ba1eb14ded364a074b39b168a9dacd10cc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-20 12:20:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"15b82f490e0ebcd905317e58b19fb8629ae919e7","author":{"_account_id":33712,"name":"Daniel Meyerholt","email":"dxm523@gmail.com","username":"dxm523"},"date":"2021-12-20 15:28:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1fa1282f9b43ecd3bdf4ac875616c051800ce20a","author":{"_account_id":33712,"name":"Daniel Meyerholt","email":"dxm523@gmail.com","username":"dxm523"},"date":"2021-12-20 16:47:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ab57f2474f40ab878878defcf35ca8129a6fec8f","author":{"_account_id":33712,"name":"Daniel Meyerholt","email":"dxm523@gmail.com","username":"dxm523"},"date":"2021-12-20 17:37:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"586c579836ceddbd24eb62904bdc6598903bfe47","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-21 12:16:54.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b1b73bbdbc5251b9e80187920ece09ada24f1e41","author":{"_account_id":33712,"name":"Daniel Meyerholt","email":"dxm523@gmail.com","username":"dxm523"},"date":"2021-12-23 16:32:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2607725c8a14b0e5c98496bc51126eb51bdbc0d0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-27 13:24:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d36bf002fad571f94c08959003f62eedf260f610","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-28 07:07:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"05efc84060072c5390dde7447cbf89af3eca9cf5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-28 07:41:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"408b2b7ae89ea2f07ebfcecb50f30343f6e30ccb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-28 09:36:04.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fca4e9546ec44431ee72a37f12ecd669b0a2a5e4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-28 09:45:35.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d304c8fc037983bcb5cfa03b6674e9e48f963ff8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-28 10:15:15.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c49a04bc55a500b2603a1a04ee77013fd198ce53","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-28 10:20:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"99f1774037c984d271fbd33c2022b0a43c684054","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-28 10:23:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c09f36dfb7e0cbd25c06356e626343ae63520590","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-28 10:34:06.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"ce9db653bb4ec4010765344b2a88781b5a9afbfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-28 10:37:23.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b36848db32c343aca400d9f2e09c23710b5e4a6c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-28 10:40:44.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2dd7507635c5f01d59063af221c89786c12d8134","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 12:01:59.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fa92bed048b24b5580567e0cdec95ede : SUCCESS in 27m 08s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/655861de129540d79f27dbe9db7f79c7 : SUCCESS in 50m 26s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/64d7ac6459d9419e9aa6597ccd2f98ab : FAILURE in 50m 46s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/77fa9cb5dcb747efb53abfc2bcbe4bb3 : FAILURE in 57m 26s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1a4e11675d984134ba2c3fd577a9a8b2 : FAILURE in 1h 20m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/287f9a9c54ee4379b13421089b811131 : SUCCESS in 28m 05s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dfafd25ca93e42a38adda9dab90b468b : SUCCESS in 18m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2140ba1c531f4068ae8835fb0e44bbf2 : FAILURE in 46m 23s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7ace6e5d4c6c45c7ad86e4e267647f6b : FAILURE in 1h 07m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf0459b661af44cab427d77533b10c17 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1347270472f40028e9e9dede6404de1 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4126465f3e7944a9ade4dc0be674b525 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d457580b2aa4e50b079324fc30a842f : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83cd12600d4f49e9b2fc3c602df00912 : SUCCESS in 3m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3ca70bd74c6249d1bbbd9615313b73d9 : SUCCESS in 24m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c23b4c5e273f46c289ed0b4083dfe47b : SUCCESS in 58m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/657d65c9d99542c6bdbe0e82d9b908c6 : FAILURE in 36m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/82c19870f60b45fa89737e5ca10127d3 : FAILURE in 39m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5dc9aa6b824348b3ac0abf4a021d5556 : FAILURE in 1h 00m 52s","accounts_in_message":[],"_revision_number":7},{"id":"c3be2eda63249c1024fb0908d1bbf9e962961afe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 12:21:09.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4c484d45d24741a49a862da2f1f62f65 : SUCCESS in 1h 05m 16s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9f620ce4b8604bc399115b4b7c3995b1 : SUCCESS in 53m 29s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/29e752831e2f4aaf895f5bbc7b1e36ba : FAILURE in 1h 41m 36s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5c9c3e9578148748848004b81f76320 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f805c48b3f3d4a76930d9001ce59dc7e : SUCCESS in 4m 46s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7b87be04ada545e28ec0562013df7064 : SUCCESS in 1h 25m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7e5244025c8291614b10257979cd30249b0d9138","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 12:31:17.000000000","message":"Patch Set 7:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/3d540249707d4392999e51176e7dc43e : SUCCESS in 46m 54s\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/a1063dcc45aa410383d52fca867c239f : SUCCESS in 35m 56s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/7d57edd9ad3a40ddb61ed16be3790343 : FAILURE in 29m 28s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b611e37c9d2c47f4b6a3101860be8012 : FAILURE in 1h 38m 59s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/766f34f085c9469597b25adceb904bc7 : FAILURE in 1h 00m 35s (non-voting)\n- kolla-ansible-centos8s-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/b432b284a8c7459facbf71ff62f322bc : FAILURE in 1h 06m 18s (non-voting)\n- kolla-ansible-centos8s-source-ironic https://zuul.opendev.org/t/openstack/build/678ce7862e43469690302f7930ed4c5b : FAILURE in 1h 04m 34s (non-voting)\n- kolla-ansible-centos8s-source-swift https://zuul.opendev.org/t/openstack/build/d33f38767cbf4a728557fabb58672149 : FAILURE in 1h 25m 08s (non-voting)\n- kolla-ansible-centos8s-source-mariadb https://zuul.opendev.org/t/openstack/build/b0190a8a45704e9e881b8d1e0e635aff : FAILURE in 29m 12s (non-voting)\n- kolla-ansible-centos8s-source-masakari https://zuul.opendev.org/t/openstack/build/e67d39d9e9aa4e5caf684887937c8629 : FAILURE in 54m 23s (non-voting)\n- kolla-ansible-centos8s-source-ovn https://zuul.opendev.org/t/openstack/build/d119cd0cc3974326b05d005cbce03844 : FAILURE in 1h 09m 03s (non-voting)\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/81481182b1cf42638449b61732866524 : FAILURE in 31m 02s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/1293bf26aece469c8528f86a878c848f : SUCCESS in 32m 32s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/72f2a5fe90544b00a7e3fda9957a1316 : SUCCESS in 34m 00s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/4995e73250784b76aa2c956b2cd742a1 : FAILURE in 52m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1af0cfac05bb485c890badc84ec36f8b : FAILURE in 1h 19m 27s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d37a48c4d885475a8cbf9d337353099b : FAILURE in 51m 30s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/a76b9b7b2c3840d7b6480029f73e6fe8 : FAILURE in 41m 28s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/8ea91826e0e040e6a04792741c27168d : FAILURE in 42m 26s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/df2a6fbacdc84d19991d6ade74c84ec4 : FAILURE in 25m 05s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/8ccdc9fca3eb46a5bab44ee48ac7bd2c : FAILURE in 35m 37s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/6aceb5587bc94b58a8c065aa69d4758f : FAILURE in 38m 08s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/a9b4ba7354ce4a09adba66b8d021ecf9 : FAILURE in 54m 53s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/5ad2be14c0254d8b9695b82587520264 : FAILURE in 19m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5c2276efbdd853897ff2b7a38632b47ae449cb3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-28 18:10:14.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"718dffb2defb0a5f5f6f1a05339c277bb7a96a67","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-28 18:10:26.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a0fb40d475c762dbfe7e54ca3c590433ac1344b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 19:45:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1a890889410c45eb9dadf964491ceb3d : SUCCESS in 1h 26m 35s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/48d766baaf4a4047952bfb23309f38cc : SUCCESS in 42m 12s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/03db89df30e948cc873227019e92217c : SUCCESS in 1h 26m 21s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/50bcd773fa09424db2c2f7c35e09f9b1 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6273904620d43a683708abec04771f9 : SUCCESS in 5m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/237c8f5fe5ca4c89a303b5978797f471 : SUCCESS in 40m 32s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2f00d4d14c0305b53a07c9795ddbc490441fc565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 19:50:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/79051aa90f1e4aacb41e48e39ec08c03 : SUCCESS in 27m 54s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8c938db77d6547d98b49358e8a7bad67 : SUCCESS in 37m 12s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9e2549c6f15746b38955ec4060d8c670 : SUCCESS in 1h 09m 13s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a841d0f15f714f45b8022286031bd56c : SUCCESS in 58m 43s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ceb8c3a97d6a47909f6492f2d68dc03f : SUCCESS in 1h 28m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ecf4fbbb65804197beaf4076802495dc : SUCCESS in 26m 48s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fef0b9ddf3ba45c492721c6cad15d553 : SUCCESS in 13m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b06698aa4fee40a0869d16cf334d79dc : SUCCESS in 58m 21s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/18538a604dab4e3a8a94e1e486bef8bb : SUCCESS in 1h 35m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6df163a6524d4eb0a952a63c1b76f4f2 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cea16160f053455fb43221cf17a3d217 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d280e4a0a72245569f45c6672267f3c5 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66d2ff485a334e6e90ce167bb47cb2b7 : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a7c3a517c6845188bbf4b767d5e8efa : SUCCESS in 5m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ace29087834a4d90955258a8c0b875cf : SUCCESS in 20m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d071a5619688438eb85ac8807615fec7 : SUCCESS in 30m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c9d7651e58b744709003e41ad0e68e37 : SUCCESS in 43m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2e7b9c1e88a14ee0a27f0121cb5f2e0b : SUCCESS in 44m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/30b4f284361e40cfb245c95103adc2a6 : SUCCESS in 1h 07m 57s","accounts_in_message":[],"_revision_number":8},{"id":"1eee1f9b78abba353bd033cf392b133d4b926afe","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-28 19:52:13.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1467e0d3cb68dcacb15e72c445032e2a7367bf3b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-28 19:52:37.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c8b1d4f468cc3271e3f372405c64c0554cca0f81","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 20:11:55.000000000","message":"Patch Set 8:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/1f14c592d2b24ae9b5c7e5a342ab6c20 : SUCCESS in 33m 35s\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/877d53a40e25499eaf71e40d378d04ed : SUCCESS in 35m 45s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/0ea74e2a0b3b487fb02500617ccb8621 : SUCCESS in 1h 32m 21s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2ff8540ee7d24d1baef835d450f5af27 : FAILURE in 37m 32s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/6e9369aab51141aba011194f0011dc5d : SUCCESS in 1h 47m 22s (non-voting)\n- kolla-ansible-centos8s-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/1f1cbb72c18444cabe69ab252e0eeff2 : FAILURE in 1h 06m 29s (non-voting)\n- kolla-ansible-centos8s-source-ironic https://zuul.opendev.org/t/openstack/build/8285be6e9ae24a5b838231f45cc3f612 : SUCCESS in 1h 19m 59s (non-voting)\n- kolla-ansible-centos8s-source-swift https://zuul.opendev.org/t/openstack/build/273aef252cad49ebbe0fec5a7332a530 : SUCCESS in 1h 36m 40s (non-voting)\n- kolla-ansible-centos8s-source-mariadb https://zuul.opendev.org/t/openstack/build/8741ecaed1f648a7952b5e122bcd9cd8 : SUCCESS in 31m 47s (non-voting)\n- kolla-ansible-centos8s-source-masakari https://zuul.opendev.org/t/openstack/build/75569b68eec44f229d0e2fabe13695ec : SUCCESS in 1h 09m 44s (non-voting)\n- kolla-ansible-centos8s-source-ovn https://zuul.opendev.org/t/openstack/build/2be706c51f6243d8b9e9b3e7980a6cdd : SUCCESS in 1h 21m 39s (non-voting)\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/b7891288fea14ba781303f7a3ed704aa : SUCCESS in 34m 47s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/f52c5b0b3ccc411baabcbf316f08938d : SUCCESS in 35m 40s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/aacdd3d59c4148dbadcf1cc0dd6e6e68 : SUCCESS in 33m 33s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/b5bad3bce96442f083895f07bee9aa50 : SUCCESS in 1h 17m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2cc30a226dd443e2be02696d9a8b559c : SUCCESS in 1h 36m 01s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/337d3e99bf96462a87e8102d1d0008c8 : SUCCESS in 52m 40s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/8780ea2e338443c191833f5df67cafc1 : SUCCESS in 57m 18s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/a53d96760bfe4a8488917b9c04a3e027 : SUCCESS in 1h 57m 31s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/e832c3c562b64fb6bf23274f5bc4fcc5 : FAILURE in 24m 22s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/f37992c7a82644a7832cecca9ef04e84 : SUCCESS in 40m 55s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/1511fe1028144952b67f9da3a4f6cfc8 : SUCCESS in 49m 33s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/6cc27e7a48104663ae4c38ef4640246a : SUCCESS in 1h 23m 34s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/7f588720dade44258c920df6f8faf4e8 : SUCCESS in 23m 01s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2da313a14850eff2cc30d0b82781164142cf6f31","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-29 09:09:08.000000000","message":"Patch Set 8: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"5a2ee1d2d44f5266c9c02bc22dc5b033e7f80536","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-29 09:12:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"59608e0d3f5f00519e2003b350d29fcd14e70ea7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-29 10:17:25.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"0134c61525bf96749371f781a046b875ac9eb88f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-29 10:25:16.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ab442cebfb3982d45236724d9c58791e16845b40","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-29 10:25:49.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"afce3385a081bc282e76a9d0b5ab19b0494d4fc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 11:42:35.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8905be98719347c9bf978c8db9eb5e35 : SUCCESS in 36m 28s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5e6e1526cf7e427faaf19ea52bbfc6a5 : SUCCESS in 41m 59s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/df8b3764df304b09a9dcf164518c152e : SUCCESS in 1h 08m 01s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/1dfbc0ca023249ff987f90a814646c3e : SUCCESS in 1h 04m 20s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ed0e1b769eae4395b8f2181f35c02636 : SUCCESS in 1h 20m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9641d3feb68f45eba62dc7a3c5c8e2b4 : SUCCESS in 21m 18s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fa078af097f64d168c32811adf5a977e : SUCCESS in 16m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/118b36b26d9945f8ae87045e413af22d : SUCCESS in 1h 10m 42s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a407959316ab46358ba49d0809775e6c : SUCCESS in 1h 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09d15e2ad02e44588214c2af4e5dded5 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3328247acf7d444799282097b100636a : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55b4b0f3a5fc4b48a258f9fc8c6d62b3 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fb7291ab05a4d5e8200954e8d29397f : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34220d8862be4844975f2f03cb5063ec : SUCCESS in 3m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1f4eba2197ea47caa57266fd8cd5fae1 : SUCCESS in 18m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ee74d729c40b40508df60657adbd9964 : SUCCESS in 32m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2cc49a7de6164f3f9144cdbd664f2144 : SUCCESS in 49m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a68c0e0455ca47faa96319925d833963 : SUCCESS in 57m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b3308073d86f469092b904bf307eecf0 : SUCCESS in 1h 07m 28s","accounts_in_message":[],"_revision_number":9},{"id":"56a1fcb4e5bd696d8ae8d0a0a66b6ebe8f9de635","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 12:06:04.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/dcce3348cf1e4487b101f3f80761784e : SUCCESS in 1h 07m 09s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a3b25ba3d63343d7b0457574a79241c4 : SUCCESS in 1h 02m 52s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a7f24acdf2a842b98b0295bf1fb23c01 : SUCCESS in 1h 38m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c3fa9ede4f340f9bf4f38bc71c507bd : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66c8d978813147078e5ee3af5de5a5b0 : SUCCESS in 4m 53s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5280296ab6ea43668d86159e66c5c23d : SUCCESS in 1h 07m 57s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"48d6a45906915938097774702853a28b25aad8ce","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 12:22:20.000000000","message":"Patch Set 9:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/764600ab5cf44b329c0f28625544ed24 : SUCCESS in 35m 22s\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/e695f0ba57c94f8598f6062d1c3e760d : SUCCESS in 32m 33s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/8b061f4a2046486db488894d33fa4985 : SUCCESS in 1h 27m 29s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bbbbaff94a9141169ed6591aafd28af1 : FAILURE in 1h 55m 04s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/19c98816c98645b9968e29179bb2880a : SUCCESS in 1h 23m 46s (non-voting)\n- kolla-ansible-centos8s-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/6c9cd4b1564642e9b5527f18e70a9da2 : FAILURE in 1h 14m 42s (non-voting)\n- kolla-ansible-centos8s-source-ironic https://zuul.opendev.org/t/openstack/build/8e106467445a42e5bdcd70452436e057 : SUCCESS in 1h 08m 16s (non-voting)\n- kolla-ansible-centos8s-source-swift https://zuul.opendev.org/t/openstack/build/525db7b87a28483a86291a1d4de8f3eb : SUCCESS in 1h 18m 40s (non-voting)\n- kolla-ansible-centos8s-source-mariadb https://zuul.opendev.org/t/openstack/build/dae6c85abd144d49be3844e23025bf3b : SUCCESS in 33m 52s (non-voting)\n- kolla-ansible-centos8s-source-masakari https://zuul.opendev.org/t/openstack/build/1816d3f6cf474a8584f3f1356677f42b : SUCCESS in 1h 11m 24s (non-voting)\n- kolla-ansible-centos8s-source-ovn https://zuul.opendev.org/t/openstack/build/412ba4be8313411bb32de3d28d7e34c6 : SUCCESS in 1h 19m 09s (non-voting)\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/2d866f82fee14604bdaa832ba4a98d30 : FAILURE in 36m 39s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/7573f189a93e4d47b1f6bda91e282db0 : SUCCESS in 34m 10s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/d4b7221e7a4d4603accc16b4c29fea37 : SUCCESS in 40m 30s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/437c46a8bec641528e7e1571467d7d02 : SUCCESS in 1h 15m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e5cfa32b05d24c19870d07075bfb1015 : SUCCESS in 1h 20m 18s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/83b0214b4fb44a99b4752cbc27835507 : SUCCESS in 1h 07m 48s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/014cb657e79f4b418a5212010a8f2422 : SUCCESS in 56m 01s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/7434d2d1b2d045b4a0433ca7edd519f2 : SUCCESS in 1h 02m 24s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/47d805e4a70c45339c3a40400bbb2a29 : SUCCESS in 24m 28s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/cb18336c0b9842aba2d74c754dfb615f : SUCCESS in 54m 56s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/7265c25ac51c44179cc3899d18cc848f : SUCCESS in 38m 33s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/8d9fa57bdee941da9476fd824551f8d7 : SUCCESS in 1h 02m 54s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/cf3950571bc147689976418a28207423 : FAILURE in 25m 19s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"71b8c1f000885e088b52d3e840136920f7b3f5c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-29 12:49:24.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"563ea82ed33573362c922d56de07c57c152f1431","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-29 13:12:57.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0300614480314b9d6241d1fa50038de89b2e8380","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 14:03:41.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/352ad43e1ce242299cc3208c069cb276 : SUCCESS in 47m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/52fd83b290d9442589d118813707b17b : SUCCESS in 34m 17s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/41c1c2ee7573483b89dfa22cc66b51a0 : SUCCESS in 1h 13m 24s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7267323045042a3aa2887676b94eedc : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e672cac1c07c493f873c3257fe7fdb69 : SUCCESS in 9m 06s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e1d9bc6d3c96459185561379ee170b57 : SUCCESS in 1h 05m 00s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b729cdd21d758c328f59489da7ec12ede338bad7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 14:30:34.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ed20344189564ca78f64ca725edd4c32 : SUCCESS in 26m 01s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e50198d4db754ec1a7db9548e8621352 : SUCCESS in 40m 29s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7c5bd3ad99e74bffb20bd059c740338a : SUCCESS in 1h 14m 22s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/5544cb3bdf6343088e7d6809a0b895fc : SUCCESS in 1h 05m 00s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/887d7ec8930541b3abefcee9a7b13ab7 : SUCCESS in 59m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dd719b0744dd4bc2b3777105fd436da3 : SUCCESS in 33m 23s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0d6b42ddf8f74d57811c6859a4fd5f89 : SUCCESS in 19m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6ec7a3e460844d458445790e7c73b3f2 : SUCCESS in 1h 09m 51s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/69ccb6e18e5b40ef9f769cff15956992 : SUCCESS in 1h 35m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2358d9d552e146eeb1c8de228310d82e : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7ba40ab80a0492b80da701891132072 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef7bcfe10c53456bb840689b6d21dc7b : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce10d42ccb794675aebbe110ecab8a04 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91692b05d86149a98eba20b571f0867a : SUCCESS in 4m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c0ad113c02ef4145972b7c026874a454 : SUCCESS in 24m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/13ce7a2f7f294ff9b4112140e13366f1 : SUCCESS in 31m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2c8addb6abec4c3ead7dfbbc2eaf5478 : SUCCESS in 51m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5524b36663154420b556be33535c25e4 : SUCCESS in 50m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/73256963a1d74bb4905585d8ae001537 : SUCCESS in 51m 17s","accounts_in_message":[],"_revision_number":10},{"id":"ff0bdc04e63df12304310d81489e16a0bdec1ee9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-29 14:35:08.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"26163eab93da084ce0035477668451e89bb3b5fe","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 15:00:28.000000000","message":"Patch Set 10:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/3e0fc987d7ce496aaa3992ae0f899526 : SUCCESS in 43m 09s\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/2dc249d93bc640ccb4332f683bdaaa82 : SUCCESS in 36m 19s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/3462c38b8e154e719515b96daecb25e1 : SUCCESS in 1h 22m 20s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/3bd5811d6018428b9f0a3a89e60aff54 : SUCCESS in 1h 45m 58s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/41070aa004c5411aaec076f1861ad385 : SUCCESS in 1h 27m 22s (non-voting)\n- kolla-ansible-centos8s-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/3282efae9ff6406da2592ca14e88dd9b : FAILURE in 1h 10m 16s (non-voting)\n- kolla-ansible-centos8s-source-ironic https://zuul.opendev.org/t/openstack/build/b68abdb8690d457881d6fdd1bb0d3c03 : SUCCESS in 1h 13m 21s (non-voting)\n- kolla-ansible-centos8s-source-swift https://zuul.opendev.org/t/openstack/build/e18d424924ca4672bc7f8bbcd04b72db : SUCCESS in 1h 07m 11s (non-voting)\n- kolla-ansible-centos8s-source-mariadb https://zuul.opendev.org/t/openstack/build/d4567825f9df49f2869e3eae319b008d : SUCCESS in 32m 57s (non-voting)\n- kolla-ansible-centos8s-source-masakari https://zuul.opendev.org/t/openstack/build/641d4fde798644a7a8144919a7fdbb59 : SUCCESS in 1h 06m 18s (non-voting)\n- kolla-ansible-centos8s-source-ovn https://zuul.opendev.org/t/openstack/build/204e5014a0b6441b9e4455c2ddce5548 : SUCCESS in 1h 23m 52s (non-voting)\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/4dbbbe085c3a4733b52430bcb71e6639 : SUCCESS in 44m 31s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/d5f192a7c31a4e8aa83906ac5b76f2bc : SUCCESS in 31m 17s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b99b1f96c4f4b63bc6a2a3bad0d17ad : SUCCESS in 44m 52s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/c965a0bb42eb43f8a81d22a959bab4c6 : SUCCESS in 1h 21m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/966b7d3aa87a4006a84779adc32ad4b8 : SUCCESS in 1h 22m 31s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/1bc610809ede4f50b057e5908f4e2908 : SUCCESS in 1h 01m 05s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/054f6835e39348f0b1f61fbeda4080ad : SUCCESS in 46m 38s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/adc9bf2dcbeb443299a582c8ec57a779 : SUCCESS in 1h 11m 19s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/5b8f4a4f966643a99e1e355e8e737409 : SUCCESS in 26m 54s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/88b7cab426b6462589ee0ecc90ca62dd : SUCCESS in 45m 27s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/ee6faa0739984a6c9a138562fa416579 : SUCCESS in 34m 10s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/e29f2a822c6140249efb65c2c2215b49 : SUCCESS in 56m 44s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/85392c4d972444a3bff98e390c0c1d3a : SUCCESS in 27m 54s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"daa9c47c4f17522ca172ee1537542edc3052d63c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-29 15:15:30.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"3accc08c020c934cb16c79adf7bb116a39f0b403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 16:38:53.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/14f680c0ca5a4f6694d9a084d6d1bd4e : SUCCESS in 28m 42s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/dcb3647345df420581af835c895c2d0f : SUCCESS in 37m 55s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8dcb0464097b4230aea372edec3731e1 : SUCCESS in 1h 15m 26s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8324ea3b2ac54f6a96783232bacb9431 : SUCCESS in 1h 10m 07s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a63f0d43b1f7490fbee183a5c661d561 : SUCCESS in 1h 21m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/683f8ccbf2614dffa434a1e6bd81d99c : SUCCESS in 28m 07s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6fae3f39183f4632a18e792778a8dcc9 : SUCCESS in 17m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cd3b790a4a324881adefcbc1ffa92894 : SUCCESS in 1h 04m 40s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/708c7b42f3cc4a3f8990050047403634 : SUCCESS in 1h 09m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e957869be2204fd7a11c9315f68d6f9d : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/740f452d17ee4eedbe4dd3b36d6063dd : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79b95e53490641aa88e2002502d3afbf : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f96fd3e407b04b819259bf5af01964ad : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ed0bd475e224d84b5982beb89578fa5 : SUCCESS in 4m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0a94a479943049d29afdbaeabdee2b6d : SUCCESS in 18m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a2a2499b8bb541798f0c1002e421e142 : SUCCESS in 32m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dc6c30e4686f45199b7639b0d4edb3dc : SUCCESS in 54m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/27313f9ff2f8474a81101f5222bf8ac6 : SUCCESS in 52m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/80cb7a2d722946a19bd002ea25242ee8 : SUCCESS in 1h 10m 08s","accounts_in_message":[],"_revision_number":11},{"id":"9e0437b96a09dc29bfa215f87c330ab799382ec4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 16:49:47.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/13011f06bca8479bb4042111924047bb : SUCCESS in 44m 53s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5f23d503050241169c81f1b8dbd8c81f : SUCCESS in 31m 22s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/88a3a0b3013f44e89928d46d17a45c82 : SUCCESS in 1h 27m 48s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1eae6e6571c48a1864dba1ba0b7c544 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/86e4351cb3054cda956345c0c5d33d09 : SUCCESS in 7m 28s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c81633681a0a45ab86db25dab74e0b21 : SUCCESS in 46m 15s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"2a6c8b614a41d7d861b8955a6de2e65bed24dde5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-20 17:50:30.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"84254eb2f648c1e779b9b17fe9b3b411c1a8785d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-25 09:28:07.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"c95baf314974b08cb0baea58d547589174c3f44e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-01-25 10:39:49.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"84431e8557616d5d9f8d39b800f9a1a6210c0aa9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-01-25 12:16:36.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"0855948dc9899cdcc6dd3435fbff0f49b6d9c9ae","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-26 09:47:14.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"4efb5b44d4983910bf290c03cfd0556ba48b387b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-01-26 20:18:55.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"fab55c902c2c3ee613e5a44726f3d8ba846d2e53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-27 09:16:51.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"1af46543eebc6dc042cf480e93b66d0bcb0702a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-27 09:18:30.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"f9f09f46785eabee51396a3739c0c95d780be2e0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-27 09:20:52.000000000","message":"Patch Set 13: Code-Review+2\n\n(9 comments)","accounts_in_message":[],"_revision_number":13},{"id":"970b1560ff4f850c82c61624600ba4bc204dd191","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-01-27 09:34:59.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"fc8e85c13d2ebf7669aa5cd3a796676e97eb395c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-27 09:49:09.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"e19179810f899c62a6b59a74735eb10ac56c7cda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-27 09:49:32.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"ff5e5b0c4982dd2ff643ed9724a161709c5c5607","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-27 09:51:07.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"3691616c52c8ed5554e3ab0e3df9f03f43a501f6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-27 09:51:23.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"534165ceff129178a1e91d746f5323c91f7b01c9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-01-27 10:45:52.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"030a2ab616fc11beda0f6534f5c4c4cab42e7839","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-01-27 10:47:08.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"31f4c5aca63530441699cb74dc7c794704635981","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-27 10:58:58.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"8afcc54f3020dab260001356844b1e93456dbf77","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-27 10:59:45.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"e707ea31222078818cf1cde1cc1de47f76908454","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 11:06:32.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2a789c4d63104bad869b200893e69c38 : SUCCESS in 1h 16m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c5992e7347e14e8b9725abf74ce6f7b5 : FAILURE in 31m 17s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f92743eef9d94c14b82ccc58dd38ef87 : FAILURE in 28m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f18c03f95576485e90a58639a8d35ee2 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/982f63ed03a44be5adda3f5484909336 : SUCCESS in 4m 55s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b14b5f30dded49569bbab3845b658a87 : SUCCESS in 53m 51s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"cfa70ea0e551dc4ca52b1b67b2dcf3d8057f9fa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 11:13:43.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ab98bc11630a4e4c8829ce91268c0109 : SUCCESS in 28m 43s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/46fcd139494a44e6b9cfc4a377df6769 : SUCCESS in 45m 56s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/46c1059a5ad94c54aa2b9a5db930e0fc : SUCCESS in 1h 08m 42s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/33036aee6821407887c5d770f972eb4f : SUCCESS in 1h 15m 09s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3b6b1f75818843348deaa21e855bf933 : SUCCESS in 1h 19m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f0b15e0037ba4e04a52de725108480a2 : SUCCESS in 32m 15s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8fc8ee83e5b54ec7a97d6b1e0d5b7f6d : SUCCESS in 16m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cbb3eeb63446479590ac689d49fb0d45 : SUCCESS in 53m 07s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d43eef7e7d394336a416be9a53eaa18b : SUCCESS in 1h 15m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67cdce9cd4a942619d4a0483f2049066 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39885f4a3c3b4379a7abfb6dd027ed16 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/350e5704228d4110a720418af05c9181 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fe427bf2783475e9ec22a881c891f87 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b8f7ca8b9de4fa5a338b9f3b3f77a1b : SUCCESS in 4m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9a9115beee2a4ca9b96f1ee4612e87ed : SUCCESS in 30m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a8ce42e1b75e41feadd948c3e60fafae : SUCCESS in 21m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/eaf2eb7b530c41dcb21d4de35e0b5378 : SUCCESS in 59m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e3560e1b01f5430094f23ecf720501b0 : SUCCESS in 43m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cf1cf62871bd4d8aa06b022efe4f6423 : SUCCESS in 1h 13m 09s","accounts_in_message":[],"_revision_number":14},{"id":"05aca251c25cfa5ed900b19603d3fa48f66c2291","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-27 11:33:00.000000000","message":"Patch Set 14: Backport-Candidate+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"cf73b09c22d242dbbbd9f6883c58335e33f03fbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 11:37:37.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"6db093d793c6e6abc95f9e94aa572e388072257b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 13:16:43.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3764ef3401124aa6a74034a7ec441d36 : SUCCESS in 34m 44s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a2c0019262c548309ba06fe72b259ad9 : SUCCESS in 1h 04m 54s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1e0ae87ca2e24f8d8d98745d4f6164a3 : SUCCESS in 1h 32m 35s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d6055f518cfc4941a7e5751a0aebe7d1 : SUCCESS in 36m 21s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/27f76ec846bf4d1e84876466cd2b5580 : SUCCESS in 55m 43s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a42d4040a91340d994e49d49f65bd5f6 : SUCCESS in 1h 11m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e63a4a4c8d0a4eef9b7b1403463ad3aa : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15470b5d9ad34905a7db07268adc3651 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6263b885d2be4ce1bb881fd27dd1e3a7 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c0c70e4bd6747ca81338765ba9ae277 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5781fe289f894779b6d4cb5b0ed5275a : SUCCESS in 5m 02s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e8589ef907f745e69f30d621d454dcf1 : SUCCESS in 33m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/92529d3ca08840ba87b19f35aebbca13 : SUCCESS in 52m 31s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/537a1cfbf2584c00b1157d0fecee47d7 : SUCCESS in 1h 10m 31s","accounts_in_message":[],"_revision_number":14},{"id":"da4f5bfbc26c6e7972b53b1ac53568f66c828116","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 13:16:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"264e4e888a1c03c95827dc5277fdeb3df1dad54e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 13:19:24.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad648f7240e04999a155c1d42e866555 : SUCCESS in 1m 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bff2c97feb7849869d0239e733273127 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"a8e6924beaea3e268435156f6bbc34812a69a01c","revisions":{"3adf2e9bd3963c7c2006cf6564e4fbafe221f3a1":{"kind":"REWORK","_number":1,"created":"2021-12-14 12:12:28.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/95/821695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/821695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/821695/1"}}},"commit":{"parents":[{"commit":"5dede104bcdfa72a8da32c46fe0bd27f54221d4f","subject":"Merge \"horizon: add various dashboard plugins for ubuntu/binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5dede104bcdfa72a8da32c46fe0bd27f54221d4f"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-14 12:10:39.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-14 12:11:45.000000000","tz":60},"subject":"Cap elasticsearch gem for fluentd","message":"Cap elasticsearch gem for fluentd\n\nLatest version of the elasticsearch gem no longer works with older\n(OSS) versions of Elasticsearch. This is fixed by capping the version\nof the elasticsearch gem installed into the fluentd container.\n\nCloses-Bug: 1954759\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I3045786e863b098d7339d1066aef6c857aa5f97f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3adf2e9bd3963c7c2006cf6564e4fbafe221f3a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3adf2e9bd3963c7c2006cf6564e4fbafe221f3a1"}]},"branch":"refs/heads/master"},"1642a0e43f1473dbf5d7edfc7b4a263ea05c9aad":{"kind":"REWORK","_number":2,"created":"2021-12-16 13:19:40.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/95/821695/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/821695/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/821695/2"}}},"commit":{"parents":[{"commit":"5dede104bcdfa72a8da32c46fe0bd27f54221d4f","subject":"Merge \"horizon: add various dashboard plugins for ubuntu/binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5dede104bcdfa72a8da32c46fe0bd27f54221d4f"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-14 12:10:39.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2021-12-16 13:19:39.000000000","tz":0},"subject":"Cap elasticsearch gem for fluentd","message":"Cap elasticsearch gem for fluentd\n\nLatest version of the elasticsearch gem no longer works with older\n(OSS) versions of Elasticsearch. This is fixed by capping the version\nof the elasticsearch gem installed into the fluentd container.\n\nCloses-Bug: 1954759\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I3045786e863b098d7339d1066aef6c857aa5f97f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1642a0e43f1473dbf5d7edfc7b4a263ea05c9aad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1642a0e43f1473dbf5d7edfc7b4a263ea05c9aad"}]},"branch":"refs/heads/master"},"7ac219cb8d941b62e309765d033261a9060b3239":{"kind":"REWORK","_number":3,"created":"2021-12-16 15:03:00.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/95/821695/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/821695/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/821695/3"}}},"commit":{"parents":[{"commit":"5dede104bcdfa72a8da32c46fe0bd27f54221d4f","subject":"Merge \"horizon: add various dashboard plugins for ubuntu/binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5dede104bcdfa72a8da32c46fe0bd27f54221d4f"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-14 12:10:39.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2021-12-16 15:02:59.000000000","tz":0},"subject":"Cap elasticsearch gem for fluentd","message":"Cap elasticsearch gem for fluentd\n\nLatest version of the elasticsearch gem no longer works with older\n(OSS) versions of Elasticsearch. This is fixed by capping the version\nof the elasticsearch gem installed into the fluentd container.\n\nCloses-Bug: 1954759\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I3045786e863b098d7339d1066aef6c857aa5f97f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ac219cb8d941b62e309765d033261a9060b3239"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ac219cb8d941b62e309765d033261a9060b3239"}]},"branch":"refs/heads/master"},"71098afc1a527619d244f77d223b602ac2f971ff":{"kind":"REWORK","_number":4,"created":"2021-12-17 07:07:09.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/95/821695/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/821695/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/821695/4"}}},"commit":{"parents":[{"commit":"5dede104bcdfa72a8da32c46fe0bd27f54221d4f","subject":"Merge \"horizon: add various dashboard plugins for ubuntu/binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5dede104bcdfa72a8da32c46fe0bd27f54221d4f"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-14 12:10:39.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2021-12-17 07:07:09.000000000","tz":0},"subject":"Cap elasticsearch gem for fluentd","message":"Cap elasticsearch gem for fluentd\n\nLatest version of the elasticsearch gem no longer works with older\n(OSS) versions of Elasticsearch. This is fixed by capping the version\nof the elasticsearch gem installed into the fluentd container.\n\nCloses-Bug: 1954759\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I3045786e863b098d7339d1066aef6c857aa5f97f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71098afc1a527619d244f77d223b602ac2f971ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71098afc1a527619d244f77d223b602ac2f971ff"}]},"branch":"refs/heads/master"},"e3db4f94d186d3859c1a7aa821b81852cde0b957":{"kind":"REWORK","_number":5,"created":"2021-12-28 10:23:18.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/95/821695/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/821695/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/821695/5"}}},"commit":{"parents":[{"commit":"06ce70731e0d0c506b42e84f2eeb20097b7911de","subject":"Clean up fluentd image recipe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06ce70731e0d0c506b42e84f2eeb20097b7911de"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-14 12:10:39.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-28 10:19:00.000000000","tz":60},"subject":"Unpin td-agent and cap elasticsearch gem","message":"Unpin td-agent and cap elasticsearch gem\n\nWith [0] the version of td-agent being installed was pinned, which isn\u0027t\nsustainable in the long run, so we drop the pin.\n\nLatest version of the elasticsearch gem no longer works with older\n(OSS) versions of Elasticsearch. This is fixed by capping the version\nof the elasticsearch gem installed into the fluentd container.\n\n[0] Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\nRelated-Bug: 1930867\nCloses-Bug: 1954759\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I3045786e863b098d7339d1066aef6c857aa5f97f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3db4f94d186d3859c1a7aa821b81852cde0b957"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3db4f94d186d3859c1a7aa821b81852cde0b957"}]},"branch":"refs/heads/master"},"fdcf260a56c4ca4be5fc9936e00cff6f6f425979":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-12-28 10:34:06.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/95/821695/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/821695/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/821695/6"}}},"commit":{"parents":[{"commit":"b8f5f67880983144043de161de8e37095fc45fa0","subject":"Clean up fluentd image recipe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8f5f67880983144043de161de8e37095fc45fa0"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-14 12:10:39.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-12-28 10:34:06.000000000","tz":0},"subject":"Unpin td-agent and cap elasticsearch gem","message":"Unpin td-agent and cap elasticsearch gem\n\nWith [0] the version of td-agent being installed was pinned, which isn\u0027t\nsustainable in the long run, so we drop the pin.\n\nLatest version of the elasticsearch gem no longer works with older\n(OSS) versions of Elasticsearch. This is fixed by capping the version\nof the elasticsearch gem installed into the fluentd container.\n\n[0] Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\nRelated-Bug: 1930867\nCloses-Bug: 1954759\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I3045786e863b098d7339d1066aef6c857aa5f97f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fdcf260a56c4ca4be5fc9936e00cff6f6f425979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fdcf260a56c4ca4be5fc9936e00cff6f6f425979"}]},"branch":"refs/heads/master","description":"Rebase"},"efe8fdb9ac47c7c8b8a57824d742415767d11536":{"kind":"REWORK","_number":7,"created":"2021-12-28 10:37:23.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/95/821695/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/821695/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/821695/7"}}},"commit":{"parents":[{"commit":"b72571de104c262ba1acd5a11b7272212241d7c7","subject":"Clean up fluentd image recipe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b72571de104c262ba1acd5a11b7272212241d7c7"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-14 12:10:39.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-12-28 10:37:11.000000000","tz":0},"subject":"Unpin td-agent and cap elasticsearch gem","message":"Unpin td-agent and cap elasticsearch gem\n\nWith [0] the version of td-agent being installed was pinned, which isn\u0027t\nsustainable in the long run, so we drop the pin.\n\nLatest version of the elasticsearch gem no longer works with older\n(OSS) versions of Elasticsearch. This is fixed by capping the version\nof the elasticsearch gem installed into the fluentd container.\n\n[0] Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\nRelated-Bug: 1930867\nCloses-Bug: 1954759\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I3045786e863b098d7339d1066aef6c857aa5f97f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efe8fdb9ac47c7c8b8a57824d742415767d11536"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efe8fdb9ac47c7c8b8a57824d742415767d11536"}]},"branch":"refs/heads/master"},"64b6bb9bde1afb6b2091633011a8f8904bb01874":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-12-28 18:10:14.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/95/821695/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/821695/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/821695/8"}}},"commit":{"parents":[{"commit":"f91e44a27f4d2f8f21a4790d3bb3254fc794c934","subject":"Clean up fluentd image recipe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f91e44a27f4d2f8f21a4790d3bb3254fc794c934"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-14 12:10:39.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-12-28 18:10:14.000000000","tz":0},"subject":"Unpin td-agent and cap elasticsearch gem","message":"Unpin td-agent and cap elasticsearch gem\n\nWith [0] the version of td-agent being installed was pinned, which isn\u0027t\nsustainable in the long run, so we drop the pin.\n\nLatest version of the elasticsearch gem no longer works with older\n(OSS) versions of Elasticsearch. This is fixed by capping the version\nof the elasticsearch gem installed into the fluentd container.\n\n[0] Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\nRelated-Bug: 1930867\nCloses-Bug: 1954759\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I3045786e863b098d7339d1066aef6c857aa5f97f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64b6bb9bde1afb6b2091633011a8f8904bb01874"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64b6bb9bde1afb6b2091633011a8f8904bb01874"}]},"branch":"refs/heads/master","description":"Rebase"},"4a4cbed75b60ba35c0f6d59a2dc8c5b24b5dc951":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2021-12-29 10:17:25.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/95/821695/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/821695/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/821695/9"}}},"commit":{"parents":[{"commit":"f91e44a27f4d2f8f21a4790d3bb3254fc794c934","subject":"Clean up fluentd image recipe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f91e44a27f4d2f8f21a4790d3bb3254fc794c934"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-14 12:10:39.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-12-29 10:17:25.000000000","tz":0},"subject":"Unpin td-agent and cap elasticsearch gem","message":"Unpin td-agent and cap elasticsearch gem\n\nWith [0] the version of td-agent being installed was pinned, which isn\u0027t\nsustainable in the long run, so we drop the pin.\n\nLatest version of the elasticsearch gem no longer works with older\n(OSS) versions of Elasticsearch. This is fixed by capping the version\nof the elasticsearch gem installed into the fluentd container.\n\n[0] Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\nRelated-Bug: 1930867\nCloses-Bug: 1954759\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/823155\nChange-Id: I3045786e863b098d7339d1066aef6c857aa5f97f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a4cbed75b60ba35c0f6d59a2dc8c5b24b5dc951"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a4cbed75b60ba35c0f6d59a2dc8c5b24b5dc951"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9d91dfdf373ca25663dce69690db37ee713226d9":{"kind":"REWORK","_number":10,"created":"2021-12-29 12:49:24.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/95/821695/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/821695/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/821695/10"}}},"commit":{"parents":[{"commit":"f91e44a27f4d2f8f21a4790d3bb3254fc794c934","subject":"Clean up fluentd image recipe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f91e44a27f4d2f8f21a4790d3bb3254fc794c934"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-14 12:10:39.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-29 12:44:44.000000000","tz":60},"subject":"Unpin td-agent and cap elasticsearch gem","message":"Unpin td-agent and cap elasticsearch gem\n\nWith [0] the version of td-agent being installed was pinned, which isn\u0027t\nsustainable in the long run, so we drop the pin.\n\nLatest version of the elasticsearch gem no longer works with older\n(OSS) versions of Elasticsearch. This is fixed by downgrading the version\nof the elasticsearch gems.\n\n[0] Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\nRelated-Bug: 1930867\nCloses-Bug: 1954759\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/823155\nChange-Id: I3045786e863b098d7339d1066aef6c857aa5f97f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d91dfdf373ca25663dce69690db37ee713226d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d91dfdf373ca25663dce69690db37ee713226d9"}]},"branch":"refs/heads/master"},"4b6990699b6f5d6402348abd49c273490e87ae28":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-12-29 15:15:30.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/95/821695/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/821695/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/821695/11"}}},"commit":{"parents":[{"commit":"79b74ff220aad8db98c10c124064c85d21e2d4ea","subject":"Clean up fluentd image recipe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79b74ff220aad8db98c10c124064c85d21e2d4ea"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-14 12:10:39.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2021-12-29 15:15:30.000000000","tz":0},"subject":"Unpin td-agent and cap elasticsearch gem","message":"Unpin td-agent and cap elasticsearch gem\n\nWith [0] the version of td-agent being installed was pinned, which isn\u0027t\nsustainable in the long run, so we drop the pin.\n\nLatest version of the elasticsearch gem no longer works with older\n(OSS) versions of Elasticsearch. This is fixed by downgrading the version\nof the elasticsearch gems.\n\n[0] Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\nRelated-Bug: 1930867\nCloses-Bug: 1954759\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/823155\nChange-Id: I3045786e863b098d7339d1066aef6c857aa5f97f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b6990699b6f5d6402348abd49c273490e87ae28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b6990699b6f5d6402348abd49c273490e87ae28"}]},"branch":"refs/heads/master","description":"Rebase"},"e6bea87402202e7b4cd00b30ad96c962e72b6956":{"kind":"REWORK","_number":12,"created":"2022-01-27 09:16:51.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/95/821695/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/821695/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/821695/12"}}},"commit":{"parents":[{"commit":"b6e057d694d90d50ea278411d0edf20afe427e13","subject":"Use distro provided GRUB efi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6e057d694d90d50ea278411d0edf20afe427e13"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-14 12:10:39.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-27 09:16:39.000000000","tz":0},"subject":"Unpin td-agent and cap elasticsearch gem","message":"Unpin td-agent and cap elasticsearch gem\n\nWith [0] the version of td-agent being installed was pinned, which isn\u0027t\nsustainable in the long run, so we drop the pin.\n\nLatest version of the elasticsearch gem no longer works with older\n(OSS) versions of Elasticsearch. This is fixed by downgrading the version\nof the elasticsearch gems.\n\n[0] Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\nRelated-Bug: 1930867\nCloses-Bug: 1954759\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/823155\nChange-Id: I3045786e863b098d7339d1066aef6c857aa5f97f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6bea87402202e7b4cd00b30ad96c962e72b6956"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6bea87402202e7b4cd00b30ad96c962e72b6956"}]},"branch":"refs/heads/master"},"78e1a195043302e8bda111e67a7ed6db9acaa564":{"kind":"REWORK","_number":13,"created":"2022-01-27 09:18:30.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/95/821695/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/821695/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/821695/13"}}},"commit":{"parents":[{"commit":"b6e057d694d90d50ea278411d0edf20afe427e13","subject":"Use distro provided GRUB efi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6e057d694d90d50ea278411d0edf20afe427e13"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-14 12:10:39.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-27 09:18:21.000000000","tz":0},"subject":"Unpin td-agent and cap elasticsearch gem","message":"Unpin td-agent and cap elasticsearch gem\n\nWith [0] the version of td-agent being installed was pinned, which isn\u0027t\nsustainable in the long run, so we drop the pin.\n\nLatest version of the elasticsearch gem no longer works with older\n(OSS) versions of Elasticsearch. This is fixed by downgrading the version\nof the elasticsearch gems.\n\n[0] Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\nRelated-Bug: 1930867\nCloses-Bug: 1954759\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/823155\nChange-Id: I3045786e863b098d7339d1066aef6c857aa5f97f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78e1a195043302e8bda111e67a7ed6db9acaa564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78e1a195043302e8bda111e67a7ed6db9acaa564"}]},"branch":"refs/heads/master"},"a8e6924beaea3e268435156f6bbc34812a69a01c":{"kind":"REWORK","_number":14,"created":"2022-01-27 09:49:32.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/95/821695/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/821695/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/821695/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/821695/14"}}},"commit":{"parents":[{"commit":"b6e057d694d90d50ea278411d0edf20afe427e13","subject":"Use distro provided GRUB efi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6e057d694d90d50ea278411d0edf20afe427e13"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-14 12:10:39.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-27 09:49:32.000000000","tz":0},"subject":"Unpin td-agent and cap elasticsearch gem","message":"Unpin td-agent and cap elasticsearch gem\n\nWith [0] the version of td-agent being installed was pinned, which isn\u0027t\nsustainable in the long run, so we drop the pin.\n\nLatest version of the elasticsearch gem no longer works with older\n(OSS) versions of Elasticsearch. This is fixed by downgrading the version\nof the elasticsearch gems.\n\n[0] Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\nRelated-Bug: 1930867\nCloses-Bug: 1954759\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/823155\nChange-Id: I3045786e863b098d7339d1066aef6c857aa5f97f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8e6924beaea3e268435156f6bbc34812a69a01c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8e6924beaea3e268435156f6bbc34812a69a01c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
